)]}'
{"id":"openstack%2Fswift~922870","triplet_id":"openstack%2Fswift~master~I48a345eedbf2369d481d18c0e2db37845673b647","project":"openstack/swift","branch":"master","topic":"py312","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-07-16 02:35:00.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-22 16:50:27.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I48a345eedbf2369d481d18c0e2db37845673b647","subject":"Use entry_points for swift-init","status":"MERGED","created":"2024-06-26 19:22:08.000000000","updated":"2024-07-22 18:19:42.000000000","submitted":"2024-07-22 18:19:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922870-py312","meta_rev_id":"f638070c14b11971843a5055f91b21c29329e113","_number":922870,"virtual_id_number":922870,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-22 18:19:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2024-07-16 02:35:00.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-22 16:50:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-26 19:50:46.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2024-06-26 20:36:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-26 20:39:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-16 02:35:00.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"575480da6795aafea2290dbccd2959d9cbaf2dd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 19:22:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac6aedb78168379151850f60ec4bc24d8339e66c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-06-26 19:50:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7bf269019cb43147618e51abdea054557b6b23b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:36:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac4b4c74f88046a29620fa7a402c4165\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/319094dce0be4787a5d44b3619704376 : SUCCESS in 18m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a86b7e4a5ba24b299287eba7f46ae75a : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/56448352e573452d9f9a2aecece24948 : SUCCESS in 22m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b133578e51545a4abb63685768faa99 : SUCCESS in 1h 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"de7a38e0e1d099bd39259f91eb447e88b0bb6827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:39:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7164c3bf7754cbf81c51ea6fe539c86\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4d378cc6eb104b319d239d7338676cde : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4dccc8d4b14b2d89de0b92b66f9426 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/11e0691391514c27999e775fd05cd27c : SUCCESS in 1h 02m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4503ecb5bf8e49c6b9763907bd7e97f4 : SUCCESS in 53m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b626b4a81f9f4ddfbab3dbd993021af1 : SUCCESS in 1h 03m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3935302fbb2452a8c8ffa12252f09f0 : SUCCESS in 22m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5c22fc5ba0344e38d95049116da73ad : SUCCESS in 22m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb9c930104a8431d851ab40f3e4f1750 : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f2711e6cf6f4b86adbffb8a0aadc578 : SUCCESS in 11m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e693c051134747258d76092f23434825 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/acd4372af1254f6a8189516ed827865c : SUCCESS in 18m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/90eb899941134e21a49fb73f5d09b74b : SUCCESS in 18m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1e078d4f4b1e4853a21364474f83be61 : SUCCESS in 16m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c877f54a269d43aa89d13bbc129225cb : SUCCESS in 18m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8b65dab089e446c08a356e37a33f18c5 : SUCCESS in 24m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0408f190f3941eea9b06e90e1db1b75 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0e0b72a3ed1143ccb5dfed3a2f31c3eb : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25f5b6d16f9146c9addcc20472ff300d : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/54de44a5a5bc46fab4381b459e85e0e4 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bfe9f5c664a3466a84306756e321aec9 : SUCCESS in 20m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6d1607c96d6c4aa79ab5030cf72c1481 : SUCCESS in 19m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/775a310a43f040589dc8b89fedf30a8a : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1d8e6e5412ab4d458ae73e1e7d8dc57a : FAILURE in 4m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/26243d02b489438ca646bf0913d8e6f6 : SUCCESS in 6m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/53925e9aca22447292ddecd3f411fc86 : SUCCESS in 52m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0fc30bb10744ff683d84d8815027746 : SUCCESS in 55m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aace647af91f41e0bd409e1b8fed9397 : SUCCESS in 42m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d143643715dd451fa7de2ae056bcbd08 : SUCCESS in 19m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0be3d9b9b214a4489fba2ff748ed40e : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23143c288a7d409bb56872083a5f00cb : SUCCESS in 18m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b361bda5d7ba4fa49793c05a2dd23d09 : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1b26b4f221188c01a8a43b3734afb84282f4da66","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-03 21:47:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9b9905e59d89ee64c9f3f0222fd4d52c24a30fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-15 21:58:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6710aa5c598e5cafbeba6a2f61305a52c63fbca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-15 21:59:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"53429420b70aa2267963ade8a5d9eb3b1734625b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 23:11:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97c68be11352492698ea3eeacf00c8d8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd41f03d2582488782cb598c45617642 : SUCCESS in 1m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9df898d45024301bf5c67961105f4f2 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/21ac907108f54d4fb01e20bc1e663522 : SUCCESS in 1h 05m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26d6fc165ae74a0abcc23969f50abdd4 : SUCCESS in 48m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a2f68382bf54f5788cc5493071919bd : SUCCESS in 1h 00m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d68ee192335a4e308a0f76b66d534e2c : SUCCESS in 22m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbb17869732f40f7bc00be4c74e38be6 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de8e7737ab6144d48a1563685666ede0 : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7fb38bb2e84f47e8894af2d25b22711d : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a3074fddbc294f4fa1a2f77bfc4bdb9a : SUCCESS in 13m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c758c0b98604f7a8e1077c83535be55 : SUCCESS in 15m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/09fc083f82f341d9b7973efb1ee7146f : SUCCESS in 14m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8227a7b6065d4ebd8044540e44d8e330 : SUCCESS in 27m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/da765e9eeb5d4fd69230b0d54343e892 : SUCCESS in 27m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c6b8cf94fbe42bcab77d3dbdbb90cdc : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f74eb03643e14e119158aac6e68d7aac : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/622b57f47a5d433b9bcf02045436589a : SUCCESS in 19m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/597a54c129e540b48dd792beb0c9514f : SUCCESS in 19m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c526519df1864ce7951b2c93fd490266 : SUCCESS in 19m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a19dcbe0a34948c8bd60d86fc23143f0 : SUCCESS in 20m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/149492ef8986432084228ceb04773f22 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4fc63620b3e04044bbd61906b8636788 : SUCCESS in 14m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f0b49a173a2340d0a66ccf1192876a01 : SUCCESS in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/622eb8bc28074c64bd835d32508907e0 : SUCCESS in 47m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdb5e46f632149938195f11b7b1b28f9 : SUCCESS in 46m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de13ab748b114ea791ca0876a943eb94 : SUCCESS in 54m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9681e7af3acd4a2ebaf8094c86c6145c : SUCCESS in 18m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e69e817a2c54277afb09772ae96af04 : SUCCESS in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd4bca1ffaf54fe7bcd859df24cc163d : SUCCESS in 18m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26aa898f6557491e9d6644821331cf57 : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c57b9087d800696527c12e88c64e1e60c80cbd74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 23:32:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d82082e27d1a47c1af773166dac0293a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/371aba28252646c8bba19ceb12ee2804 : SUCCESS in 19m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53e200926b2d4fab8d3a5e0303fba506 : SUCCESS in 55m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/145f193d812b47519bf5a23f919bed28 : SUCCESS in 23m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcedc8c7d57a4397a6a5bb7f7da3d979 : TIMED_OUT in 1h 31m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ce3300c136ec70169fe59132a49070faccd81b89","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-16 02:35:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"657e7cbd4d8fdd84e91f0952da0f96c4c63ca03d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-22 16:50:27.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6d6decd66d8786dc49fca50fbd0419a74c028457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:51:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e58cbe7c12a8eb5c793e6529c48413898c1227c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:19:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30d6d4b6bd564aa3853d37a215f4ddeb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/be7ff19ca89647f19f59e8763dd464ec : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b317abf0201043859133fb05c7b3d02a : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/d7136d72fb644117bb90446048ce917b : SUCCESS in 57m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1bb49560da094b7daaa520c7bc9e89fc : SUCCESS in 58m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46a38d3eebd7405db3c65a839f1ea171 : SUCCESS in 58m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/35f52400b5434c02b8f40fc3c5e714eb : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b55c40f68e484f0683508f4262898b59 : SUCCESS in 20m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0babe09ceb154fa88d4e34c4966efbbd : SUCCESS in 19m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8bce603ac0094c8e99d8630adb89bb9c : SUCCESS in 20m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f8461e10ad6245de926e333224783a0c : SUCCESS in 28m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/aee8ef64224149618dc60c0e6b293e48 : SUCCESS in 27m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aaf1a1739a9d42799a262789a07563bd : SUCCESS in 20m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/becfd2cf5804429383c5fd65d640690f : SUCCESS in 16m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a0ece9f8cb0f4567bbf47319ceacd3a3 : SUCCESS in 19m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/37814448600d4cb0a40fc15ec9c61860 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c266e88afa764180ac3de19fe839e973 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/80802a4574a4451d811d2d00864e9d51 : SUCCESS in 18m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45b8172c730a47a2a716c5db7d125563 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7dbb28efdf994c7b9f40e9e006bff910 : SUCCESS in 10m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/12270a6551154622859323576347529e : SUCCESS in 49m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/42aa2210eb6846f99c3d0a3220521050 : SUCCESS in 50m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/87756bd48ab0401da10dd3500c0c5b7d : SUCCESS in 36m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97020424d5e14fe29c1a778a0928252d : SUCCESS in 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e82bfca1cf4c40950297a2d52c289a : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5beed88d86a4a53925fdcc92d058b15 : SUCCESS in 16m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5704b95743141eb921e8151e26002c3 : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f638070c14b11971843a5055f91b21c29329e113","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:19:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"006fa6b3a7504b828f74e39cfeee635a20f3b709","revisions":{"f8c98fc5307877cc9cd2444e65b087570bde018b":{"kind":"REWORK","_number":1,"created":"2024-06-26 19:22:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/922870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/922870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/922870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/922870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/922870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/922870/1"}}},"commit":{"parents":[{"commit":"e36acc994304100ffbfe4bd8d4023aefb8c9e242","subject":"Use entry_points for a bunch more executables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e36acc994304100ffbfe4bd8d4023aefb8c9e242"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 23:28:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 23:28:30.000000000","tz":-420},"subject":"Use entry_points for swift-init","message":"Use entry_points for swift-init\n\nI *think* swift.common.manager is a reasonable place for it?\n\nChange-Id: I48a345eedbf2369d481d18c0e2db37845673b647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8c98fc5307877cc9cd2444e65b087570bde018b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8c98fc5307877cc9cd2444e65b087570bde018b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e36acc994304100ffbfe4bd8d4023aefb8c9e242","is_merged_in_target_branch":false,"change_id":"Ia43d7cd3921bc6c9ff0cee3100ef5f486fd3edcb","change_number":922151,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"006fa6b3a7504b828f74e39cfeee635a20f3b709":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-07-15 21:59:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/922870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/922870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/922870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/922870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/922870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/922870/2"}}},"commit":{"parents":[{"commit":"8f9d7db7f12527f0d848d0c89c60755ee14bc58b","subject":"Use entry_points for a bunch more executables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f9d7db7f12527f0d848d0c89c60755ee14bc58b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 23:28:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-15 21:56:29.000000000","tz":-420},"subject":"Use entry_points for swift-init","message":"Use entry_points for swift-init\n\nI *think* swift.common.manager is a reasonable place for it?\n\nRelated-Change: Ifcc8138e7b55d5b82bea0d411ec6bfcca2c77c83\nChange-Id: I48a345eedbf2369d481d18c0e2db37845673b647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/006fa6b3a7504b828f74e39cfeee635a20f3b709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/006fa6b3a7504b828f74e39cfeee635a20f3b709"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8f9d7db7f12527f0d848d0c89c60755ee14bc58b","is_merged_in_target_branch":true,"change_id":"Ia43d7cd3921bc6c9ff0cee3100ef5f486fd3edcb","change_number":922151,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
