)]}'
{"id":"openstack%2Fswift~922872","triplet_id":"openstack%2Fswift~master~I2718d8b96193d13f913824fc1091d72737d755ea","project":"openstack/swift","branch":"master","topic":"py312","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-25 03:49:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2718d8b96193d13f913824fc1091d72737d755ea","subject":"Pull swift-*-info scripts into swift.cli.info","status":"MERGED","created":"2024-06-26 19:22:08.000000000","updated":"2024-07-25 03:50:28.000000000","submitted":"2024-07-25 03:49:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"922872-py312","meta_rev_id":"66aa70d4ca5e7eb96c2d11403cf585ec2c350b80","_number":922872,"virtual_id_number":922872,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-25 03:49:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-07-24 06:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-07-25 02:33:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-26 20:50:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 06:24:11.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"c06bb2bb9108732d7abbbf39d34c8d73a02fe1d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 19:22:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90ff0bf42fbd51a035682bc7075c200853d73ce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:50:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2947f2e18684cdf90cac6cbcccfa69d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/66f0622b7a1f4db6b29b27e16c820dc0 : SUCCESS in 1m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0ea95dd0d24f869ffdf5001be62a55 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/417091db367c497890ee904264483e11 : SUCCESS in 56m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1e61054aa87419f999db037afd70d30 : SUCCESS in 44m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d67d3531a9394e4d8c4e3c4b25fd97a0 : SUCCESS in 1h 20m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af43f7b86212479ebe59e1362b98fc7a : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0944906c81c34ab985dba18d20647b2a : SUCCESS in 27m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ee9a0a8db284b198865e2686ceaa9bb : SUCCESS in 18m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/993f9b26738a4072b58ecf5446efaa49 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/88fd4c856f7a47e9bdce20d4380ea4a1 : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2c698cb786424b848b3a48a1e4e79fd6 : SUCCESS in 18m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8000f487cf054e8a9855ffdd53543cb2 : SUCCESS in 17m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/54d4d7e1648b492aa68177fe4dd95bc4 : SUCCESS in 16m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ea24615ae01f410481bb6505bda47bbb : SUCCESS in 17m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/03c4ace2d03e429589945301ea73ff23 : SUCCESS in 21m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f53fa26a23084cabb2e66879acf7c563 : SUCCESS in 20m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54b3e5f5d4c1455faa6943ae2e21852b : SUCCESS in 17m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/38ec2c615815442787dadb212c1932f6 : SUCCESS in 20m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e8ac0f75c9eb4192b187e24b93281511 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7f686dc8c13244278834b16a9a9f8d97 : SUCCESS in 18m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1b52011c3d0745e88e2cc81e9ffaab17 : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c94dbf0b5b1f4b6bb738ef5c72fe7610 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f397eac0439343efba1a256c89ba60c1 : FAILURE in 7m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fe7b20d13c1c47b19228c068b820e1be : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bc00c5cb7704e06a820aba6f87ce9de : SUCCESS in 1h 01m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/790ad6677cd34604b3bace97c91199c1 : SUCCESS in 47m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/258158c1edc94e45a43da7ad8397e842 : SUCCESS in 48m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e116e21b08d74af7bb15cc9799b020f0 : SUCCESS in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fcbbe361b27481a9f78e7adf2de8b5e : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/595ff9e43cf94e66882571008b114bbb : SUCCESS in 25m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc4272e99d0a468691b276ee43a771b2 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ae424673d72ba48cdbc0bef710c90c612ab1e7d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:50:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41231ec72335496cb5bcd8f71a8c66e4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75e4be535f3d460f8528458b5e2b8192 : SUCCESS in 18m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ddb472a86cd94e1f83fc3f49e208beb7 : SUCCESS in 1h 26m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c92a3570785248b9babac4c8499860ec : SUCCESS in 20m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c944285881e140c0b66c2261606d0207 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8bda728c33c2fda65352a49de08409b42635ce2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-15 21:59:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a94fd5edc1ac0ac6035e298733cf3f70d8425114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 23:11:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbb53fe750c04a3eb75a77b7b5ebf571\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/66d3241ef58c4939ad2aaab0726520de : SUCCESS in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a9b517f98594462b89dab306f876c4f : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/679ef0b16d55490fac820ed8b3a256b1 : SUCCESS in 1h 00m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/12880b04592642b586d3af2ccd631f60 : SUCCESS in 47m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/631906e478fa4dc58a2ac2526587ea7a : SUCCESS in 55m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebba19d7b2cd4e398205741fe2971c59 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14170777b97947de8c41606a30f1de47 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d450841c06b42c69d9c62a7c67283b0 : SUCCESS in 24m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4f2411bcfd349e6b07c4667df9f686a : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59326f5cd12c4c5a8618c34ff4afc51c : SUCCESS in 21m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9b1de9bebb754258a4666712cc8e1e72 : SUCCESS in 13m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/72ade29ee29a41edb9e491c9c43f2691 : SUCCESS in 20m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/73ad9b8f446143a39556ea65378d4de9 : SUCCESS in 15m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bc2e8e373da84eb1af7b8f666cf9ed82 : SUCCESS in 34m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f231b3a1cc814336bb7d553ee670bdc7 : SUCCESS in 19m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df54cfd7022e4547a762a6fbaa0af994 : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a08a8cce3a34e9da4043ee3c5acb215 : SUCCESS in 21m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c45c097321b643b98451ffc489836fce : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/21c38c99d3954d17b5701482d590d97c : SUCCESS in 17m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d990affcacf34f4bb9af66a2dce96a83 : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d5ecd5d09923421b8fdc84c32beecf4a : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45dd2caa4707426881ddfe44ab9bb86c : SUCCESS in 13m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9e913c1619c9449e927bd051230fea65 : SUCCESS in 6m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fa401d8ce75463f8ba0bc5cbaccd00e : SUCCESS in 1h 04m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/42bd3049927a4c05ad05f5a1cafa3621 : SUCCESS in 46m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/57ce33f8f8e1417784a7f52fd838cccc : SUCCESS in 53m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7f2bf2529874a299e31251e2f40b606 : SUCCESS in 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70bc140eee9641e2a49bfc7ac7029a12 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/757f8303631544449d68145dd0f52736 : SUCCESS in 23m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5535fa5bdc78463ead3eeeba8f3c813d : SUCCESS in 45m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f7c25074e05f07a1b2fc86b93d7e6182310f2fa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 23:32:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a84f95854d4249e192e9a15d3e714ae5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5da67b064b1f41539ee90d56b00da4d0 : SUCCESS in 19m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b0e887a83f24668a198ffaa36f24f05 : SUCCESS in 1h 06m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/df5a3678c1984e2f80b84348c0982f85 : SUCCESS in 21m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ddf432f59c542ad98b73d1c52a9eba6 : TIMED_OUT in 1h 31m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d19d5fbec48c407ed558cc7f4ae6b60e303b11e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-23 21:47:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f296c068f627c5e90b9b14460d4c78f10abfc1c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 23:00:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/181ac4d8f9f143a7b949e3b33288edf0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0859288e1e534d8ea575478771a46444 : SUCCESS in 18m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b28303265c8f4375bb28d2b19e9ea286 : SUCCESS in 56m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/23d95b1751e34b8c9f54594ce4a5ab7a : SUCCESS in 20m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/95b72460820a4977aeeb7ac7ada5912a : SUCCESS in 1h 10m 59s","accounts_in_message":[],"_revision_number":3},{"id":"ca9a6392b44d5e8349fd7dda2444061a5155d92b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 23:01:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cca151acbe14bba8efa829fe4b91d8b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe28c4121dad4b9aa62ac3fb4f4d1c6e : SUCCESS in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/490da045279b44b1ad31fff5088b9cb8 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/84714a51bf7d4603b4cdeef5ac219434 : SUCCESS in 1h 04m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/55b2a12b409a449eb47a91b87fba18d5 : SUCCESS in 51m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e646ca2a9dd14520bf6c566f396c5c66 : SUCCESS in 1h 07m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ab964d927ca46e7a57ed274031ebc95 : SUCCESS in 27m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22cf6192f50a4f37946613b7dcc5eb0d : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cc8ba8c23c243feb33848245374ce2e : SUCCESS in 27m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e21e31b086f14cb08ed983c47a11fd98 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b47e76461c24041a620906b4d3c0f7d : SUCCESS in 21m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1103b19560c24942854d7b8f8b33de07 : SUCCESS in 24m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cf70b77cad7541eaa5419c009c72ddde : SUCCESS in 25m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9f1df3e9f9cf490a9519f8e570b3c525 : SUCCESS in 20m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/36bd8b738f6a41018bcbd03c225b7636 : SUCCESS in 22m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/15fd04c160364af688ff2e98335aa33b : SUCCESS in 20m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/710c964e1e8741aa8aedc4c0cd1e2dc1 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/de5d7a21871c40a289cfd7e93afac4fd : SUCCESS in 26m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/63a76dc28726427f8fc2ad51ae80b988 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ea3bd73cfd50492cb59cf4c0a52f9d97 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1b058cb1a576486cbba5360f11a03434 : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/af1734701e824cd6922b16cc772e871d : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49e9dc57aef84db49fbbc7bf41266c00 : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9c0646ec2e8e438da4740fa0cfdd2257 : SUCCESS in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bdaad17883e4551966bdb40dc17bef0 : SUCCESS in 51m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/631c7858b910448689d1a1a2368c364a : SUCCESS in 52m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3e3222fe9344225bee1729c2de2583a : SUCCESS in 51m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c757c2a1b4e7427089df3888ea975f4a : SUCCESS in 20m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448ed81e4ac546649fe59f914d5be608 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77c8287de9d64e3c8c7d80413213110d : SUCCESS in 18m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9fa799932954642ad68674907af7489 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":3},{"id":"6637d4451a1b6123da8f7bcb0a764cdc52bee6b1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-24 06:24:11.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c4f0957e340c0b585e6d80aea7f19e3d2f10bc0e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-24 06:25:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"34639ddceceb1f511a24fbb291e23ebc7e02f0fe","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-25 02:33:33.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"24b1a15f4f1159b5adac44dcf8709f4e2a3a7bc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 02:34:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"061035760ce70837f7bc6af8158d692d11b9060c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 03:49:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ab2326256b140a1afd8c13ef75333d8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6f11d83a6ebb4037860d9bd3b5887d8e : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72c21523d424780a81fbf3c541c8dfc : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/08027cd4102442cfafeb460a27729119 : SUCCESS in 1h 08m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7d639af4df34b5c8ec7d2b94552e43e : SUCCESS in 45m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b074945bae2d4b6ab4fc87760ac0a2ab : SUCCESS in 46m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f7899832e0d34ee4800170dbe602779f : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/19459ea8e925463490e78881af44fd94 : SUCCESS in 20m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fcadc74b69a44762a0d8cdfee736f097 : SUCCESS in 17m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4d0e3cbb6c2e40efbbc72f399011d27d : SUCCESS in 19m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/af742af259b148a1a501b6a7bf426fac : SUCCESS in 24m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a57c0db8dd34484eb449ae92d9cc8fde : SUCCESS in 26m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/20cee584329d4ee794772362cfd0469a : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e8cabcb11604c94ae82b4377c84dc92 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a1fb80fe62e4ad7a2e334c1b712274d : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/acf246da6c3e427f9f23258f3cce97c4 : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6310a280ad174a0b9bcb9f209ae23686 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/513280a8cdf445fbb52cf7bc2537b549 : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2ade219e047448cb9cb6b1af69b69f67 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/33c721700cfa41eba7c922b8ff90024b : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d240d838de54235b20fa69f347d5f5c : SUCCESS in 55m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1478b4c478f46cc9f40c244b882e393 : SUCCESS in 49m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a7e73063285458aa74b7de4c417d019 : SUCCESS in 48m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b109ecc6f5754b639c916c85609cb8df : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db120eedb1c244ce8f56360e9e5e3709 : SUCCESS in 8m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb71c06837b74e438f66b9ffdcf67a73 : SUCCESS in 18m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dba2c4a8b8a846f7bb63c9f2a7d428bd : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":3},{"id":"ce93d566e5631d584e595b7ae916f42be43be41e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 03:49:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"66aa70d4ca5e7eb96c2d11403cf585ec2c350b80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 03:50:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f6f635f3f54472b9e8b1097b0596886\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a682ce70387f448f8107404e3342ddfc : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1780610c9a4b497eb209cac4d51f37ab : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2c5dc64d2562030ba4976b1c88462e1ebf8717c4","revisions":{"88a936534376aea0010fe1bedd953d084831b486":{"kind":"REWORK","_number":1,"created":"2024-06-26 19:22:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/72/922872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/922872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/922872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/922872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/922872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/922872/1"}}},"commit":{"parents":[{"commit":"3329e3aab7bbfe3005666b424f194011a25d1e7a","subject":"Pull swift-dispersion-populate to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3329e3aab7bbfe3005666b424f194011a25d1e7a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-26 19:15:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-26 19:15:07.000000000","tz":-420},"subject":"Pull swift-*-info scripts into swift.cli.info","message":"Pull swift-*-info scripts into swift.cli.info\n\nChange-Id: I2718d8b96193d13f913824fc1091d72737d755ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88a936534376aea0010fe1bedd953d084831b486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88a936534376aea0010fe1bedd953d084831b486"}]},"branch":"refs/heads/master"},"4ab3a63ff180083bb6990c075926fa65eff11186":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-07-15 21:59:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/72/922872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/922872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/922872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/922872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/922872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/922872/2"}}},"commit":{"parents":[{"commit":"2e06b771edfc874448939c25748fb248f01884ab","subject":"Pull swift-dispersion-populate to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e06b771edfc874448939c25748fb248f01884ab"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-26 19:15:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-15 21:56:29.000000000","tz":-420},"subject":"Pull swift-*-info scripts into swift.cli.info","message":"Pull swift-*-info scripts into swift.cli.info\n\nRelated-Change: Ifcc8138e7b55d5b82bea0d411ec6bfcca2c77c83\nChange-Id: I2718d8b96193d13f913824fc1091d72737d755ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ab3a63ff180083bb6990c075926fa65eff11186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ab3a63ff180083bb6990c075926fa65eff11186"}]},"branch":"refs/heads/master"},"2c5dc64d2562030ba4976b1c88462e1ebf8717c4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-23 21:47:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/72/922872/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/922872/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/922872/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/922872/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/922872/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/922872/3"}}},"commit":{"parents":[{"commit":"da12a55bc5e943827dfb3186a161fdeb02900edc","subject":"Merge \"Pull swift-dispersion-populate to its own module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da12a55bc5e943827dfb3186a161fdeb02900edc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-26 19:15:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-23 21:30:32.000000000","tz":-420},"subject":"Pull swift-*-info scripts into swift.cli.info","message":"Pull swift-*-info scripts into swift.cli.info\n\nRelated-Change: Ifcc8138e7b55d5b82bea0d411ec6bfcca2c77c83\nChange-Id: I2718d8b96193d13f913824fc1091d72737d755ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c5dc64d2562030ba4976b1c88462e1ebf8717c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c5dc64d2562030ba4976b1c88462e1ebf8717c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
