)]}'
{"id":"openstack%2Fswift~922905","triplet_id":"openstack%2Fswift~unmaintained%2Fvictoria~I1a2f22d592fc7245e40c645000026e4ba1fca528","project":"openstack/swift","branch":"unmaintained/victoria","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-27 20:18:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a2f22d592fc7245e40c645000026e4ba1fca528","subject":"Run swift-tox-func-encryption-py37 job in the gate","status":"MERGED","created":"2024-06-27 03:52:55.000000000","updated":"2024-06-27 20:21:50.000000000","submitted":"2024-06-27 20:18:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"922905","meta_rev_id":"6a95e4f639afa0d0c4e21667e7a85dff2e73707c","_number":922905,"virtual_id_number":922905,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-27 20:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-06-27 18:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-27 16:40:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-06-27 18:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-27 05:02:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-27 18:35:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"cde58988bb7d1cdfedcb9341602b83656ca2460c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 03:52:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2e33066d5ed763779bb5404a6a758ed93a3578b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 05:02:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c79ccf0a5fa402eb46a29da412e689a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eae77b3d66b4fb8b978560167dfbe00 : SUCCESS in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/4191a35e9a864121bfc9661cf8a57d46 : SUCCESS in 1h 05m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45ccf798ed064ddb9a10dd255dd8fc44 : SUCCESS in 42m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7de1ceb2061a4811935a99064d3fe778 : FAILURE in 3m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e849181d750f413a9a7a8f97888e6ee4 : FAILURE in 3m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c4570ec3d28244acb702a90ef9c5a5c4 : SUCCESS in 13m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/54b9a4ab15004a1e833d3d0f51106880 : SUCCESS in 12m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7917e7148ab24d57b9a338b426ca5ddd : SUCCESS in 13m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bd34f99655a944c19382a039e643ac3d : SUCCESS in 12m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4fb1ddafd3ec403c97b3c61376f75847 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c56a734f59784a87847b510ebb88440a : SUCCESS in 15m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/026e88fdb1d94aae9412632b91bf2c70 : SUCCESS in 16m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/47912ad4619a4b1d948777cf2607145c : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cd521edd170d433fabbaea6274373235 : SUCCESS in 17m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9def3b7197d848a589d9a1e6fd080e79 : SUCCESS in 16m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9429786296c9431185604c882fdb42e7 : RETRY_LIMIT in 1m 57s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c318b00168e0470baba6b5c3bcf6f06d : SUCCESS in 38m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/11004a609cff41d0ad470a392af8eabf : SUCCESS in 41m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bef57425c4d40a5b85b550dda8ae2c0 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3a52c5d287c4c5c9cc961d13ca9c698 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3eef39e5c4c4c448caf0cca998c94e4 : RETRY_LIMIT in 2m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20b597b49d1c43f68457b6e2adc9f8b3 : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8387b4c608bb59afa1fbc818997b236aa61da807","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 16:40:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7c53136e7d00213c03fb1187401dcf6e14375b19","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-06-27 18:35:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd161ef0281a459f8bd3d55b4c1c3dc654e65c1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 18:36:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/nova on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"nova-emulation\" job stanza:\n\n  job:\n      name: nova-emulation\n      parent: devstack-tempest\n      description: |\n        Run compute tests using emulated AARCH64 architecture.\n      # NOTE(chateaulav): due to constraints with no IDE support for aarch64,\n      # tests have been limited to eliminate any items that are incompatible.\n      # This is to be re-evaluated as greater support is added and defined.\n      irrelevant-files:\n  ...\n\n  in \"openstack/nova/.zuul.yaml@master\", line 333","accounts_in_message":[],"_revision_number":1},{"id":"1894ef69abf4553a1d4faac81b3e91b9b57949e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:18:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed31999c37514a02a224e5725413d75b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f378f95cf2b54dc68d7d5a5199d05d8d : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/0f87caba74b14b31b3332d71d4a7439c : SUCCESS in 1h 05m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/796c2e70b4054ebd9dc89811c18da66e : SUCCESS in 55m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/38ab46177c344faa9ea3656244d36edb : POST_FAILURE in 5m 31s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3cf41058b3354016ba365339346f0c67 : POST_FAILURE in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f1133cd844b04f60abff65f2d2fa15d8 : SUCCESS in 13m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3561fc50238b49cf9d6b243a3a937f3b : SUCCESS in 14m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d9005e55ae494f4daa0cdab53082fc7c : SUCCESS in 13m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dc606b921fd2497dab7d5dc5a0ef4ccd : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c49a57245704428acdc26c69db6dfd1 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e736825a87ad4440bedc7a3cf4e9c254 : SUCCESS in 16m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7a58600274f54a69b8c0875979736e0e : SUCCESS in 16m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b76480b187db45fea5cb8bb695f5cff4 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1cbc0d8cca864b7dbfa4e78aae7b4f79 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3e5d91a9eba34b7b847ba2a386cbd894 : SUCCESS in 16m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93771dd3cb07445c8325a314a28b2756 : SUCCESS in 43m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90ec9946dd5d4d348a4f394979481ed2 : SUCCESS in 43m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1348ca76094a4073bbf3605a1eec83c5 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a654b29c077b47cdbf9d24dc36fb5f09 : SUCCESS in 6m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/460bb2fd0c0d4e42aaec0606acf4d90e : SUCCESS in 50m 03s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/nova on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"nova-emulation\" job stanza:\n\n  job:\n      name: nova-emulation\n      parent: devstack-tempest\n      description: |\n        Run compute tests using emulated AARCH64 architecture.\n      # NOTE(chateaulav): due to constraints with no IDE support for aarch64,\n      # tests have been limited to eliminate any items that are incompatible.\n      # This is to be re-evaluated as greater support is added and defined.\n      irrelevant-files:\n  ...\n\n  in \"openstack/nova/.zuul.yaml@master\", line 333","accounts_in_message":[],"_revision_number":1},{"id":"c69c8de2c2604f9dc2ed52585edf26a2147f896f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:18:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a95e4f639afa0d0c4e21667e7a85dff2e73707c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:21:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d6d3e1001054572820d36c5aa71d74e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e0e5fa4c9e5440ea50b0cf454421d1b : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/04407b22e3db41909da2c8266114e452 : FAILURE in 25s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/45995979b4724d0686bd1043728c6031 : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"283b81c2bcafe55c9200166a6d993f51b023cd58","revisions":{"283b81c2bcafe55c9200166a6d993f51b023cd58":{"kind":"REWORK","_number":1,"created":"2024-06-27 03:52:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/922905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/922905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/922905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/922905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/922905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/922905/1"}}},"commit":{"parents":[{"commit":"42950baf14bfa701ed552de5a1bfed43c068a46e","subject":"Clean up centos-7 based jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42950baf14bfa701ed552de5a1bfed43c068a46e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-17 22:46:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-27 03:50:53.000000000","tz":-420},"subject":"Run swift-tox-func-encryption-py37 job in the gate","message":"Run swift-tox-func-encryption-py37 job in the gate\n\nThis used to be the name of the py2 job, but now even that\u0027s gone;\nI\u0027m not sure how Zuul let us get away with this.\n\nChange-Id: I1a2f22d592fc7245e40c645000026e4ba1fca528\nRelated-Change: Ia9ae0fc226dfc9b40157faebac100c10a9180c62\n(cherry picked from commit 318c1a5660d3fd52502dd6340d0a1cba952703a6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/283b81c2bcafe55c9200166a6d993f51b023cd58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/283b81c2bcafe55c9200166a6d993f51b023cd58"}]},"branch":"refs/heads/unmaintained/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
