)]}'
{"id":"openstack%2Fswift~922983","triplet_id":"openstack%2Fswift~master~I3e40ebd78a9f8e93c905b24a12f5f54b2d27c2d9","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I3e40ebd78a9f8e93c905b24a12f5f54b2d27c2d9","subject":"docker: Remove \u0026\u0026 chaining in install scripts","status":"MERGED","created":"2024-06-27 21:34:43.000000000","updated":"2024-06-28 00:26:57.000000000","submitted":"2024-06-28 00:25:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922983","meta_rev_id":"741cf42e4d622726693a036f9e3811d95a065675","_number":922983,"virtual_id_number":922983,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-28 00:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-27 23:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-27 23:14:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-27 21:44:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-27 22:54:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c63191472ed35ff84edeb73b38870ae81ae4229c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 21:34:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7c5091604b373fedff9a0bad25c411baa2bf82f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 21:44:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68156d4d2af047449f716c9fb8016e75\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d617ca7ed1d440bbda0f2a602499274 : FAILURE in 7m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4fc7a76df59e4948b3cd2cbba7ae91b2 : FAILURE in 7m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b13a24ea96b48a987ffdfd55f178b37 : FAILURE in 6m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1ee17b4985c497daf23bffbafc7f26d : FAILURE in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3077a197d059a4667abe8910b3670f5f9aa31510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 22:54:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6ab0f45f5174cda89df1b9882288398\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43f4cc1d7ae4c9bac2282219098b5ed : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/c20cefa5d03f42c89b6686f3c9bac771 : SUCCESS in 1h 00m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70fecfaff1ef4f0a90dded8e0457deb1 : SUCCESS in 51m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a2f3fb74df24db3a7378921e6e142df : SUCCESS in 1h 01m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee6164fe808c4509800c894be88ef0b9 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac0606335b7b4b819801274ca2f8996d : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07c0935f062e4252a04e04e60c2d04b5 : SUCCESS in 21m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbfc00c32edc43a2b3ccd07527af9eff : FAILURE in 7m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a8ca4aed465f4b4a92291a7038083069 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33e2817c08cf4537a4209baf5dd7ee34 : SUCCESS in 19m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb37c80c68d843c1ba653db198f37649 : SUCCESS in 18m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7f432ec51bad4331b69b8728158a8132 : SUCCESS in 18m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cb356ec776d5424c87e49d35d82e6527 : SUCCESS in 14m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a10a172c5bec41fa9a8a06a4091a8286 : SUCCESS in 20m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3426a644584c488ca20e76a12dc94397 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6fd1d889d8a54b41bd7ab25d0aead8da : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8308e156dfaa4f93861d84b3e4be6389 : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/802a57861d6c4db18dc25f876e8865df : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e9fae97d4c3747c38c5554990781377c : SUCCESS in 16m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e8f41556b93940e69ea4314ad9576446 : SUCCESS in 18m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0d04843d20ae4d43ba4ebf20370507b5 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9813a0cd7a6e451282c0e73376f41959 : FAILURE in 4m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/40b7213b19fc4366b03e650feab25657 : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/689016b0e6af4253b1e187ab2135067a : SUCCESS in 1h 07m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6e446e70b664c78b46d5023fce405e2 : SUCCESS in 50m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d8e2d4950854d5eb757215d726e172f : SUCCESS in 41m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1226a0720de4314b04a52829e04c559 : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b128d3502874b56b026808206c01e8b : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be27f78a77dd4109841f49d143edc436 : SUCCESS in 18m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a091e4d2a20494aa5e0270e7c95e779 : SUCCESS in 50m 21s","accounts_in_message":[],"_revision_number":1},{"id":"804c080870485ab931b430815961c3de86912322","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 23:14:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f3f1f50fc30467817cd2dec5de323dcb297cd49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 23:15:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82ecd58043170577f6524693907ae25e12647b7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 00:25:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a73b2649294afa9602c02fb5012209\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aeba790e2d54cdfb41b872a3b4681d9 : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/78aebcba7214450eaa778c82dd0aced2 : SUCCESS in 1h 05m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7a9eb9e567f45369afde522c0c3066c : SUCCESS in 57m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/734c0b88334948a584f2afe03814bae1 : SUCCESS in 57m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f117f5b5ff5c468997f115f0954f69bf : POST_FAILURE in 13m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/79dba267fabc4653bccef1fc4bcf8379 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb0878681e5f47148c7d853db097d3a2 : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f4e3d97a4a64706917d2e8e5304458c : SUCCESS in 20m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e651aad65ee543849bf701bfc245d4bf : SUCCESS in 18m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b1bb1130a7444b82851dd633fa0d07e0 : SUCCESS in 17m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e5f16718836d404bbd010cdaf0d96ffe : SUCCESS in 27m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d21888a66b1742f3815dd1a3296056fc : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed7c836f0b6b46e6800a57d71577cfae : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7f76a96c756d41d8ae717a6daaf1e53f : SUCCESS in 21m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/89e0ea1c4acd4c85ab06455f1094dad7 : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3632127043074f3aa0e12d371d6e1034 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7eb6c88f6ca74f45915b2824048071b7 : SUCCESS in 13m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2930bf44bba46efa8f58076522389e3 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aad92e6775434387860483611b962a3e : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef1e8c08b02942d0beeacd9355b67ba8 : SUCCESS in 1h 02m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4047172501f744c08e9d2d15cafe37f5 : SUCCESS in 45m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d87f196754a4e98b5bd36a933d96c09 : SUCCESS in 43m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb124d6eee4049fbbb0440bbff86c6ec : SUCCESS in 20m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b68061729f46058fd9db08931c1e4d : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc346e1a71bc49ee831c204d744624ef : SUCCESS in 18m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f9c8fc251c94224aabf22bb020f2305 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"83afaf70238f0e8dcbfb40c7307edbfcabf8af0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 00:25:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"741cf42e4d622726693a036f9e3811d95a065675","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 00:26:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f7aa1b0e61425ca394878d13d9fef5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c75f9134cfb45098716a80e6efb98c5 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d8e00ac34d684c4b97b74c535076766e : FAILURE in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/31ee508d6c8948159122695dd1fea5fb : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f29067631ee2e01c8fcc056bce13b689cb663e59","revisions":{"f29067631ee2e01c8fcc056bce13b689cb663e59":{"kind":"REWORK","_number":1,"created":"2024-06-27 21:34:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/83/922983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/922983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/922983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/922983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/922983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/922983/1"}}},"commit":{"parents":[{"commit":"37169aa6ac58082737f31c893d923427428329d1","subject":"Merge \"proxy-server: log correct path when listing parsing fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37169aa6ac58082737f31c893d923427428329d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-27 21:24:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-27 21:24:37.000000000","tz":-420},"subject":"docker: Remove \u0026\u0026 chaining in install scripts","message":"docker: Remove \u0026\u0026 chaining in install scripts\n\nSomething about them threw off the exit-immediately-on-non-zero-exit\nbehavior when building.\n\nRelated-Bug: #2070029\nChange-Id: I3e40ebd78a9f8e93c905b24a12f5f54b2d27c2d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f29067631ee2e01c8fcc056bce13b689cb663e59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f29067631ee2e01c8fcc056bce13b689cb663e59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
