)]}'
{"id":"openstack%2Fswift~922993","triplet_id":"openstack%2Fswift~master~I6119e054289eac263ff5448d7d118209f98678d9","project":"openstack/swift","branch":"master","topic":"py312-wrap_socket","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-06-28 17:23:11.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-07-01 23:04:16.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-01 21:52:42.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I6119e054289eac263ff5448d7d118209f98678d9","subject":"Python 3.12: do not use ssl.wrap_socket","status":"MERGED","created":"2024-06-27 23:38:02.000000000","updated":"2024-07-01 23:05:37.000000000","submitted":"2024-07-01 23:04:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922993-py312-wrap_socket","meta_rev_id":"d429800b1283aa707f5fe508deafa3c8e0738bb0","_number":922993,"virtual_id_number":922993,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-01 23:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2024-06-28 17:23:11.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-28 16:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-01 21:52:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-27 23:42:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-28 00:28:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-06-28 00:40:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-28 04:23:55.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-06-28 16:27:26.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"d3d1b28ba8e0a667cef12efd5e87de53f894a10b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-27 23:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"220a5f05cfd6c877f90b0cc05157a9f10edfb502","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 23:42:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7354c94289614aa9942b7b1e2cdae3bd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb5ddeed3c6048e8834421c8a4f17e89 : FAILURE in 2m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43396eaf06cc43d7b59823934e86c2a1 : FAILURE in 3m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/86ae68cdf9b5419f847e7ae1596222d5 : FAILURE in 3m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5cf7fb928f50496caca049aca53d0e2e : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"30c9732b7dd5cf5b4add571dc62367d8e1101569","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-28 00:28:02.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b7f08d329c145064686461b65c28f2d4ea8a122e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 00:40:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9dc5baaa4fe04cb0baf5fe05b68c3313\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b9cb82101cf4b0284fcc870996c2a01 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/d7bea51aae804af195b0d9b12d5a7525 : SUCCESS in 46m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ac6587336864446a8cc400b4736e626 : SUCCESS in 55m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58c159e05794404480788d45b1cba33c : SUCCESS in 1h 02m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8743eef596c477188ba05fb3fec0d76 : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aefbfa4d4e5c4532a1d264ca891a4186 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f97abf6cdbc4a3ca24a5bb9c4e629e9 : SUCCESS in 21m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86bce2774edf42119403babb0512b259 : SUCCESS in 9m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5f8ef8bf0c9e443fa2abd2a985473dfc : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40f8e784675940859dd7aa826eb33689 : FAILURE in 19m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/972f524aa4ee4e73a1a4b62275894bcd : SUCCESS in 18m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7de99bbdd96e487abdae9a7c4c486138 : SUCCESS in 17m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cac1e2f3afee48868ea1fd9e9b52ed9a : SUCCESS in 17m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f32a5db8515949de8fad480a5ffa98c3 : SUCCESS in 22m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28d8add084da4b3c961eb14dfc9c17e0 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f7b3ec0e49441499125377971c18af4 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f31bbd37dabb4546a2aab2a28e383504 : SUCCESS in 18m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/36644f7567404b2f8ca91351ebacbd84 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e7579f86d060408ba007423551f5a7f2 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e1563a6c46c948a0ae525f36568c4ad7 : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/582c89caf46d468eaca47d5d84add3f1 : SUCCESS in 9m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e558f6a85a5466bb66a125c83aa96f8 : SUCCESS in 14m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4c0e80593e0d4ae6951baf5b89fc4743 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed369cbfb5754a82b8bcbbc15be0c1a7 : SUCCESS in 51m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8fd3646a776f44e88848d2f50216a9e4 : SUCCESS in 49m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ff9626fa35d49b895d2d7ce2b5ab44e : SUCCESS in 36m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b414229f35c5463cb7f060bf55542ce5 : FAILURE in 18m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eca3c53b3cd34b12ab68f7449740b1ec : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab2495bcba3549d5b17a9075064f111a : SUCCESS in 18m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f936df77919241019434835b55051a5a : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7fe4cada44bc2ca3dd1245d04e442bbd3d235eb8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-06-28 04:23:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f88e089a3ada1bd4236b060be89c2511fa61485f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-28 16:24:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"53fd2ac8225dca87db5f37152d37bb9f4c03c002","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-28 16:27:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"388988280e27ffb051e3c26b8dcb9aabca42d59e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-06-28 17:23:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aefea48871ac870e192288a5d4ff9080b0ae128d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 17:35:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fd6af786a984a0b9fde8ee2b93d1761\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d45662f645e42fea614c59a9ffe9043 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/f76cc3518c62457eab6acb839882c8d9 : SUCCESS in 1h 01m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b606a7248038447bace167b3e8e5131a : SUCCESS in 49m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3718e2c44bb54f9bbc849eec0981651d : SUCCESS in 1h 00m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66e155bcdd9549abad68b2372e532e22 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71a1ecaa71a34ad0959456f8d8bfdbcb : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1847b4fbeb344a6180063bcf6796cd14 : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4a82e5fcc95429ba25f4e6a62f77311 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f79e0c0f33694904b494a20e157ebbae : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e63a0bfb153e4fe7bf0f0f856aea1783 : SUCCESS in 18m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1ddbbbe13e064db8883e9b21e5c759f6 : SUCCESS in 14m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0b660e91766d4e6d8cffdfdc44098b71 : SUCCESS in 17m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/de326189a74440d38e924f578b3600e1 : SUCCESS in 24m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b7c8d1092f9f40ca8c05f60f0fbe28d7 : SUCCESS in 18m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a34e61b9275e45b7a33709f5808f6f2d : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af11c9ce7bd6454bbdcfdc084f045dcc : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8586bb69666849ee908ef916607e0d77 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e57c5e114bd74166b0512581764b00d3 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0192e1a4aed4476fb5369a2d81a72b47 : SUCCESS in 18m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a95fd1b198744e2f8f93f775c95f694a : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bcfb49a41dcc40b486f54d5775e74f24 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c3658c2100a742f68b6e8d4c81e254ac : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6823fd2363946df8ec56aab7c8f45ba : SUCCESS in 59m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef22a3d9b6564e41a626f63899a2fc28 : SUCCESS in 52m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/28ef167c20e0436986b1bc4ee65b5b08 : SUCCESS in 58m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d1e74f5986142c0b0a0d689dc6a07a9 : SUCCESS in 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0daca0e67da94a789844463b9930905d : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01c227a0f77445af986866d2cfc048cb : SUCCESS in 16m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3435bcc1f5ee4e36891e4f1b63dbead9 : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0e706c928920b8227cc592c334799c8794acb164","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 17:37:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bab4a03ef0544c0b3b3a8ab07f8759d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f1672f2493e467d938c6b9a6f46c6e4 : SUCCESS in 15m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf8c8dd24093425fa618eef4c396aaef : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/00663412879943b4b218e28bfae60389 : SUCCESS in 24m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e0b7934332643d589b3f28721843e06 : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":2},{"id":"65d5a065fd1e060dc04e98a1a2b2eae0ea70fe0a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-01 21:52:42.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ccfc885aa6aab6c912bd7d2dea75f3663d14096f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 21:53:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3ddd10e0579d0b02d0b3d183af6972a1e594e11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 23:04:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e636d6a731c74fcb8c7ccd8e58b6b814\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b526997454c423c85030e58f80c7323 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/3850d095b4064e04a8c5181e2f3b2b71 : SUCCESS in 51m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9ed8b3b32b04c09a9b140ff57e97ec4 : SUCCESS in 53m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/689aa25162ae437788bf551c29217272 : SUCCESS in 1h 05m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1f0f086c6d174dbda4d546d7e83724ce : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5511520fb771423b89feaa1e12563a08 : SUCCESS in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/131e4e8dc90648f0a1d4e9773adc57f0 : SUCCESS in 20m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/00cf8271e2184fb99369a5ececec7837 : SUCCESS in 17m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9b1d3aa623fc4f068ddaa2f10f46675a : SUCCESS in 16m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/98431247f8414339a49af1e597ab6121 : SUCCESS in 22m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f444d2038fef42809d2370a5682c28b0 : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/831744bba11b4952ba11b9ffb553c3a4 : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fbd2a103ca5e48eead38150243bbd2e9 : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3d20d157bea54f53a1d5e79b1d5b20ac : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4504007b98e44fbcae3dccbcb92bb3ae : SUCCESS in 17m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/06f6adaf6f28438eaa1ea33cd767e87b : SUCCESS in 13m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ff8ccc8d0484730a768302ec2dbedfb : SUCCESS in 5m 40s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cde0e7ee6339442389dd5d2fe859902b : SUCCESS in 5m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/360fcb8af6c841d78fe3d898a90475fe : SUCCESS in 50m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ead1abbf9f0842cc82c485929aed71d3 : SUCCESS in 46m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bddec4721e0544c989a97488adf05e62 : SUCCESS in 41m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de4f0ed99f3c448fac1aa8225b3dc44a : SUCCESS in 17m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed22e202a16414db5d709702fe4036e : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6eda166d94644ef393d64fca3137737d : SUCCESS in 15m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51c1056492924446bd1bc63e1b27ec90 : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":2},{"id":"dc72c210bfb2dcdce02f4f14029fc58291c22b0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 23:04:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d429800b1283aa707f5fe508deafa3c8e0738bb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 23:05:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a7485ac56547efbb70ac4f10ea500e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ef7b696e8d0421d83fec3956c1778ab : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6d7ecff5c4304d2d800e4f46e8ccd4e9 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"64086c30915349b1b6440aa2de36f753bbb0741c","revisions":{"ef5201ce522b4250433671ec33c8bd7d01dc7f36":{"kind":"REWORK","_number":1,"created":"2024-06-27 23:38:02.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/93/922993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/922993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/922993/1"}}},"commit":{"parents":[{"commit":"37169aa6ac58082737f31c893d923427428329d1","subject":"Merge \"proxy-server: log correct path when listing parsing fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37169aa6ac58082737f31c893d923427428329d1"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-06-27 23:37:09.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-06-27 23:37:09.000000000","tz":120},"subject":"Python 3.12: do not use ssl.wrap_socket","message":"Python 3.12: do not use ssl.wrap_socket\n\nThe ssl.wrap_socket method has been removed in 3.12.\nSSLContext.wrap_socket should now be used.\n\nChange-Id: I6119e054289eac263ff5448d7d118209f98678d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef5201ce522b4250433671ec33c8bd7d01dc7f36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef5201ce522b4250433671ec33c8bd7d01dc7f36"}]},"branch":"refs/heads/master"},"64086c30915349b1b6440aa2de36f753bbb0741c":{"kind":"REWORK","_number":2,"created":"2024-06-28 16:24:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/922993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/922993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/922993/2"}}},"commit":{"parents":[{"commit":"37169aa6ac58082737f31c893d923427428329d1","subject":"Merge \"proxy-server: log correct path when listing parsing fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37169aa6ac58082737f31c893d923427428329d1"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-06-27 23:37:09.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-28 16:14:58.000000000","tz":-420},"subject":"Python 3.12: do not use ssl.wrap_socket","message":"Python 3.12: do not use ssl.wrap_socket\n\nThe ssl.wrap_socket method has been removed in 3.12.\nSSLContext.wrap_socket should now be used.\n\nChange-Id: I6119e054289eac263ff5448d7d118209f98678d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64086c30915349b1b6440aa2de36f753bbb0741c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64086c30915349b1b6440aa2de36f753bbb0741c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
