)]}'
{"id":"openstack%2Fswift~923315","triplet_id":"openstack%2Fswift~master~I022404cd90d9755a09c20619c3a72588d3367467","project":"openstack/swift","branch":"master","topic":"cooperative_token","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-07-02 19:30:04.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"hashtags":[],"change_id":"I022404cd90d9755a09c20619c3a72588d3367467","subject":"common: move memcached exceptions to the base file.","status":"MERGED","created":"2024-07-02 15:42:57.000000000","updated":"2024-07-03 05:52:42.000000000","submitted":"2024-07-03 05:51:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923315-cooperative_token","meta_rev_id":"e8077280b527479de486b510230615c54e2efb9a","_number":923315,"virtual_id_number":923315,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-03 05:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-02 16:03:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-02 16:03:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 16:03:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-07-02 17:16:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 19:17:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b22a105633a5cc6d10e5a23e1053106541d13f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-07-02 15:42:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a165db99e8a2298b7da071c4f242093fb265d29","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-02 16:03:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef90a4655b2f701f80d2fa9b7ef0bdfba6721e33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 17:16:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/818b633f6329483296e8d8dba9e9722f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e185fce44334db5b9a599a435dbdefd : TIMED_OUT in 1h 30m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db56de268a1346c390c97698b26b9cb9 : SUCCESS in 1h 20m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/63cd0f04aebf45faaa2060765898c47e : TIMED_OUT in 1h 31m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd836e145528466f85baf26c4cdf206a : TIMED_OUT in 1h 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"89d199a0b770942c18142e21fa651441f6d9a785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:17:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56f9165851ac4c16b9ebe833eae41c23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21c633009e24b1ea232c2f36ed802a9 : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/8498d39d83c84325b2ec3f937efe0083 : SUCCESS in 57m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e52f562582342c7b3134682be303ee4 : SUCCESS in 50m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86fa677256094fe8b3cd86a06fbb922e : FAILURE in 54m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66ecd9219f28419291c567acfc624306 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cff242c62f57431080fbf8f2abcd2577 : SUCCESS in 22m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d74897231a745e693493dc8a4f31469 : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1fc500e669e49949dd1b14d08b6e2a5 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c086d9e55fa942f0a636dcdd61cb5aa9 : SUCCESS in 18m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29a4b52217714d9eba0ad96fb3ec3d55 : SUCCESS in 15m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2e7831a6b5204601bbedc821beb80970 : SUCCESS in 15m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/96a2e6b189014d7ebaa4fcfdd26ba314 : SUCCESS in 18m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/212bfe36b1e642dcb70b869db61ca1ae : SUCCESS in 23m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0274a310b8c74a92b19f41f5906d35fc : SUCCESS in 19m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8fd6c2fb8aea49078ea1130390d01e22 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5990d10541b84da993fb7cbe3cb41d24 : SUCCESS in 20m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d5967ff01a644df79b84228251988dec : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/07b8479046c749b382c490156c51d65c : SUCCESS in 17m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/20d4b27d132d40e8adaca08942f61314 : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6efd88f554f94278a64e2c4154e7d6ac : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3b3cfce52f2342b3ae107c132d2bb8ea : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d7052fe787da46d4be259728328428f9 : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c4ac2207bdc4a3393cdb0bcaf55c1d1 : TIMED_OUT in 2h 01m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c191807fbe044f397c540cf1b2a45cc : SUCCESS in 52m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/017a1a48086a467daef94977f6276dbd : SUCCESS in 1h 00m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fec92672850942f9bad63b89ef6b0c18 : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce1ae8bcb01445f894bae1191d26765b : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30fd3cd8ffd64ed08ccaed509438f00d : SUCCESS in 27m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22ef19a5e7e7491fb2e0df928e3cc7e1 : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a1956797697acb1aa0a52e7a0c60d9e8114d53d8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-07-02 19:30:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6786335d7fa35fd4f72c0728c97be82d7fc6c973","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:34:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78947db0889045bba8717ffc990975fb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3485e13e5d494c6fa52a2c5c2ecf3779 : FAILURE in 3m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fe12ee7c8104788903b43d6451ef5ed : FAILURE in 3m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ce69d1eda0b47de81361960e6504f06 : FAILURE in 2m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4540c73d2104c01895ad5d54a1436e3 : FAILURE in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4e5e213f7863b15a387e8e662a87e60345c74b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 20:37:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8f48d30c2a24c9585d53487364a8829\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb1f2ed7e4054b69942682241c609994 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/208355c31f9544c8a01462b1436ff1f2 : SUCCESS in 1h 06m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4dc23134e7df40b69c929c5bcbd2667e : SUCCESS in 55m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/982d43a470fe40aaa5b843be53c27f29 : SUCCESS in 59m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ba9a63c5a9c474d8b81f5cdfb2129e5 : SUCCESS in 24m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b25b055ff0c945b1bb04dd37baa8644c : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ebd2c0230674e339d024a2be6a8fd1a : SUCCESS in 20m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a32e640428ed4ae7999c7f6a802cb55d : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bada10d641464e5aa9501ed014aa7373 : SUCCESS in 18m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eb27d779162b401ea0dde644c588164a : SUCCESS in 14m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf6c19064e041e59ce869d936f5d9fd : SUCCESS in 19m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9e81e8fe6771445ebaa97397144cc45f : SUCCESS in 17m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c928d3d559b44f60b821d11e15e9ba4d : SUCCESS in 23m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/04c90ab8a07d4605930903c67c8f8750 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6bd352670dbe4add9616bf9a5b2641f5 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d31b590427e446ccbb086dc04c777b4c : SUCCESS in 18m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bfe97282295d40eaae5298666a2ced4e : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fbeca272408c4824872266c495f48f8a : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/630b4a7ef7c44304998dddb2779765b7 : SUCCESS in 18m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee077c38e21543058dafc8dd2942e084 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4cf66d19b07346899a13010d0c123bfd : SUCCESS in 14m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/73b2780fa70b4352b4e684dbbd29a72c : SUCCESS in 6m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bb1bbae4f2e4de6b2f969a3d1ef3804 : SUCCESS in 54m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c164ba4bd51494ea2f52fa3dc7cf21f : SUCCESS in 51m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eca36137c63044489475c2d19a4945fd : SUCCESS in 57m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d98671e78b341b5afb338a2a16f7957 : SUCCESS in 22m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd736b6cd89647df83300742453bea43 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87420368a66b404b832b2a2a49460c2e : SUCCESS in 17m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/712fdc73826c4afa86d0aa42b59d527e : SUCCESS in 48m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ab442952d9a1a9d10a144912906eed00527cef15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 20:37:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c693aa5187372b958caae75861ea07dd11ae270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 05:51:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce6717bc5c046f9bbaa599b1c1f9326\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb9c0d7d6254feba894248c82d9028c : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/8dd44d75524a49918d538100305e0eee : SUCCESS in 1h 09m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1eb65b5f9ce644f69cc28ec537ea2372 : SUCCESS in 50m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/470d6d4259c24908967c49a397b9a5ab : SUCCESS in 1h 01m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fe1dfbd3b086445895d68045636322da : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ecfae7a4df440d49c031179e4994f36 : SUCCESS in 17m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35777e39fb4146878e12a78586ece99d : SUCCESS in 18m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/509f0c02745d4b4bbe25c67d4ddc5c29 : SUCCESS in 17m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bcb082e302a142b0846e303db4efaf66 : SUCCESS in 15m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ee4dd22d62af4525b3f61b639e3d88f4 : SUCCESS in 22m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/11ebf4ca0751485eb126d92c0eca0ebf : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0eda2065a22c4f0b977d9035386a2c5a : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9a9bebbdbd4b44069ddacbd33ffd78dd : SUCCESS in 19m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c743f19bbf984be3aa8b5c8efe750577 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/78b1855e633d4748afbe5353574333f0 : SUCCESS in 18m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/46695361125843f9a2ee771f2427a637 : SUCCESS in 18m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/34b748b36fd14c3d996211b924e3ec5f : SUCCESS in 7m 06s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e604eedb10cb4dddb9b8b3feb7e5e815 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/33730dc26c3b4bc7853ae5a3e1006594 : SUCCESS in 1h 01m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17d571ae58e947f9ba8c82503b2efe33 : SUCCESS in 47m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f609e58e0d14bfbb7d94d67ba4aa421 : SUCCESS in 47m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b7f3f04180145dfaba03edb2b246a97 : SUCCESS in 19m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba05c590e1294b7aba4ccf36a4893c56 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc68da86d56741a78b0c3684aa6a18fc : SUCCESS in 18m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/561b67971db04f7bb0495b92c074241c : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":1},{"id":"87b498f3aa63af86ff0f04c85d1abbcf08e5ff43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 05:51:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e8077280b527479de486b510230615c54e2efb9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 05:52:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f347330b1814e4983f4d96bc8c0d858\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a46fcd4a3a643c0b2af1b81b9261d58 : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fb113b2c7fc941b6af81483f8b0095d9 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fd36e395dbe05fb77a4e76bc6ea4c83597d86ff","revisions":{"0fd36e395dbe05fb77a4e76bc6ea4c83597d86ff":{"kind":"REWORK","_number":1,"created":"2024-07-02 15:42:57.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/15/923315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/923315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/923315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/923315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/923315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/923315/1"}}},"commit":{"parents":[{"commit":"789d0c44d9744554e584e9a56b99e6139326e936","subject":"Merge \"Python 3.12: do not use ssl.wrap_socket\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/789d0c44d9744554e584e9a56b99e6139326e936"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-07-02 15:37:12.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-07-02 15:37:12.000000000","tz":-420},"subject":"common: move memcached exceptions to the base file.","message":"common: move memcached exceptions to the base file.\n\nChange-Id: I022404cd90d9755a09c20619c3a72588d3367467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fd36e395dbe05fb77a4e76bc6ea4c83597d86ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fd36e395dbe05fb77a4e76bc6ea4c83597d86ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
