)]}'
{"id":"openstack%2Fswift~924108","triplet_id":"openstack%2Fswift~master~I5d756147e68722b06a234e91be7c0491916d77ab","project":"openstack/swift","branch":"master","topic":"followup-db-state","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-07-13 03:37:51.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5d756147e68722b06a234e91be7c0491916d77ab","subject":"trivial: refactor _get_update_target","status":"NEW","created":"2024-07-13 02:19:57.000000000","updated":"2024-07-14 17:27:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fd4d3eaef7a2be7520b1d6bb5f1bf6958a6513c9","_number":924108,"virtual_id_number":924108,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-14 17:27:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-13 03:37:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-13 03:37:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a4e85860f80043cadab7e81eac238089c329525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-13 02:19:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dba499fab1c322b15227b8a743c3f3fcd13963a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 03:37:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7d423a8209f4680bdb417042ff64d01\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ec7223b6c8e4f978631f588486e18c1 : SUCCESS in 1h 03m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0b022a4d41e41dd942de65fea3371e3 : SUCCESS in 1h 15m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/684a064affee4b7290277b3f8cace83e : SUCCESS in 25m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c535fbc1874425dabed941e64bac0fc : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bae8744f5b516f4bd15123a3c5c48e9e1451ec2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 03:37:51.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad3e0d80696441cc8e7d5c913bd43980\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2aad78e4bdf4abda2ecff8fd91c4366 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/e4146ef85b3f40fcbfc7659a1f58fabe : SUCCESS in 1h 02m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7765cdcd5853471aa3809b1af89a9eb2 : SUCCESS in 53m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29b88289935a496ca8076bdadd20231b : SUCCESS in 54m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e962c7e4cfd54eb0b6c2e106dd1ad022 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77cfb84a1d964ae7be0d3525496c7207 : SUCCESS in 21m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44095a0867d048208861f199cefd9e68 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32a5efc314b1462fa9d0ab9b6e375b20 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b0babb53f6b64357be858abe5cb1f11a : SUCCESS in 18m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a28f3d0a940f4304befbae77e7ceb07e : SUCCESS in 18m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/be1f13e1890b4e909b81cc6bd3d4292c : SUCCESS in 20m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f468be747504472296ad987ae18d7dcd : SUCCESS in 19m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/61e67aaa5f9046e198992cc96d32b9c4 : SUCCESS in 23m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fcefec936df44b49aefeb51676d465d4 : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0bf5de68a4794add9e16c15beecd44d8 : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/917186ebdb134b209bedd1df4d694923 : SUCCESS in 19m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/237248cbdec542c69cc59cf75c56ac0d : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b252caddb0a14e3cb3e4ea8333d6a949 : SUCCESS in 20m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eb478ff739ba4596b343d74f769f9203 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bfbc7b57fe354e5297f9c44be6a44c95 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1106f707d8143bf8868561dd2ed03cb : SUCCESS in 24m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e9b01e23cdb145bfac8b6607781e58d8 : SUCCESS in 9m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed052ed38ff24aa9a11913096100a3e7 : SUCCESS in 1h 05m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6f04fc9c59446719e222bdfe039ca3c : SUCCESS in 48m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74ff71bd4bc149329a7764b6c8131021 : SUCCESS in 49m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f959f555a1eb40bd8fec8f83d2f0e920 : SUCCESS in 22m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec36170c4e7e48558d6422ccc2e9dbe9 : FAILURE in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c93843c2b7044e12b6fb3037fdf46436 : SUCCESS in 28m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92114966988e4c9bac8977e5a2cc1a81 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"da6a736c1a2be41677b859f403901867edfe5a62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-13 19:44:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fc39d7f0e3ddd89d142267e0fa500af4484ae83b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 20:12:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ff61d92de14ad7bf40e4cba310c00b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f8740f2c1094d74a80b074003d3ac57 : SUCCESS in 17m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a481ae868054698ad047cef8cc459dd : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ce9f773489b4500834f1be88747d34d : SUCCESS in 20m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce58ba6be4414c20b011c2004938be50 : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a738ceb71ee030496c89a92de328157839ec42de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 20:47:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4da046a640948faab8ea02cf345e13c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e14eb2d52a8749e1a24965437cae0877 : SUCCESS in 6m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/2e071264ef2d4ae98df1ca1a69d8240d : SUCCESS in 40m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d9a1802c47c04dd09b0aba25df77186f : SUCCESS in 50m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d153d11ea134161af88f2f2bc929981 : SUCCESS in 1h 00m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/235150ceea2043eb8670e6c138aa0f39 : SUCCESS in 25m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fd948e16b8948bd9331f8efa9302537 : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d337ad16bcc4b24b8767ff6a3a15911 : SUCCESS in 19m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d11c88df0824d09980719e514917ade : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ff7b464c96f47d8ad44659c9469f551 : FAILURE in 20m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c1e0562166c4433390c2e795f5d23c51 : SUCCESS in 19m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3013be09216d4ac298d9fb70120abdc6 : SUCCESS in 21m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/60413dbac4a445178c31928ce4a5fb2a : SUCCESS in 14m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/606ccef151c944ba937872ee9c356f0e : SUCCESS in 27m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/47dde275cacf423a9fff2d070166a158 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b727fb64a28a452da9301a56bb400712 : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b28058ae3b454b2c833e653b2202bfe3 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f50bc621c4d4478c8409b06bce0e9c53 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b75bf708db9540f8bfc7e45d9118e5d1 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4eff367293eb452794ba35a518cdc401 : SUCCESS in 22m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92bf90c1c39748168a3edb69c086a91b : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c713ebfa9a274ac6b187792b9684e4ff : SUCCESS in 12m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd98360dbb634262b3d86d7a6a79ad7f : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2cad308aa9b45b69c085dba0aded2d8 : SUCCESS in 55m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf885ee448384ac8bc4f0af626fdcef9 : SUCCESS in 53m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/df804d70651b457f8e0205992494912d : SUCCESS in 32m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f34e36945cdb41f88104118cd2b38d5e : FAILURE in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592f34946e914421860fddd9144ea13f : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e81e1c5ec52246e98a7efa2db1b1e568 : SUCCESS in 17m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01e74187e17c4055a9cb3828e265a5b3 : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":2},{"id":"04fbfbdb7a84d81f834682d8c2af6a549d62d1b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-14 16:02:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4304979452d2529b9e371c128a16b948c0d7569d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-14 16:24:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3ad5a245514c3987cd7e1efaa0eaa7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fa00be05c7d47ad83a69e52ed710dfe : SUCCESS in 20m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b70aff0c6574a7393f973b1cfc18089 : SUCCESS in 18m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a1080c1ee7f4fd9bd80d998490cac70 : SUCCESS in 21m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab1bc526f517417b9424c5fc585e57e8 : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":3},{"id":"fd4d3eaef7a2be7520b1d6bb5f1bf6958a6513c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-14 17:27:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78034a47ae51481d80d1a863a8f44abc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26c77c5e23649f597e1df2ca1d80e8e : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/50c231fedb6044649e336de0053d28db : SUCCESS in 43m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bce40f538d33415abdccfea9c93adf58 : SUCCESS in 33m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/152282295acc47b7a6f078ac08b14f6d : SUCCESS in 40m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bd39d929a69446fa418547bff3e43f7 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/916b29259f6a42a0bbc8bfe8b221f1f9 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c258d50a5bc40c680b89f5471f1aa05 : SUCCESS in 17m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85a9da7cd19a4b3a864e5ea0ec855ea4 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/888309f1d18244e98b62e640b165a84e : SUCCESS in 20m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb846e56cc4a4be99de85f09a81b52e6 : SUCCESS in 15m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/53ecff2b85874ba1a1b442e41df768a7 : SUCCESS in 15m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b938749132be4c83a127c486d9a537af : SUCCESS in 26m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0fcfca965bfd4901b58bb26ee048491d : SUCCESS in 17m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5667dd7d92a40eeb3a4af6f7915570e : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/618bfe5145cb40358b9863c8cec37ba3 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ca8130c1c39c41ac9f2c7f2d67ae18b2 : SUCCESS in 20m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f4cc42fdb45546c3af0f336ef8bc8811 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/475f787bb49c47f3a9082e9a24ca3936 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e4af588822664476add91a46a6d2bf80 : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/426d8f263ead46a3a0e7bd5dfaf59200 : SUCCESS in 5m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4783a62891e046c5b17d918446008b90 : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/641efa0accb54760be665030fc8b3ea4 : SUCCESS in 6m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b5121101f2c4de8a5aa7fda8dd74f36 : SUCCESS in 55m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ba90e9ad04e4303ad5f20bea52a6f6d : SUCCESS in 52m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad06a8973281475e8f504e80594e2dd9 : SUCCESS in 38m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aa85e9f49d04ed08971cbe6fdddf8cf : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f98f7631940a4859819d67d63a6da618 : SUCCESS in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52490b776fd34b1a8f135f123852eb3c : SUCCESS in 14m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8894b5e917db444eb369ab78cdfe261e : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9a4da903bf4702080240a52f98cfc99d69b956f8","revisions":{"a191800bcf620e1056200b8c9bb8184ca8601bf2":{"kind":"REWORK","_number":1,"created":"2024-07-13 02:19:57.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/08/924108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/924108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/924108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/924108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/924108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/924108/1"}}},"commit":{"parents":[{"commit":"e8affa7db5817e50cd82130a8f4c250725b80884","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8affa7db5817e50cd82130a8f4c250725b80884"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-13 01:27:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-13 02:19:33.000000000","tz":-300},"subject":"WIP: add missing tests for _get_update_target","message":"WIP: add missing tests for _get_update_target\n\n... and reword some other tests from related change that added db_state\nto async pendings.\n\nRelated-Change-Id: Ie665e5c656c7fb27b45ee7427fe4b07ad466e3e2\nChange-Id: I5d756147e68722b06a234e91be7c0491916d77ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a191800bcf620e1056200b8c9bb8184ca8601bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a191800bcf620e1056200b8c9bb8184ca8601bf2"}]},"branch":"refs/heads/master"},"092a8f01079cefbcec0f03aa57fd28278d659f2d":{"kind":"REWORK","_number":2,"created":"2024-07-13 19:44:58.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/08/924108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/924108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/924108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/924108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/924108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/924108/2"}}},"commit":{"parents":[{"commit":"e8affa7db5817e50cd82130a8f4c250725b80884","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8affa7db5817e50cd82130a8f4c250725b80884"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-13 01:27:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-13 19:44:26.000000000","tz":-300},"subject":"trivial: refactor _get_update_target","message":"trivial: refactor _get_update_target\n\n... and reword some other tests from related change that added db_state\nto async pendings.\n\nDrive-By: refactor some test infra\n\nRelated-Change-Id: Ie665e5c656c7fb27b45ee7427fe4b07ad466e3e2\nChange-Id: I5d756147e68722b06a234e91be7c0491916d77ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092a8f01079cefbcec0f03aa57fd28278d659f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092a8f01079cefbcec0f03aa57fd28278d659f2d"}]},"branch":"refs/heads/master"},"9a4da903bf4702080240a52f98cfc99d69b956f8":{"kind":"REWORK","_number":3,"created":"2024-07-14 16:02:07.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/08/924108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/924108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/924108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/924108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/924108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/924108/3"}}},"commit":{"parents":[{"commit":"e8affa7db5817e50cd82130a8f4c250725b80884","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8affa7db5817e50cd82130a8f4c250725b80884"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-13 01:27:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-14 16:01:58.000000000","tz":-300},"subject":"trivial: refactor _get_update_target","message":"trivial: refactor _get_update_target\n\n... and reword some other tests from related change that added db_state\nto async pendings.\n\nDrive-By: refactor some test infra\n\nRelated-Change-Id: Ie665e5c656c7fb27b45ee7427fe4b07ad466e3e2\nChange-Id: I5d756147e68722b06a234e91be7c0491916d77ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a4da903bf4702080240a52f98cfc99d69b956f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a4da903bf4702080240a52f98cfc99d69b956f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
