)]}'
{"id":"openstack%2Fswift~924183","triplet_id":"openstack%2Fswift~master~Ibd7781107884e472ce01cccc87147895b30aec85","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-03-25 10:53:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibd7781107884e472ce01cccc87147895b30aec85","subject":"sq: fix a bug","status":"ABANDONED","created":"2024-07-15 21:43:22.000000000","updated":"2025-03-25 10:53:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2bb38913a129618e5c75839cca720ec892513416","_number":924183,"virtual_id_number":924183,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-15 23:16:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-15 22:06:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-15 23:16:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c5dd00ec85f935122c5f24678728162983573b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-15 21:43:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84f4ed3bbaa61a87895029ad54e4518afc3dd275","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 22:06:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e47291241dad405a97c2740198111c75\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b9bcb385d1345e4ba3298205514ee47 : FAILURE in 20m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1d74561f9c646adacf431d2b809b9a7 : FAILURE in 17m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e54f525cf0a443929408e002ba8b5b0d : SUCCESS in 21m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3ec9ffcaf3541a0aab381dbc38281d3 : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b03f8fabea9d4eeb94584393d9b0b83ae91ff980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 23:16:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ee9790759034dbc85ac43a578b62dfb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f491327efb145fb816947181e881f02 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/693b339f87ba414485954f1fcaea9366 : SUCCESS in 1h 03m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7cd18d8c06540da9d0330f1e3f0fbff : SUCCESS in 40m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e220a8df86744226a432e0df3ddfd15a : SUCCESS in 52m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c246cb9575424dc3aa58b7274a8aa8c9 : SUCCESS in 14m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52b9c69aeba54e51b1ac0460204c8a26 : SUCCESS in 27m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f3bf4c63ac44b679877ca2411da4480 : SUCCESS in 21m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f88f94282ea44529673d1f33eeca9c0 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7090366998f4bd7a992e22042d091d3 : FAILURE in 19m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b3ba51dee9349b390add6bee17210c4 : FAILURE in 15m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a53a197b66c849ed9b0b5d67f81e4e7e : FAILURE in 26m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/96f64f838ec54e0a88921cfc46b94fda : FAILURE in 22m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a1497c349ed249c79516fc1e15ccc0e8 : FAILURE in 24m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/df459ef417204798a148d3f4f78c9372 : SUCCESS in 26m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/33bb62a312284224a92baf898e1b0333 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d8e38478444a45f2a1857c1ac45a1696 : SUCCESS in 20m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d9a428104570439b968dc6836ececd46 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e21bc06798f343e2b31cfb3c873ed73d : SUCCESS in 17m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc7d801bf4e14aaeaa9f241f03619be3 : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d8bf9f2210a458290430c84b2d2d916 : SUCCESS in 9m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f43cbf3d55341b5b5230408871eaeac : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/04bee31dbf35458bb78d84f16bc7336d : SUCCESS in 10m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/17b71d888fd84d77904a94c222a30177 : SUCCESS in 58m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/797b7cf5621846e3b609c851be2e9f09 : SUCCESS in 47m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30b46bcc8cd94f6ab291cb7c48503e93 : SUCCESS in 50m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bbf847814d74d75babd51c38018cab9 : FAILURE in 27m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7f52b740234f20a08014b1ec4885e9 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d89b4c0db1844c6b6c96c6299265c2c : SUCCESS in 26m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a367db67bd746b5a5619bafaa9dad18 : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2bb38913a129618e5c75839cca720ec892513416","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-25 10:53:53.000000000","message":"Abandoned\n\nAFAICT this is all covered in https://review.opendev.org/c/openstack/swift/+/909882/48","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"332d90de948df085b675d422d70cc176a4f094fb","revisions":{"332d90de948df085b675d422d70cc176a4f094fb":{"kind":"REWORK","_number":1,"created":"2024-07-15 21:43:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/83/924183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/924183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/924183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/924183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/924183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/924183/1"}}},"commit":{"parents":[{"commit":"137ce227de8bd03e2493e8e4689975fade1c993d","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/137ce227de8bd03e2493e8e4689975fade1c993d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-15 21:40:24.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-15 21:40:27.000000000","tz":-300},"subject":"sq: fix a bug","message":"sq: fix a bug\n\nand write a failing test\n\nChange-Id: Ibd7781107884e472ce01cccc87147895b30aec85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/332d90de948df085b675d422d70cc176a4f094fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/332d90de948df085b675d422d70cc176a4f094fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
