)]}'
{"id":"openstack%2Fswift~924214","triplet_id":"openstack%2Fswift~master~Id18176d3790fd187e304f0e33e3f74a94dc5305c","project":"openstack/swift","branch":"master","topic":"log_prefix","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-07-22 16:16:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2024-07-22 18:19:31.000000000","reason":"Change was submitted"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2024-07-17 15:33:49.000000000","reason":"\u003cGERRIT_ACCOUNT_36606\u003e replied on the change","reason_account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}}},"hashtags":[],"change_id":"Id18176d3790fd187e304f0e33e3f74a94dc5305c","subject":"proxy_logging config: unit tests and doc pointers","status":"MERGED","created":"2024-07-16 11:13:22.000000000","updated":"2024-07-22 18:21:06.000000000","submitted":"2024-07-22 18:19:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924214-log_prefix","meta_rev_id":"429f2ec21d8a5869b6a927a94838d89e912d6267","_number":924214,"virtual_id_number":924214,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-22 18:19:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-07-22 16:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-16 19:14:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":1,"date":"2024-07-16 22:01:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-07-22 16:16:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-16 12:25:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-16 19:14:14.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"},{"updated":"2024-07-16 22:01:23.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"CC"},{"updated":"2024-07-16 22:01:34.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"REVIEWER"}],"messages":[{"id":"3b8d267ec27f56e0ffbba7196ff8beeea975e59b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-16 11:13:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"113b9c6004849348762062e42997f9b2911d47d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:25:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f235203af8f44c8294149736c5140354\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69cf981d45304c9d8c12bd8e5f537b44 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/6958c9eeb0134422b7b091108ac7aa2d : SUCCESS in 52m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/daea09d6840544ceaf554cda760e8771 : SUCCESS in 58m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d32590e3cd7b485594e0cc9197ff386a : SUCCESS in 1h 00m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9022dd5a54a745e5ad546d63a2cecd82 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b33fd6f12d84f2d87b6b302572608a0 : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9dc29975dab540749ef96e9aa6bb6fdf : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1465dc5afd5b4a48bfc1a7ce86f7bd9e : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c84c4f1484654dedaa408d7db3f93f78 : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8853f133417c4c42b9636bb13cd95c1c : SUCCESS in 15m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5126d976aa8640b1a6746b62540a917e : SUCCESS in 19m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bf7a08e5e46149a9a5d9759f4850692e : SUCCESS in 16m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6da4f495219d49c3ab100e9c43949361 : SUCCESS in 22m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4758c638d6ff4d74904e19cd2948a052 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f99e33b115dd4ba09f701ce8d5828d38 : SUCCESS in 19m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd0431569a5048feb187d7c4288ea4f3 : SUCCESS in 17m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c6e0eebd9d234d988bfe5b493d3746ac : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ff8610f8dac42178259c4de06ec7074 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/31e2820325fc4089a2f5f5abf3274ca2 : SUCCESS in 18m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2bd676f376924765912caabc75b41c68 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/71faa1f75cb5478b859e6361ecf2b201 : SUCCESS in 14m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13cba15db7ca4cf3b251684a5c243a4e : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/220feed6b4e94c03bf46ba25b15652f4 : SUCCESS in 52m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c3d05f513ec4d90a2be38b24a462688 : SUCCESS in 38m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee2c23c9c6aa40e3a8e94e2231719d7b : SUCCESS in 39m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab5937cc906a4dc5afdf29b1d0a6884d : SUCCESS in 22m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa0929c541d444888171a273dda0be8 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f70a0041237843cdb7a5fa7520e2bb7c : SUCCESS in 17m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/175883bd7de643b4b94471c87256b9e4 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"15a510daf5228874a467646856181be7725ce9a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:46:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b800b0e897a4eef9edadabc0db5b674\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0080dc3e6daf43308a7cd92073f9be50 : SUCCESS in 1h 10m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9db6085bf7f4506b78d4f1cce57dc14 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9404f53a1d514a4db5a832e6173ca10e : SUCCESS in 1h 11m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb45c2054bea4c84a0eaf2de7f5453a3 : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"01c7159b50d6373300d3e95edd71ef89174426df","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-07-16 19:14:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ca8e526dd32bc40638611c669ca27462f5cc6c81","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-16 22:01:23.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c62ac47d49cffbaa2a3c9c850857935df05768d2","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-16 22:01:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a0968900f4258ccc868ec25606c8e89b0186754d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-17 11:34:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1192bbd878ee8b951540aa4d24d286efe6924e04","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-17 15:33:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a9ff3e3041302a20289f51a800f0ec433d99df3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-22 16:16:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed312a990d7e1b23d04a9f95ecda266e1bc070a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:16:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"518cd64691546d34b44ac2ffa95e0b1aeff5e698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:19:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13e07860dc9c4b1cb292ceed5aeb4530\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4b8aed425954339a62da8532cbbcddc : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/bb0a8706b3c5478c89139ada756197fb : SUCCESS in 58m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/972ed5e2ff7341299094039a2800b4ad : SUCCESS in 49m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a002c54214a5444fa77ab8fd49f3612b : SUCCESS in 1h 00m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b78bde7cb8ab42dd8eaabd5229af06e7 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c4afa1c2ebc43899bd54807bfd85d2f : SUCCESS in 19m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56b2cd55035541da81df6f98d8089357 : SUCCESS in 18m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/30ef627a1d034f619360ddb29a3717ba : SUCCESS in 18m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4ab4dd8764994bb39fd721ef0607d2c6 : SUCCESS in 27m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c987b82d0a354986843fc45bf5d32f37 : SUCCESS in 31m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6cca2e3c226649ffac7e9603802adc90 : SUCCESS in 22m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/199717e4e3614ed9b1c1d1220feb9803 : SUCCESS in 23m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/70c4041126824c0a88ffa0aa0081478c : SUCCESS in 17m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e379b2f776204803806400c6dee6c31f : SUCCESS in 22m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ca1c2b42667e418e9ac4998b640948a7 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ac837750a1c54f2aad8756d463fc8a95 : SUCCESS in 18m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f1cb3bed2c694343a2ed87d8c4124595 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fc1e39a2f3ae4090afbcc647a0253891 : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/df7222475d1046bbbf8601b5b240e869 : SUCCESS in 1h 03m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0a3269a563544079c8247eea5c5575a : SUCCESS in 44m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c8ac2c4d75e349b6b7d26fc775ddc400 : SUCCESS in 46m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6395378637d64afcbf97d637e5df2672 : SUCCESS in 22m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5bae2e89d242d885f357bd1e4a710d : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5aa89df585bc4a339fc6b4284d54f2b1 : SUCCESS in 17m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb614f20c5b5459e8a971cd1141c697c : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":1},{"id":"980133357e1da9305477e49d486ac7a6138bbb0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:19:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"429f2ec21d8a5869b6a927a94838d89e912d6267","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:21:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e16121db7ec144c6890aed55fbeae595\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c4766e02554d5c8bc810a2f9404e72 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/da52f319390c46bd81b7e440b49cf4a5 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d555755423c179d23ef02ab2e0b856fd2b04fa71","revisions":{"d555755423c179d23ef02ab2e0b856fd2b04fa71":{"kind":"REWORK","_number":1,"created":"2024-07-16 11:13:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/14/924214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/924214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/924214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/924214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/924214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/924214/1"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-16 09:42:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-16 10:33:58.000000000","tz":60},"subject":"proxy_logging config: unit tests and doc pointers","message":"proxy_logging config: unit tests and doc pointers\n\nAdd unit tests to verify the precedence of access_log_ and log_\nprefixes to options.\n\nAdd pointers from proxy_logging sections in other sample config files\nto the proxy-server.conf-sample file.\n\nChange-Id: Id18176d3790fd187e304f0e33e3f74a94dc5305c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d555755423c179d23ef02ab2e0b856fd2b04fa71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d555755423c179d23ef02ab2e0b856fd2b04fa71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
