)]}'
{"id":"openstack%2Fswift~924304","triplet_id":"openstack%2Fswift~master~Ie240f3aa01ce5823b18bdf28fea88e414dab5b54","project":"openstack/swift","branch":"master","topic":"db_state_in_async","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-07-17 10:14:04.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie240f3aa01ce5823b18bdf28fea88e414dab5b54","subject":"sq: wip: first attempt at an async cleanup","status":"NEW","created":"2024-07-17 08:00:03.000000000","updated":"2024-08-15 08:49:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"1902e261dd9769c3e4ddd54ab45a7ff4d748559d","_number":924304,"virtual_id_number":924304,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-15 08:49:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-17 09:32:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-17 10:14:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f5abcd2c566d07ab3af4d5666adb41a4f31f95e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-17 08:00:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e18f95f1bdaae5563e8393195e59aec7fa2cdd5d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-17 08:04:44.000000000","message":"Patch Set 1: Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cebe20c01cf21f9f577b2bc6b93b1547173b1084","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 09:32:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7031c6dc0c6d4b0a8fd09cb8dd8ba18d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3521a4fe554241dca1413bb0c2be7fe6 : FAILURE in 51m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0a072bd441a466cafab3a98b671f894 : FAILURE in 1h 15m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/70a1e311b5634f52a48506ec76de9382 : TIMED_OUT in 1h 30m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/be0e5c1e4d324f91bf51505316df55a3 : TIMED_OUT in 1h 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"feb347b8b5a9e1e9658b8c306e951d3a4da90fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 10:14:04.000000000","message":"Patch Set 1: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbcb2581d5a0466f83c47af209800f41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb4c248e74e4f5882eaa59e7c82fba7 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/6172b02473834a8094a2260c724cc35a : FAILURE in 1h 01m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8ccd6e07bcfc494c8c4835ccaea2b497 : FAILURE in 27m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e0e781a04754a61a58989065ae5731a : POST_FAILURE in 28m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b96de4aab284f8d998076a4e29be54e : TIMED_OUT in 1h 30m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad472ba98ee945d9a7cf924a2cb035ea : TIMED_OUT in 1h 31m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df3780ec255c490bb89a7531c1be5cb5 : TIMED_OUT in 1h 30m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb5801d3bbcb40ae8cb1774e6bb7e757 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d2613f1ab134ff6baf78bbbb1f488d5 : FAILURE in 26m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6297c4a94524e75be8fe32f37bea9e3 : FAILURE in 22m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/982855554985433e9fb5cff9d676e6d4 : FAILURE in 25m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1416afe0931640b6bacd2d775bd3229b : FAILURE in 33m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1c2f7b0dc29a4910935ca9898a94d67c : FAILURE in 29m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c942460629e54ed4bb8bc17c638951f8 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7ca875b2d8e41d69f0e5ed29d6776fa : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/398a573f0a35451fbbe73ecfa89719b4 : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ff8bbd77a0348e9be8ea2ffae5d664f : TIMED_OUT in 1h 30m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b02894c951f24efc8529dbe1c6b40605 : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b6ea863b50354af1a960dbb4d6901328 : TIMED_OUT in 1h 31m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/626f7836d14b4c289aa944ab3a30d8fc : FAILURE in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8a5fefc980744b5a765df06377d350e : FAILURE in 1h 04m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1048fde5578c49bca2aba94b699eadef : FAILURE in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc46f3cdc0544e7a9862ae8571945c16 : FAILURE in 1h 37m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e230d58635a447f69effbb989e748dff : TIMED_OUT in 1h 31m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/73544294a7a04a11b4c35720ff9d4d66 : TIMED_OUT in 1h 33m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fac6a1b62b64f2a91699c3fed43c26b : FAILURE in 23m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d57f163fc3e43be83dc54e84b232ed2 : FAILURE in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24a05c3d0a434c15b432db23c55530d1 : TIMED_OUT in 1h 31m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5eef9f2f7250420ba415da0d369739db : FAILURE in 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"213c8ee9e8f96e4e9d49342bc3e72ab10a3adf40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-15 07:03:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"abb24cf493fc02b62022ed8bd626d1af44f26cae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 08:37:14.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea0f401d379346c5856a3d4d7f9f8554\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07afe8b3ee6844418ecfe7067c59e424 : TIMED_OUT in 1h 30m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7458db596e2a474ebb4212832b260e14 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/052c4ea918c041998d6bebb753d153d2 : TIMED_OUT in 1h 30m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6693ad613b394b718eea6e5c55041216 : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1902e261dd9769c3e4ddd54ab45a7ff4d748559d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 08:49:50.000000000","message":"Patch Set 2: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4dd37e8a5774a5ca89f73f3a6d35c20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7af5d3a3064689b8d26007474c4d17 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/e5b2c289b64e43e4b77bef6321581ac0 : FAILURE in 57m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/be96bf3d968844d386d4b91c949c65e6 : FAILURE in 15m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb2efccb4b04456e85fcd91bbdbb7fc1 : POST_FAILURE in 14m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e8c56a94a2b4866afdc6d0e85e96197 : TIMED_OUT in 1h 31m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8e0a3dae43d4cffb971c0c73d823ffa : TIMED_OUT in 1h 30m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d05fac8c7534bf580ab7551894a1877 : TIMED_OUT in 1h 31m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e7052d8954041be9e27f3bdc4431a5f : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/409d8b7903264fe296a9b2a5b5bee01a : FAILURE in 17m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f01e159a3435488da5f47fc6d16fd51d : FAILURE in 16m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/20e329253a70487d865f95d325823439 : FAILURE in 18m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5cf154b6377449c9aa1889037f080a06 : FAILURE in 18m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d63492012ec54ee0879a08542c37118b : FAILURE in 20m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aad46b5ed860409a939e9668fab0c641 : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cec923b06a43412084736d0285a9788f : TIMED_OUT in 1h 30m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e2c1e59a4c2b4e6db452fc6cf7d93529 : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4aaaeea1d8cd4c298a066239c6892b96 : TIMED_OUT in 1h 30m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f0a9f95d34244d94b87f17eb768a8115 : TIMED_OUT in 1h 30m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/40ec40cd976e49f0990c4da377a40b75 : TIMED_OUT in 1h 30m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8aec1b5b4fd4379bf8fe1e0769f806f : FAILURE in 5m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/031b60a230a04f46a15848ad049562bc : FAILURE in 1h 04m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c2aae5d4708c4fe6a8ba38dd7df8522c : POST_FAILURE in 22m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/767cb76ec1cf41058c3b8745c47bf0a0 : FAILURE in 1h 31m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ad6fd672a8049e790378fd743dcb5bf : TIMED_OUT in 1h 32m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/896d4c283abb400b9c1be1db0ca94128 : TIMED_OUT in 1h 31m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d19bdaa2a9164b58b94d7fcf1277cea3 : FAILURE in 24m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efaa42eb002f4a79b57f1809dc20c992 : FAILURE in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11214f09f90649c0be8a48f4cf438e1a : TIMED_OUT in 1h 31m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/784be6a423b04f3c926e74ef2df5b981 : FAILURE in 26m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fca0a51d12b09b56548fed73c0df17856d66129c","revisions":{"e70f9e64baf5350fcd69cd2649cda5089c9ff7ee":{"kind":"REWORK","_number":1,"created":"2024-07-17 08:00:03.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/04/924304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/924304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/924304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/924304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/924304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/924304/1"}}},"commit":{"parents":[{"commit":"14696fdb2954e2a37bda4773645655c62e75201d","subject":"obj: skip to async_pending when root is busy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14696fdb2954e2a37bda4773645655c62e75201d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-07-17 07:58:56.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-07-17 07:58:56.000000000","tz":600},"subject":"sq: wip: first attempt at an async cleanup","message":"sq: wip: first attempt at an async cleanup\n\nChange-Id: Ie240f3aa01ce5823b18bdf28fea88e414dab5b54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e70f9e64baf5350fcd69cd2649cda5089c9ff7ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e70f9e64baf5350fcd69cd2649cda5089c9ff7ee"}]},"branch":"refs/heads/master"},"fca0a51d12b09b56548fed73c0df17856d66129c":{"kind":"REWORK","_number":2,"created":"2024-08-15 07:03:25.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/04/924304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/924304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/924304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/924304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/924304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/924304/2"}}},"commit":{"parents":[{"commit":"14696fdb2954e2a37bda4773645655c62e75201d","subject":"obj: skip to async_pending when root is busy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14696fdb2954e2a37bda4773645655c62e75201d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-07-17 07:58:56.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-08-15 07:03:06.000000000","tz":600},"subject":"sq: wip: first attempt at an async cleanup","message":"sq: wip: first attempt at an async cleanup\n\nChange-Id: Ie240f3aa01ce5823b18bdf28fea88e414dab5b54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fca0a51d12b09b56548fed73c0df17856d66129c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fca0a51d12b09b56548fed73c0df17856d66129c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
