)]}'
{"id":"openstack%2Fswift~924313","triplet_id":"openstack%2Fswift~master~I5d62e14c228544af634190f16321ee97a8c4211c","project":"openstack/swift","branch":"master","topic":"p-test-non-negative-int","hashtags":[],"change_id":"I5d62e14c228544af634190f16321ee97a8c4211c","subject":"trivial: assert non_negative_int treats float as valid","status":"MERGED","created":"2024-07-17 09:25:51.000000000","updated":"2024-07-19 13:02:16.000000000","submitted":"2024-07-19 13:01:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924313-p-test-non-negative-int","meta_rev_id":"94d42f5ebc18bfb22aec7e1774a0ada140a76545","_number":924313,"virtual_id_number":924313,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-19 13:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-07-19 10:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-07-19 10:23:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-17 10:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ff4ac162bb73179273942796c738ba57d2015da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-17 09:25:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b70f29530d284b5bbfa074b8a5075dbd9ca042a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 10:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1c1b5777a9e404c9f3c890e4ce650c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa57b72662a4d26995020f5618793c7 : SUCCESS in 8m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1e998f51c8b450ebbbcd430298470f6 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/beca1ea7d896428c83a87916d8a1c578 : SUCCESS in 23m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a65e02884dd4a2aa874b029f8ef26d5 : SUCCESS in 23m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c69d565dfc75464fa31fbcd52affa4d8 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3329a6f72064f679c0eb0a10e5aff4f : SUCCESS in 23m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/71414d45f2194bf9af9b197b9df76d11 : SUCCESS in 20m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5b00565aff1c45ca845d25fad18067b5 : SUCCESS in 21m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dae3d722915b4e879307419ca8407668 : SUCCESS in 22m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4d51c7da98ed4bcfa252dbd3257f96be : SUCCESS in 40m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9cce194f34404cd48735439fbf2e6684 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/12232e6c87144d899b7d22a1d1f4a18a : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/16bacb458b804c26a334cca1a13b67cd : SUCCESS in 19m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6e298c10755e43a19b9f7abe5bfdf8b7 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/917af4360c1449ed891d93bd7dd0615a : SUCCESS in 18m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/267d5369e9ae4ba08e4ee206d417d7e2 : SUCCESS in 15m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1237b615c292459fbc02cb8e6a9daa11 : SUCCESS in 52m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0eb2bbf813749aebf707b85ec0784ee : SUCCESS in 46m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fc53db6c2544bb8bfe3db61494d3d61 : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a755e0359b845a098bb122d2df0dc4d : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f8794aa039894d4b9e0768cb9ea575c2 : SUCCESS in 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5d4cf7b58da68e81a46ded72e70705c70a89107f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 10:59:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb765a70e69940f29f513cea28900a4a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2dbd39e3d589414e9adac20fde1b033b : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8c67c7866274ac8a4b978f53eb460f6 : TIMED_OUT in 1h 31m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/53f5eca5d64a4d4fb6e64f15484fb117 : TIMED_OUT in 1h 31m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/492a57c6abbc492a86515565877ec290 : TIMED_OUT in 1h 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c6f3b0784a79f2768c087ce6a8db02d43acbe970","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-19 10:23:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb18436a803d3c69a39d6874c9253aef407bf15c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 10:23:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad404c672eef701779cebd1d4dcc81a96369b473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 13:01:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/409fa5daf13f437291a038237d1f075a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/301211e4861f490bbad6bebcc2565b11 : SUCCESS in 9m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a3100203d0f44973900f5e053b87c11e : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/41c46906c7d14ebdb43e8d8cbad5d854 : SUCCESS in 15m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c6e69f62aa24667af203021d19e047c : SUCCESS in 16m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4dabada2524245f2be44b61233b06d1f : SUCCESS in 17m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/04326ad6078d40e885dd1f408ea1dd75 : SUCCESS in 14m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7f315d6240f24de4ae4a6a8539772c8d : SUCCESS in 22m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/560cd45d60f549f7a3322c055ec71232 : SUCCESS in 24m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/62a0d438608e46dbb72c9af276393a5b : SUCCESS in 20m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/503fd7b01e144a98b33714d50ec54d09 : SUCCESS in 27m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/67709826130947129dd3ae0ab93eef8a : SUCCESS in 19m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2c7ca45c80f5478fb28f390983d6a554 : SUCCESS in 25m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/df35638d2dd04d74b96621266ba4e248 : SUCCESS in 18m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f77013a656374f61a85dd7b8ff263531 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8fcdf2a7385f46c3b4cc8a66a6252cbd : SUCCESS in 7m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce215c99b5624220929eb605316f6bd1 : SUCCESS in 46m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30e3fd42d6b941a486e6f90c9341bfdb : SUCCESS in 53m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc3d9e0def0b4a2c9503d605daf39bd0 : SUCCESS in 19m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4aa8642cf6242d78b1449d1e63f2b31 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/51c21bc080954cf5b70377c1bcd8e4cd : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f90b981a79a5c1fb6ddb835aba15e0550136ad77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 13:01:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"94d42f5ebc18bfb22aec7e1774a0ada140a76545","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 13:02:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0637edaf5c764da281693ccc1ac7e40b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be1fcdf213d0407bbced3b31da7c9275 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/124797b8939744a6a7bfbd82d39d56f7 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e984de864dd0895f4bde81bed56691c48e7cea52","revisions":{"e984de864dd0895f4bde81bed56691c48e7cea52":{"kind":"REWORK","_number":1,"created":"2024-07-17 09:25:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/13/924313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/13/924313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/13/924313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/13/924313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/13/924313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/13/924313/1"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-17 09:14:24.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-17 09:25:10.000000000","tz":60},"subject":"trivial: assert non_negative_int treats float as valid","message":"trivial: assert non_negative_int treats float as valid\n\nPrior to the Related-Change, non_negative_int would raise a ValueError\nif it was passed a positive float. It now casts the float to an int,\nwhich is consistent with the docstring.\n\nThis patch adds test assertions to verify this behavior, and similar\nbehavior of config_positive_int_value.\n\nChange-Id: I5d62e14c228544af634190f16321ee97a8c4211c\nRelated-Change: I06508279d59fa57296dd85548f271a7812aeb45f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e984de864dd0895f4bde81bed56691c48e7cea52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e984de864dd0895f4bde81bed56691c48e7cea52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
