)]}'
{"id":"openstack%2Fswift~924352","triplet_id":"openstack%2Fswift~master~I92fce513d06d8b0875dabf9e9a1b2c5a3a79d9b5","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-07-18 16:58:59.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-22 16:08:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I92fce513d06d8b0875dabf9e9a1b2c5a3a79d9b5","subject":"swift-drive-audit: Work with ISO timestamps","status":"MERGED","created":"2024-07-17 18:09:38.000000000","updated":"2024-07-22 18:21:54.000000000","submitted":"2024-07-22 18:19:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"924352","meta_rev_id":"427c842c9d33ce71073960f101611e3ab73bca5f","_number":924352,"virtual_id_number":924352,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-22 18:19:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-07-19 02:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-07-17 21:50:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-19 22:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-19 22:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-17 18:35:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-17 19:19:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-17 21:50:35.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2024-07-19 02:20:44.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"ac66087b18d7957175c06797de1e9979847b7b06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-17 18:09:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27772fcab396aee462e61450ba2c26c5467cbd86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 18:35:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d1c51f4602416dafebd1fba01dc0a9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/173569a64d3748ae897fcdbc2dbab29f : SUCCESS in 19m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/301a39d5de714d1d9ee793f604c610fa : SUCCESS in 21m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b80f022cda404b60a920f0b22f809482 : SUCCESS in 23m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8da3917a9abc455aba5f7fcd1d494fd6 : SUCCESS in 24m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ca36635923cd6b6f9f7f4d77cc8fa932fd86f26c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 19:19:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28be7cb0d11f494eae1170325d076c5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def9c1c8dfac4ce3bf3884730d74b8de : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/7d047c8f08cc4c94944a5f3b3fcf5828 : SUCCESS in 51m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3caf74783d304d44a5ce517bf4a2c50f : SUCCESS in 52m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/371da8bc06e9477b8b20674a3f1dbb97 : SUCCESS in 1h 01m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89c81418ea7141cc8a60d1e672398dd8 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f465683918949fb9dfefa4f13b6d71b : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09721cd96bc84149bf801e2438a36104 : SUCCESS in 23m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de87c104fe144a2fa78156a911f3975f : SUCCESS in 9m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f76d9e1ce32d45d2a4d507e43f8591a9 : SUCCESS in 19m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe8a369d4b7c4161a9109776cee238c1 : SUCCESS in 19m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/89090663e4c0443fb8e3565eef82a938 : SUCCESS in 20m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bc693efe7a554993a5d76a94e1f92bcb : SUCCESS in 26m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ecf2da41589e43ea857f66453c76b126 : SUCCESS in 26m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cfc9da6ea50245abaa11d488a3cd6855 : SUCCESS in 20m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/63dd3e339a824a979c3b01d54cb29354 : SUCCESS in 20m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e119151f152d465ea8b990b54c225669 : SUCCESS in 19m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/af469aa144024864b8937c964cd7a3d3 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6458c83245444820a87c3921dabaa7ee : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eabad687b1584201baa027858f557914 : SUCCESS in 18m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c82630c4d0214938a6fb0acb04005add : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/67a79dd6eec54d14851add8566588ce8 : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/79d60e3fcff44452a09b210dae09df97 : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d9a52822a5c4152be02150e3242cb02 : SUCCESS in 57m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ccddd71d3944b3db0026ea5c3f7b031 : SUCCESS in 51m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f5abeb9ffe44bec9bc6118131ed2fe7 : SUCCESS in 54m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24d56d0301c94a6aa4024d2b71dfdadd : SUCCESS in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1dd317165734f848941cb48a0346549 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0c06a6f658af4b76b0798f950f697794 : SUCCESS in 25m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93c6026138164ba58ad7aaa48d11ebf9 : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7a278f964eb64b599bc7f0b324bb9534faaaf12d","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-07-17 21:50:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a46ee44c7167d9560695144da64cf0074f120795","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-18 16:55:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f8f24380d5c44c568b233e0b5319650ba322b31","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-07-18 16:58:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21d856238ff67bb1c7cd3c3f4f1c2b87f648d102","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-19 02:20:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e0391fd6e531026c99355fba769182aba02ef28d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-19 22:18:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90496d609c87b5a70031b0f3d28de02448101a70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 22:19:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4bcc7ccd70fd6f3d21f59c452bae29bc58dfeef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-20 00:20:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f938009369204deba6befddc0c84624f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63da30937b844bb4a2ce1582c1645382 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/0824ea2141e94fd4b748481c4abfc0db : SUCCESS in 1h 02m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5c40d5e87944457aabbcacb89fcc4113 : SUCCESS in 49m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db7d989b750d44ae8b2948bd1db827b8 : SUCCESS in 1h 14m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e9a1cbc03b0b4046811db42c891e36cb : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e141b565fa7f4013b8736ed2be612b4a : SUCCESS in 15m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/69ac2a04aa46463f93433b839c284de8 : SUCCESS in 15m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/abe8f91881a74c6eb8d7ca3183c2301b : SUCCESS in 19m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2a9f6872fd244b96ad83c4508c1469fb : SUCCESS in 22m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/082bf3d33d294326b32ae5fe66e7be9b : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0ca909cf74c4b6195d83eb0f6d43cb3 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d4360b63e4c34786b72d07d7a92df6a8 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e49f634c475248bd8f1434b2fe2baf95 : SUCCESS in 17m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ee525e0f848a4672858567b82c2c3c87 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2632ca9ff23f4ff2aae12f8c3b50470b : SUCCESS in 18m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8652e873b6444b268e7f00d36911776a : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85a25af2182d425cb866f212a5058bc7 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/008d23c2bfca409d9dcf1c6d002c69f8 : SUCCESS in 7m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d195aeecf2b46fc90f4695275991942 : TIMED_OUT in 2h 00m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bde330d2b0114c29a5ca0f4f5328e11a : SUCCESS in 33m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92acf3ee36e64b5497cb95057b9720ac : SUCCESS in 33m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/659240723bc045caaafeccfcfbdceb72 : SUCCESS in 15m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83147190368544c9839b19a203cb83e9 : SUCCESS in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea1a9cb7183949b29f4c1cf487a6c702 : SUCCESS in 16m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b855119d4fbd41f0848cce15262a1579 : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":1},{"id":"24018841686e00fb4e2b82c2b3fd290fc0a703d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-22 16:08:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4680fb58690da0e7b54711ed760e22715f432c1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:35:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/747a0a41014149eb9362139a40678f70\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b4bd7e3f4644b88964f54a6cb73f23e : SUCCESS in 20m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7830207ddf4c418aa867dfc5468be485 : SUCCESS in 23m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe3e4ba9fd6345c5aec73c85ee043060 : SUCCESS in 25m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d006364be88f438ea97878f8f23e3179 : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2d56c6d5b91a0934472ef67a5519676d177c24a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 17:11:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02810bb4523c4827a41245578869cf7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957d10b260174fb3bdc8c79ce2b5fa48 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/c01fff8c1c1346f8af2b4022250de82e : SUCCESS in 55m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0527e75c7faa4cd2a15f44e11c357e93 : SUCCESS in 43m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65a22c01b66d40929fb89086d8d3f73c : SUCCESS in 53m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/662ce8cfecdb41c09b507bbdecb3c129 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/faeff07c9739473bae293fab8d848c1f : SUCCESS in 22m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d53dbbcf8f454f0bbcbc63fcd2ee3066 : SUCCESS in 23m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a6ea0db24c841ea8901450fcef25ea8 : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa1f174d4c234c8493cde69cf9b71af2 : SUCCESS in 21m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64858b11f3114e419473635ad3a38256 : SUCCESS in 16m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/15676eb6500a4f42abab82da7a103c4b : SUCCESS in 19m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/15fae20d564f498f87a1f5a2450a575f : SUCCESS in 21m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f1cc4bb565474e769bfbffeee6f3da4e : SUCCESS in 30m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/486ff48ac859463eb7d801ebd533efbe : SUCCESS in 22m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e46c6a5e617a42b9b2308b42aee12a68 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a0fb46615062485abf13507e957eb53f : SUCCESS in 22m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6e972b3d98bf4851a610efa5c9629ae8 : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0bc4564ba0f94f7fb91bf662116938e3 : SUCCESS in 19m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/baca8efa5bba4d34a0975aba926f48ad : SUCCESS in 17m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/be70184d18e848f484fd846ea51bf374 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/90e7f55211b94fa1853359dac62dc2f5 : SUCCESS in 12m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ab8b5a35a4754ff7a76eab3690f341a2 : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/15ce617c1e0a49cbbc43fdbeacab0b51 : SUCCESS in 57m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fcc6ad13efea4d959fad22827b93a2bb : SUCCESS in 45m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b511d37f12694d3599f5d0e2f9850d86 : SUCCESS in 47m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea95f2b796b8439c8883799883aadf99 : SUCCESS in 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09df663e01994b4f8187497bf3991472 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1e087176b4f4b89a70576b124d62fe2 : SUCCESS in 23m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59bf841c49a44c7691f7e9b20ddcd5f7 : SUCCESS in 45m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ff9bca05c842fbdb664470237bf2450eed6a530a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 17:12:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00518c53497fc6849e31e0214749514a7987fc01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:19:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f405b3d67d83498ca0a0f717669157ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f79daf54c0614575a2161af90300b424 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/9690704f08524c829a39e529140fecd1 : SUCCESS in 1h 07m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63061f435c6e420cb7e2139b1ac43f53 : SUCCESS in 46m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a5cec274b7440d0a505f3a2ac5c175e : SUCCESS in 48m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/664850d0c0c148399afd02b7d79a6193 : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bebddd13162941bf8b9be95200a72184 : SUCCESS in 22m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bfc075eb57f1444eb6f63caea94140b8 : SUCCESS in 18m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e4d72224919549cc93f1320620c520f4 : SUCCESS in 24m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/18c618cd470d4a199c7cc9ac90a978d5 : SUCCESS in 19m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e4099b9d307540b2ae7955c262c56d01 : SUCCESS in 24m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92aea8ee80084a5c8f214a3e1880b9b2 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3716cd306b304b5aa544df0503f1cc4d : SUCCESS in 20m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b8d9c4b0ace34aadb5762ceb6f90726b : SUCCESS in 21m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a264e94c32334d1fa12b148cdb04a417 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f796e01617814f2bbbd3cd3a11d1ee80 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c03e8fbc62d143f5bf7413f918ab2eb7 : SUCCESS in 19m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8c2ef023633e44c0ae52f70179663288 : SUCCESS in 6m 42s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4f1f2019c878456aa6e86044a3bcbadc : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c53b6f879f6d43d1aeb71d8a139f0e35 : SUCCESS in 52m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8f926f4a84e4307a9589dbef8235350 : SUCCESS in 34m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd551b6417104b069c71b1f9f3c64298 : SUCCESS in 48m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/809f665887d34904b1300b945ad0720d : SUCCESS in 20m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7815fdda8aba4ed19cf73a50d161d903 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a325b490fcfa4b00bf3badb166fd15e6 : SUCCESS in 23m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16ddbf9a41214e06b3d4238e515f9610 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fcc9c831448679f60799f68e7b59c4db4fe21726","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:19:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"427c842c9d33ce71073960f101611e3ab73bca5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:21:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2c1913a5e2d41208bf76bb7a59d94c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/235556ecb0704416a6cbc32abe7a17c2 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f82d4b89a2f340c79b9d0a2772880421 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78b986cb390ad6e985b930ec6892b4efa226f53a","revisions":{"78b986cb390ad6e985b930ec6892b4efa226f53a":{"kind":"REWORK","_number":1,"created":"2024-07-17 18:09:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/52/924352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/924352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/924352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/924352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/924352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/924352/1"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-17 18:07:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-17 18:07:39.000000000","tz":-420},"subject":"swift-drive-audit: Work with ISO timestamps","message":"swift-drive-audit: Work with ISO timestamps\n\nRecent OSes are using ISO format for their kernel logs.\n\nCloses-Bug: #2072609\nChange-Id: I92fce513d06d8b0875dabf9e9a1b2c5a3a79d9b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78b986cb390ad6e985b930ec6892b4efa226f53a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78b986cb390ad6e985b930ec6892b4efa226f53a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
