)]}'
{"id":"openstack%2Fswift~924431","triplet_id":"openstack%2Fswift~master~Ie34bb404a140119cadaea53bde7dea9924e07c5a","project":"openstack/swift","branch":"master","topic":"policy-modification","attention_set":{"37058":{"account":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"last_update":"2024-07-18 20:32:06.000000000","reason":"\u003cGERRIT_ACCOUNT_37040\u003e replied on the change","reason_account":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"}}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-07-26 03:38:50.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"hashtags":[],"change_id":"Ie34bb404a140119cadaea53bde7dea9924e07c5a","subject":"container storage-policy modification operator tool","status":"NEW","created":"2024-07-18 19:03:30.000000000","updated":"2024-07-26 03:38:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"507d6278bd1170b993ee54c300909bc305eea1df","_number":924431,"virtual_id_number":924431,"owner":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-25 20:49:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":-1,"date":"2024-07-26 03:38:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-18 20:16:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-18 20:32:06.000000000","updated_by":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"reviewer":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"state":"CC"},{"updated":"2024-07-18 23:24:52.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"bc487e2b74d2a94abeb07241e9d378b9bf50cdc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"date":"2024-07-18 19:03:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5728a5b872ee4b161875107cf3a31f6d36f429b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 20:16:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2017bb6b909e4f8b8fa473d5f5069c12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9644b267cfd147f6a83da0222893f241 : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/949aad56a5e14efb854cd3474a5be562 : SUCCESS in 55m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b88af0a2a16f4fd88849066078bee1e6 : SUCCESS in 50m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fd3a3213bef4b0195ddba50768b41ea : SUCCESS in 53m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24c4c74fd5f741498175ca82e1d16724 : SUCCESS in 34m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee081a4fcf974b46930566e59387148f : SUCCESS in 33m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/576f72299f4d45e992d7a96eafde5072 : SUCCESS in 21m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a4281a55ea564d95a2b0c33c48f33b84 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b19a54eef02147b4bcec643edb839bee : SUCCESS in 21m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b03791976964e1a82b1b9c0459b8c2d : SUCCESS in 19m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b721621a222d4a5f8d332f0dde59bea3 : SUCCESS in 20m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b308d941cce34167af5286898aaa5941 : SUCCESS in 20m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/30333b66bfd949b4a91b61ec548eb767 : SUCCESS in 28m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3b37093f3037420eb6c265d8e433f7aa : SUCCESS in 19m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b7c89308ce34998930292b8aee4351e : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b1428f49bc644eafa2f733c3820e9457 : SUCCESS in 20m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/49a404e4a7ce407ea7341e838e4761ac : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eb208a2e951e410b847365e16bc7d2f9 : SUCCESS in 17m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5329dc43d36f4d4eb3303c3dc0f72e2b : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c20559bf97c34616a740d9ede218da87 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/564c8f1463214a8a9af710da4aa4822d : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fa0eb1905a094594b3f733c181ec6400 : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/314f3259cc4849e19c28d3217da70291 : SUCCESS in 1h 07m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c44394cab18444b8d656870ccd300f0 : SUCCESS in 49m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31d03456af3444d382cdb9de4cea0f73 : SUCCESS in 33m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f65a4f5b23e43a595d4a01a4278577b : SUCCESS in 20m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10398ea522ca44e2ac614347cafedffc : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91b85044b58848df875847561e8082a2 : SUCCESS in 17m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2430bac0945a4ee1a826c5f13ae32dd9 : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6ed40d9c98d4faaac0782a959bb0e86d2a0cd835","author":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"date":"2024-07-18 20:32:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24c2e4bc12c3e42b10ffd4f9b15a46f87d416d69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 20:36:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9132fe097c8c4a1593fdc823291ed248\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3809730a148d47eca1a3e0f1138593e9 : SUCCESS in 25m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d866d354707b4f549932ca25ee4536e0 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/272d50e746b544c1abba462b2ae51ce8 : SUCCESS in 28m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c77546cba9504735bb09a236ccccf5d0 : TIMED_OUT in 1h 31m 11s","accounts_in_message":[],"_revision_number":1},{"id":"19a3c15a2d9905e0ab1064482ec5b5e4f636f3d2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-18 23:24:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b9418f8ec23e24cfeaad5a4ddd3e076f9c27b169","author":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"date":"2024-07-19 10:49:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ca63237cc3dcde2c71779645b8ac672775c7d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"date":"2024-07-25 19:46:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"58c196e05c8f25ce6e631a6d8d24969694cb4dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 20:49:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1eb5fa6486d4633881e9f127fdbf3c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f4e57b3a624bf292e1e30b64bc4a53 : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/702288bbe12441ef81e673719a6dd8bf : SUCCESS in 1h 00m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ba6e19e9e5e43e6ac10843796b5ce02 : SUCCESS in 44m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d07140fa37df473d8706a70a776ef12f : SUCCESS in 45m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb5a16210d5242938b25cb30325adc10 : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/389838799e714f7ca0c5afb4ed8b27ad : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1dde578cbcc94747a976311adecfa326 : SUCCESS in 21m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/720209b4335843b2b696835b178babfa : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9fca2d79ab004cc298554482ac7e96b4 : SUCCESS in 19m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef0b21670f9849eabb825b40b9275012 : SUCCESS in 18m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/58087eacb25046398394417b790652fb : SUCCESS in 20m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/06d413b1229d4aeca10a56ef56c4be5c : SUCCESS in 17m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0e7ad188a3a9467683e2608d34225209 : SUCCESS in 19m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3fbd7df8f32447bc836730a597086e22 : SUCCESS in 20m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/67fc3138de3945c29236054499cfc758 : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d87a90d9c7646b8bd65389845dd8c51 : SUCCESS in 20m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2a28e9494b8e4459977e8efd201fb3c4 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cb6df49f073142c19448534558be24df : SUCCESS in 18m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f914bd46b1184748958690710bc91f3a : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f1e3148efb9247e1ab915f4fe70065f5 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b1d5b819e0b49d587c9bd0ffedb792e : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c9a85d9af154ffb92f9264b4f114b10 : SUCCESS in 6m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1a3ee296d8d4f44b17bf14c8dcefd82 : SUCCESS in 50m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f49f7048ebf45d2883a138b9737699c : SUCCESS in 48m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5351baf5ebd446d79962c76c3205aa70 : SUCCESS in 51m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b9dc16fd4b445739b77153266984885 : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31aaf15cbcc249d7ac0509ff469684a2 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ab5cfaa87e947ac91d19de35a41b36b : SUCCESS in 18m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8aa45422553c457da0ef7f456ed54498 : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4b1937c03e5226064be04542e058cfafa36c8a2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 20:51:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/730659f11b7840bc980404232ba1428e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c807792c8e5418380c2babc73625683 : SUCCESS in 55m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0175c7afcee94e0ca79b77e111d099a7 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/30ab8ff2b2854f5a94f603fdb974917b : SUCCESS in 1h 03m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a603c76b40444da934067fe6443485f : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":2},{"id":"507d6278bd1170b993ee54c300909bc305eea1df","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-26 03:38:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9fed84dbe61236036f4e8ee15bafedb23eeab2e9","revisions":{"8eb73af87d77fda1b30cd10260dd2a7d79a460bc":{"kind":"REWORK","_number":1,"created":"2024-07-18 19:03:30.000000000","uploader":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"ref":"refs/changes/31/924431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/924431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/924431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/924431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/924431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/924431/1"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","date":"2024-07-18 18:57:35.000000000","tz":180},"committer":{"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","date":"2024-07-18 18:57:35.000000000","tz":180},"subject":"container storage-policy modification operator tool","message":"container storage-policy modification operator tool\n\nOriginally, the storage policy of a container could only be set at container creation time. This aims to change that by introducing a dev tool that allows the operators/admins of a\ncluster to change the storage policy of an existing container/bucket.\n\nThis patch allows resellers to change the storage policy of a container by sending a POST container request with an \u0027X-storage-policy\u0027 header specified\n\nCo-authored-by: Azmain Adib \u003cadib1905@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohammed Al-Jawaheri \u003cmjawaheri02@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\nChange-Id: Ie34bb404a140119cadaea53bde7dea9924e07c5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eb73af87d77fda1b30cd10260dd2a7d79a460bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8eb73af87d77fda1b30cd10260dd2a7d79a460bc"}]},"branch":"refs/heads/master"},"9fed84dbe61236036f4e8ee15bafedb23eeab2e9":{"kind":"REWORK","_number":2,"created":"2024-07-25 19:46:08.000000000","uploader":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"ref":"refs/changes/31/924431/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/924431/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/924431/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/924431/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/924431/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/924431/2"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","date":"2024-07-18 18:57:35.000000000","tz":180},"committer":{"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","date":"2024-07-25 19:44:50.000000000","tz":180},"subject":"container storage-policy modification operator tool","message":"container storage-policy modification operator tool\n\nOriginally, the storage policy of a container could\nonly be set at container creation time. This aims to\nchange that by introducing a dev tool that allows the\nadmins of a cluster to change the storage policy of\nan existing container.\n\nCo-authored-by: Azmain Adib \u003cadib1905@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohammed Al-Jawaheri \u003cmjawaheri02@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\nChange-Id: Ie34bb404a140119cadaea53bde7dea9924e07c5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fed84dbe61236036f4e8ee15bafedb23eeab2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fed84dbe61236036f4e8ee15bafedb23eeab2e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
