)]}'
{"id":"openstack%2Fswift~924451","triplet_id":"openstack%2Fswift~master~I10e2767e88757fa74f1e7d3a66120500233d7c9c","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-18 22:29:50.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I10e2767e88757fa74f1e7d3a66120500233d7c9c","subject":"Add streamlimit middleware","status":"NEW","created":"2024-07-18 21:25:36.000000000","updated":"2024-07-19 00:07:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3d05d68ecadfc52ea2da32b587429694e973e84b","_number":924451,"virtual_id_number":924451,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-19 00:07:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-18 22:29:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2431827367f9657b565746ea267378e345b95ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-18 21:25:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd670a846918eb9e17a1afc53346dda7f6f4af42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 22:29:50.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8df3a818e4af4ef88bbbcb94ca6392b1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/862639db3ebf49188e4daab2797011e6 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e6ab8c8daa48929d66be841c077bf5 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/95017bbcd4f545bda515190d12479716 : SUCCESS in 47m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/df0ddd344acb49fbb702d7b4b356f2de : FAILURE in 14m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2306c8142ea24fd1a3aa068220f6f537 : POST_FAILURE in 26m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12836d21bba149b28f79b7a93d2f7e06 : FAILURE in 6m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/538ee81b0742498da49d6ed35f483e13 : FAILURE in 5m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/438eb58639b142578cccb40ee5b7c907 : FAILURE in 7m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae94d61b54494593aadb203aa739b89f : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/20c89293a2cb46788bdf063be543ed9d : FAILURE in 21m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9aca1d751c374ee38e546ae3409c07bc : FAILURE in 17m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/20e7be89fa66449794cab932d63f0570 : FAILURE in 20m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/20301536ead94e20adb56bde3fd48654 : FAILURE in 22m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7bd50b4a858a4bfaa37eb8852958af01 : FAILURE in 33m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4e639fe6a60347d884675aaa9b6796c3 : FAILURE in 3m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2e49b1591fff4da5a59b6e4c55c01e95 : FAILURE in 6m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a29cf60e51345f6b4bff78988e28bbc : FAILURE in 3m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9de7edc0339546bfaf90ef1a2ed1dd03 : FAILURE in 3m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/587021a87e3d434aba7f334a14ce9634 : FAILURE in 3m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cb9b28014328477781030d5cd43aa117 : FAILURE in 4m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c0cd101d922c4a14b8dda254fc583df2 : SUCCESS in 8m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/100fb9f534344afeb112e241a89026f4 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/04780369e2c1434cbf8fbb3a113e7725 : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/18df77f0cb584921aa34cf10491fb451 : SUCCESS in 57m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09a79839289847269d421b0791047370 : RETRY_LIMIT in 8m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37bfe3ab3207466c9745202821511e97 : RETRY_LIMIT in 9m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/623b03c37bb54de2ab5f62511c1d8aad : FAILURE in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b30bdc382e4af1a02cfff52eea3784 : FAILURE in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e6fa04a0df14b0ab84bec1ffb04b032 : SUCCESS in 18m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99228dac67814633a8f6ddd4cad9c189 : FAILURE in 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b683c80791220b7e7f9791a51b644ba1b234b4d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 22:33:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6afafb5d18c74b0492f95f6ea72c7594\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f25c40aa10d4835aea4984b88a0ee4b : FAILURE in 1h 06m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34075951b8f046259f75e7b7581a3215 : FAILURE in 19m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8de080ac98624cbb8840381d15327c24 : FAILURE in 9m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2178dbfb00b04d81a880e3e85b0ab992 : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"03442786f3b10a485522222c6fb11caeacb6cddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-18 22:58:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"06ba96253183b7d3343601fb0495682521f76ffe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 23:23:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a27e395e5e49d6922c8b519b696b0c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b01de189cc944e9b4919cd901bb4c99 : SUCCESS in 16m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7774cd253ec407c9f3e6dc47b883ab7 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b74ebe81a18494f82550a847889523b : SUCCESS in 24m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/edd385464adf41b2b9916c89b984eb26 : SUCCESS in 19m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3d05d68ecadfc52ea2da32b587429694e973e84b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 00:07:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28a9e46772c7438fa95520bf8d4d2458\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/62358c7c98404a38aba8c2fd81a35b56 : SUCCESS in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f114f8c0f44a2b94d967f0412023aa : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/681501e80d2141d08ebd44d3203815ff : SUCCESS in 48m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e709765ec34400ea8f8d123a6043884 : SUCCESS in 42m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30942c61fe104ead8646b6b8869f9054 : SUCCESS in 51m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5669acc62184e3d9139aa7244b0e323 : SUCCESS in 21m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37da58de37ef480d9622e7ab53479bff : SUCCESS in 23m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35258ae6449f49509ddf2f673f3fffbe : SUCCESS in 22m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/74db9471b5ad413b986e668f1f70469c : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ab2071d2ec364e19bffbe0e223d39ff3 : SUCCESS in 21m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ae1963b2d9d840e589cfacdea31e860b : SUCCESS in 19m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/796fe98d18ba4de99f9b99daf19ba3a9 : SUCCESS in 20m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3fd1bb3571724928abfa8374624e5be2 : SUCCESS in 22m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8f81e3b0439647cfa5d16f86e113d955 : SUCCESS in 36m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3f75414c213242c09294b151950d898f : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dc9046eae9d34188a4420251a46ef974 : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e919563fdaf46aeb56a0be0643c2bac : SUCCESS in 19m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c6c1b97f41ae4ba29144c0fa979e63bf : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a64d40789b5d4191892e040f8a7d38e6 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3db9633d86c44b72b16fc214c7831ce8 : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/06c14bb9a1024f6d8649430d55f89516 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ee3f4769123c4bd8afdaaa6067415ff7 : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e6e2b91e117468387a40bcb119dcf31 : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a780f9297794099aa409a04d5ac95c4 : SUCCESS in 56m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f25d080f7f8d476489db0d52e58016c8 : SUCCESS in 53m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58fa7f24f25942fa9f5bb221f8b95b00 : SUCCESS in 1h 08m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8db12c8aa69a4d57983f39e4f0c9a7fb : SUCCESS in 15m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3da09bbcc54940aab5b1ed7f7281b9bb : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8887440c95544f7b9bfe4f2e3079b8d7 : SUCCESS in 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cc757cdfdab456fa7d74a75fb06982f : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1834dd534cb78447bc823d97749a058352e1b796","revisions":{"8e44b7742202793554c41ea8697f539e0f6f99f0":{"kind":"REWORK","_number":1,"created":"2024-07-18 21:25:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/924451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/924451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/924451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/924451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/924451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/924451/1"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-18 21:21:10.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-18 21:25:23.000000000","tz":-420},"subject":"Add streamlimit middleware","message":"Add streamlimit middleware\n\nThis intentionally caps the bandwidth a connection or worker may use.\nThis is mainly intended for testing.\n\nChange-Id: I10e2767e88757fa74f1e7d3a66120500233d7c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e44b7742202793554c41ea8697f539e0f6f99f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e44b7742202793554c41ea8697f539e0f6f99f0"}]},"branch":"refs/heads/master"},"1834dd534cb78447bc823d97749a058352e1b796":{"kind":"REWORK","_number":2,"created":"2024-07-18 22:58:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/924451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/924451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/924451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/924451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/924451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/924451/2"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-18 21:21:10.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-18 22:57:53.000000000","tz":-420},"subject":"Add streamlimit middleware","message":"Add streamlimit middleware\n\nThis intentionally caps the bandwidth a connection or worker may use.\nThis is mainly intended for testing.\n\nChange-Id: I10e2767e88757fa74f1e7d3a66120500233d7c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1834dd534cb78447bc823d97749a058352e1b796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1834dd534cb78447bc823d97749a058352e1b796"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
