)]}'
{"id":"openstack%2Fswift~924511","triplet_id":"openstack%2Fswift~master~I7039456f0f3938e4b1eb626c16310621ec4a8142","project":"openstack/swift","branch":"master","topic":"mpu-expire-randomized-chain","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-07-24 10:39:29.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-07-24 10:39:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7039456f0f3938e4b1eb626c16310621ec4a8142","subject":"sq? expirer: check task_container_per_day \u003c container_divisor","status":"ABANDONED","created":"2024-07-19 14:03:41.000000000","updated":"2024-07-24 10:39:39.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cc89669a2ef0b25096ea86a2c6acac1a03ef5bdc","_number":924511,"virtual_id_number":924511,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-19 15:15:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-07-23 19:02:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-19 14:15:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-19 15:15:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 19:02:51.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"739a535343846dea2dbb458338549f0d1cbcb536","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-19 14:03:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e488d6f13c8b4efe2db5c9729e1a5d5f7f9fade4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:15:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db05513a6f9c4c149013474618b81df7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6a4bad20d3748f18fd3196abfffc291 : RETRY_LIMIT in 2m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13bddd4f0117429fa8dc6b89b454ebb9 : RETRY_LIMIT in 2m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b778def9bd040ecba839bea037c16ce : RETRY_LIMIT in 2m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/488dc4aa05294e93a2cd522afa9ecdd1 : RETRY_LIMIT in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f83978440a45b1a3650657206d7c3bc0742905a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 15:15:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5efd7b710ea846139789d4fa1f204d22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82efba8646ad4fd6bde251e2a3160d7d : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/22665397779d402dbde04043e1bc7708 : SUCCESS in 1h 02m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dbaca77ba22c48f4804b6803b481c6e5 : SUCCESS in 49m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83de42f9fe384b6e8b049d2d0267c333 : SUCCESS in 1h 04m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed47918250dc4cf4ac7b681b15238577 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bea9e54e80d47549629046795a66958 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/793dd27cc9db4cde9f753e6a53e2d9ea : SUCCESS in 28m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40057b845b80418b931806c2968bf969 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e7f8f26d99894ae7944972125cda063b : SUCCESS in 21m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c9a4a742e31e42df86a159e43a31b359 : SUCCESS in 20m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a6047c8d44714fb589522c643f5ebe95 : SUCCESS in 25m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/565648568e3d44548d5db42ff24496bb : SUCCESS in 24m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/24cdd61e761c4e319f3423da8add821d : SUCCESS in 27m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f8d1a90e6c724d32ab0e123c80e45985 : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eee1b6644c754242bbe525ede43b8a28 : SUCCESS in 20m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a2245e190384b0e8986bef5b564a863 : SUCCESS in 18m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8be39be38e144c0184332cb9e474a99f : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1927f0d0babf4412978de8268b52cd24 : SUCCESS in 19m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/04bfef95298549018ac9e4f8f002326b : SUCCESS in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/879299a8d1254e3fbe2214947b966995 : SUCCESS in 8m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bc8eba4b51a3402f9f5a3b5a6277e544 : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a06cd20533684f2c938e5e42e130a0b6 : SUCCESS in 10m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/76b806094cb9481db5bc72470d1e0775 : SUCCESS in 1h 05m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/511da98e8de24afdaac97aff48df7cb6 : SUCCESS in 47m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/38624145e0694e69b7d8d4748b0f2f32 : SUCCESS in 56m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf9824caaf804ffab4f17dcad554ece4 : SUCCESS in 21m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb4d6e43fb5a4c6a9521d95ef6d8f3ae : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b62c197b6d764395ab00b7b473b1e99d : SUCCESS in 19m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6643ebc8ea44a8e94a0fd363fcd8d5c : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5073f2910cbcae97762961f6be2b61aa56669826","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-23 15:56:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6521dcd7969efc147fd8bd01fc2cdc08e7c71887","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-23 19:02:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da2e8842503d471aa3d81ec68cb3b0b3105dd577","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-24 10:39:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc89669a2ef0b25096ea86a2c6acac1a03ef5bdc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-24 10:39:39.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f1db46cbe48d40d6ce78ae59c3cbed31c4a8638","revisions":{"1f1db46cbe48d40d6ce78ae59c3cbed31c4a8638":{"kind":"REWORK","_number":1,"created":"2024-07-19 14:03:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/924511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/924511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/924511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/924511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/924511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/924511/1"}}},"commit":{"parents":[{"commit":"17faac823373f58cde95365c4fae359ae6648267","subject":"Configurable expiring_objects_task_container_per_day","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17faac823373f58cde95365c4fae359ae6648267"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-19 13:56:57.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-19 13:56:57.000000000","tz":60},"subject":"sq? expirer: check task_container_per_day \u003c container_divisor","message":"sq? expirer: check task_container_per_day \u003c container_divisor\n\nChange-Id: I7039456f0f3938e4b1eb626c16310621ec4a8142\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f1db46cbe48d40d6ce78ae59c3cbed31c4a8638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f1db46cbe48d40d6ce78ae59c3cbed31c4a8638"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
