)]}'
{"id":"openstack%2Fswift~924667","triplet_id":"openstack%2Fswift~master~I18d2b1bc701d438036ff3c7ee337e55a8120b454","project":"openstack/swift","branch":"master","topic":"mpu-expire-randomized-chain","hashtags":[],"change_id":"I18d2b1bc701d438036ff3c7ee337e55a8120b454","subject":"WIP: sq? alternative expirer parallel_strategy impl","status":"ABANDONED","created":"2024-07-22 17:31:21.000000000","updated":"2024-07-24 16:01:16.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"97b6134a1e18ed7cce44ce5c479c5017b86442b1","_number":924667,"virtual_id_number":924667,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-22 18:33:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 18:19:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 18:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73e88a8fadc43901c32306f1a27bf32117fda3d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-22 17:31:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cc2cd5e3535864213f5b987eaa07e5ce686a248","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-22 17:35:39.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2da3d04fe5d1c3be8bc099525fda99d1e141bd28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:19:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/872baf0cba554f3d8cd2537378fa2b20\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c5dff722fe64ea9a5b95c1309eb45a0 : SUCCESS in 27m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ae3fa3bcc004e06831dadf98dd9616a : SUCCESS in 20m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8441284442154f3eb8a28017b8fe608b : SUCCESS in 34m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1738a9e544348b9b92ca2959421c942 : SUCCESS in 45m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b238d10bc2435b32745402c12ee85844c5058709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:33:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/214ee5db8cb045e9a41cf95141c46f2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7488e7cfe8324925a014f3f79fa9382d : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/71952dd0f40f400aa4f83d66b8e5d767 : SUCCESS in 1h 01m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/559a82ec93c246f38ebf22b8ab6d976d : SUCCESS in 47m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f46aa2b384a94eaa83825f88e88b64c6 : SUCCESS in 56m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d774b5baae8a4841a8897c06074bea6d : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1faf16b1ad374b69be10d21dac456d2d : SUCCESS in 21m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7427a067131f471d88ce46057cdf6fdd : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/981a78cafa30407a8e9c9931bcf8a503 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d9488d4206be447f821c38f85660dd58 : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e21565e82bd647659c58d77c42bfd264 : SUCCESS in 18m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/00569b2e9a054ea89bdfbcbec467adaa : SUCCESS in 21m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/39267d8e72a848aeb3be685d6c2f178a : SUCCESS in 18m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c54fa81c34004b3989d7b40eeac764b0 : SUCCESS in 21m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/51c62ecd46b241e0b55934495e8370bc : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d2d8dc6709ca43419114ccc9aff5cc61 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c0be1aaa2d34bd9bca9471456878f34 : SUCCESS in 17m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7763e860c5a74447ba19a17c6ac24db3 : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/791a91fdd1bc40b584a48ec3db7b4ecd : SUCCESS in 17m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/723c24e9b16843f991875ce118dd03bf : SUCCESS in 20m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/11db96fe4d994f28b6051b544b542c0a : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/71faf95d70b74a03a1ff31ff90599a70 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e1694a4233ca440a89b4b0c1f00b315d : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39e3e8670d524ac4b85ae6e95e557929 : SUCCESS in 55m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f7f9564581e467b9746863249d2e91f : SUCCESS in 48m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/034f43b1e5c541f38f0ad95fe7a89212 : SUCCESS in 43m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdd49567a6ad4cbe860f1dbe8a819428 : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32a8f0ad3324bef835b5f76493a0858 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f28795b4309b4a3489d4fe41c7a892d6 : SUCCESS in 21m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec0673f6c2484e03b25203fe657b7af0 : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":1},{"id":"97b6134a1e18ed7cce44ce5c479c5017b86442b1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-24 16:01:16.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00170496f03bf2e5ead63fc824a44ac91bd398f0","revisions":{"00170496f03bf2e5ead63fc824a44ac91bd398f0":{"kind":"REWORK","_number":1,"created":"2024-07-22 17:31:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/924667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/924667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/924667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/924667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/924667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/924667/1"}}},"commit":{"parents":[{"commit":"f5a63a1b4f12987d22cbbf613cd3d7673ae2eba8","subject":"Parallel distirbuted task container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5a63a1b4f12987d22cbbf613cd3d7673ae2eba8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-22 17:30:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-22 17:30:25.000000000","tz":60},"subject":"WIP: sq? alternative expirer parallel_strategy impl","message":"WIP: sq? alternative expirer parallel_strategy impl\n\n...perhaps simpler ?\n\nChange-Id: I18d2b1bc701d438036ff3c7ee337e55a8120b454\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00170496f03bf2e5ead63fc824a44ac91bd398f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00170496f03bf2e5ead63fc824a44ac91bd398f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
