)]}'
{"id":"openstack%2Fswift~924768","triplet_id":"openstack%2Fswift~master~I30305566d12a1b8be4c8bde4b416b798322a1385","project":"openstack/swift","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-07-25 11:00:58.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-25 18:28:37.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I30305566d12a1b8be4c8bde4b416b798322a1385","subject":"Implement context manager protocol for logging mutexes","status":"MERGED","created":"2024-07-23 17:49:18.000000000","updated":"2024-07-25 20:46:53.000000000","submitted":"2024-07-25 20:45:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924768-py313","meta_rev_id":"5f3c6503490a964d6e35b55a86e6b7f087be5226","_number":924768,"virtual_id_number":924768,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-25 20:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-07-25 11:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-07-25 11:00:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 19:07:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 10:34:05.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"701f5e63f528eba3f8dc714884ca8d51bcf6b579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-23 17:49:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb1faa3a356a75eb254c75c32fa873b8da6bf7d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 19:07:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19beb60b2d964dc3bcd0508f7222fd76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74abc227a05149148200e8d6d55524c4 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/45eb384424a346059d4d4cf5937741df : SUCCESS in 1h 07m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/61a3abf7bd0249acb03ca7e049adf87d : SUCCESS in 55m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/787eac42695c449c8c2ccf6ad269cab9 : SUCCESS in 1h 00m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3f1be6aefd4468dbcd60ae471583cbf : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f6661591701488c8a362ebb5f88d316 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a0501268c324d6692292db22575e363 : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4fbf36af84924ea080569cb50b0db0dd : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33df688ec9f1492189707f2b240ef2e8 : SUCCESS in 19m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef878a781c684770808aeaa524f03cf0 : SUCCESS in 18m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b625b00660f64c7da4aa9850ee1bc123 : SUCCESS in 18m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ad1c1e74639c4c16a8601920259fc2f6 : SUCCESS in 16m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6da813913af84184ade653ff76f9f62d : SUCCESS in 22m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3193d93d2057460ca037fdac746d2af4 : SUCCESS in 18m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb8cb87c958a45eca05cc4d9a5653644 : SUCCESS in 21m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/602793631cfd4022bdaa2ca4259dd63e : SUCCESS in 24m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e109024ba1e54299b51ca536ed6814d8 : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8be894b6df4842cba5e32815b1c74b75 : SUCCESS in 18m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c6f3527d5754c4b9b803c60ebfd7c1a : SUCCESS in 17m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/285bab6638fd432cac02308acd1ece4c : SUCCESS in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a46d2bafb2a84b3d8aa01ea3a24907cc : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fd4b5f7170af42d4b0e07ff2090d5101 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca7d77b57bab4589a76f2a1431403aae : SUCCESS in 50m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1cf181d07804918a52b7c66d526b614 : SUCCESS in 51m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/16b4be3be6594a339027f6365a90b43f : SUCCESS in 36m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20963ca9145a45d4b0e9c4f18534c15d : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f95d7da11d346638bea01d0725c3064 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62534458a7ea4d67801718251073dd82 : SUCCESS in 17m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14101bdde65848ff9eeaa111fc100760 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ff46d527890dab511d77558e695071dbf306c7fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 19:22:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90d772473e4a4e818c5f12f317d75eaa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/edbe2056e6004866bd187c4e37f770e5 : SUCCESS in 18m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/314fa5bd048440bbb0af2721bbbcd1e0 : SUCCESS in 1h 19m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d64e576a548d4650877c8cde46c39437 : SUCCESS in 22m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdc5f015123e427a9abe158e30ce40b0 : TIMED_OUT in 1h 31m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7d39bd233b1b0029dccea6bc9c609a9dcad52932","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-24 10:34:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"378d43f5ff9bc4e7748fef126e130d7ac70fa66d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-24 16:44:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"68b989e6ae18ac823898b7e8bc93ea57884f1f53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 17:10:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22f6d898d23244368a77b616a73d6697\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af78ff3ad3eb4091ac9e84f0312239ee : SUCCESS in 17m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60804e5d26ad46c587519ac17527cbbd : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/031e974403954cba9dddd1e4a85ac854 : SUCCESS in 25m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fafbd55ca4474f0d87280540c3da7fba : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":2},{"id":"61356c307db5370dce93470ec4041a5c1db19591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 17:42:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd3ec0cac4e549ae9503bab49c0b1032\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4a8229f7434977823b0102722d0c69 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/fd9b42533c29423aa096c55bad9a2f43 : SUCCESS in 56m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0930982572cb4c5ca13bace280ff78b0 : SUCCESS in 49m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c12b92da280645c2a97bfdd9fbd45488 : SUCCESS in 45m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b52fa46c2df45e0ba25271af7f92d98 : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac1bd9337c79483cab7a3658e011df0a : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef1c5830481342f9bce95ce9a9e54f37 : SUCCESS in 20m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0efec96dc6a44895a291d7cbc2c26e22 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e5e39bd1850e4917a6883b84e13f6619 : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f69d66f2647493ebbf1886394d663a8 : SUCCESS in 18m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aaddaed1fd6c4e05816a130685bdcde4 : SUCCESS in 19m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/86a2d10488cc486bb451660c26976445 : SUCCESS in 27m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/453511d38c764732bcb3141590f30796 : SUCCESS in 28m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8589d2b691d14925a57056ba676136d5 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05b6f77f2d3a4204a56c5e0bc00c5076 : SUCCESS in 18m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/444550170a784555954420c8a1193163 : SUCCESS in 21m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0701aef432084528bb8486f1f1cb9ee3 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7f3ac13afb6a45c5ac64c8e158c4f429 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a6448cff6306446b91fad120c908138d : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f196c5edda4483abb0b8247dcd1eb2d : SUCCESS in 7m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7bec936aa55a449182761ad59516569e : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d821dd9c5ac34f6493d77ace9363033c : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c58d8920b92e4a3499bce6075c66e7cc : SUCCESS in 42m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/798e6ae430ec4263ba76bbf12692a0dc : SUCCESS in 57m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e54270f1565349ddab303c203a63eb11 : SUCCESS in 51m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df226851989140d6bbc557ca8f0f2610 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46f2f03654084fb18eaa5476a99bca32 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92179bf5300245bea187653fb6c043dc : SUCCESS in 16m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bee0350f4274432194d793ab399271f7 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":2},{"id":"abc367bff9d68d523607c74294ee4cb23c3baa71","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-25 11:00:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb5be6d7b944a90c18e902af786c01c77dc93fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 11:01:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14240d7506d73c0b0804fbc6eba6a8c4b70d1702","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 12:23:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eaee4f9ef617406a8d748b51e7759df3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b76eeb2778d452a8eb5803f5ac0a5e5 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/9a1b68d92dac41eaaa3436ea858e4ca6 : SUCCESS in 1h 07m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d286671eb0c74ec58fdfabd503f0a2a3 : SUCCESS in 45m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/237b402825ec460dbe89e202130174ce : SUCCESS in 58m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6738eb83559840d283af60274585ec39 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e8158d1bc964d4ea5e33f17215fe9e5 : SUCCESS in 20m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f5ce1a929c144be98dfa7d163c660f8b : SUCCESS in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f9ef75dba37248e0ab21c8567ee49284 : SUCCESS in 21m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7788e66721734b319dd54e0a4e284f49 : SUCCESS in 28m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2c712cd83ef44af68ce51efc992f4cb5 : SUCCESS in 28m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2e74193b1b9948d78eee51e1f2df1145 : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/289497d7000d44db9656a3fd44414712 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60cd9330c53849c0b63198281a1f98bd : SUCCESS in 19m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ddcdc1f2a2d42cfa420b94ccee3a106 : SUCCESS in 20m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c4034f3059dc4c97b6f3548c087f0384 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3e1fb1032ec747eb9ae87422b6217e47 : SUCCESS in 19m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce5f4ea89a2044f0a8e533c3757378fe : SUCCESS in 6m 31s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fe7b9c5639dc475b9ed63bc2283029a5 : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c065f47c380e4f34a538ccdced0b0cf9 : FAILURE in 48m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f429742497954a408944613fc5e33435 : SUCCESS in 51m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f3d8747a755c433298a77d4f29d57ab9 : SUCCESS in 51m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96f46de88fc1431ea532ed6f384b7b51 : SUCCESS in 19m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a5fd71f1504ba58a7078af0f8a5053 : SUCCESS in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5df243d875f43e7a188e7d8d87dabc5 : SUCCESS in 20m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2adc5ec3edf8410aa736eb440fa5659a : SUCCESS in 48m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4a50e8686ea3f30b56009d943c51696d90c0d47f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-25 18:28:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e06a4204de8e3ae65e5f502641cf0fe63df0b5ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 18:52:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6527ae1f0fd4f0b925e30cf282a7f94\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b12c16a90b5c4664a2a8b6fa1cf03a90 : SUCCESS in 17m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24f6aae5975e4787a598776f5b98e31e : SUCCESS in 19m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/55187cbd5f2941ddae28aacb07aeeb85 : SUCCESS in 22m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b3f3d70de1847298ee82dac0737940e : SUCCESS in 18m 07s","accounts_in_message":[],"_revision_number":2},{"id":"c82abd3e44e52c424d2ef35494357f14b2fe0f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 19:40:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daeb937cc9db43ffbce543492314b663\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aada980755184e4c846a9200daef9b0b : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/391afde8f9214274b526c5c22cfdf1ec : SUCCESS in 1h 06m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1458b2763f0142a29fdf620a9ae5ec15 : SUCCESS in 54m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cfe262bd38f4ce6a9e75430197eb8f5 : SUCCESS in 59m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c052bb207d6543baa9d65b4e79f2ec8c : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93f8dfd971c945498640f22ab14ad7e1 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3ea5fea80e04a5881cdde6bae23335c : SUCCESS in 21m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b299464d1b34f07a96a0a3953344cef : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14435b07a861495a8216aae32e3f85fd : SUCCESS in 20m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5024a5bca91a4a13bd38deeb619ae01c : SUCCESS in 19m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ea7cf1b0db3a4b7ba7d56967d7a9ba70 : SUCCESS in 20m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6f13dbac22e543d3b1b11a0afc00dec4 : SUCCESS in 21m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a97c2f5f12644e2d8a7197e7c48665f3 : SUCCESS in 29m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eb0ed2defabf4a0ba0e051b6bef0b9b3 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/935dc14fe1a94d3d961c9c1f749c229a : SUCCESS in 17m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0c9b64bce6e4512aeac7c283828eb8d : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/de70c00028114f869060fe20e58ce941 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/392e508d235047609a4a2ecaa9306609 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/070728deece94943a02d65cb6838a8a0 : SUCCESS in 17m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb47cb5913fb453bba55627a3c7b0bfc : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf0f8aaf0f1e4c02b2c4337fbc5d7725 : SUCCESS in 14m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f21ad2091a1149269a0499f76fae6f9b : SUCCESS in 7m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5811f0c33982484aaa189bf60402e5d7 : SUCCESS in 50m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1eb27c2b7a8f481ea5dbd9a93011600d : SUCCESS in 51m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c616a84232de4180a8da05c5aa89a390 : SUCCESS in 51m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f555d5e37364628a7caeef8b6490b7f : SUCCESS in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5632c1023ed458e92459ef80a2ddee6 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/147278234d0048788aaa7018dfc41440 : SUCCESS in 18m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9950a61af2be4b5ba1431bacd9e3b616 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":2},{"id":"02fcb5a221e59748c1733a3451641d6a0132216a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 19:40:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c10e4f56e46d1d02101dc1bf0336a3a1bbc5ea0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 20:45:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8542ebca8064741a9e7c48e23cec57f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb3a41b0ee944e5b902ed2cee0030b40 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/2bcd4e49342a4b96bae74a435d859293 : SUCCESS in 46m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/17f4e376edc34665a928c14c93a7dd0a : SUCCESS in 53m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2c324d996ed46c982c7e57e37242511 : SUCCESS in 1h 04m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e1e65da629f544afaa86cf7af06baef0 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f7d8d2c43a8a4b2685cef481e0c80ce5 : SUCCESS in 16m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d54d9ebfedaa4c18a68aa0c454c895c6 : SUCCESS in 14m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/119e41cd667642d180aaed801e03f054 : SUCCESS in 16m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/381e16ad6cee45629ce9042ad06f60e9 : SUCCESS in 15m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bc46fbbfd4d54d828b798ff11e3c9099 : SUCCESS in 30m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fbe219abdbfd4bdeab175b80ff8c66c7 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc8e6a4f56b346c2bc24bc92b2e739d3 : SUCCESS in 21m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8e1cdb4e89c64a099ea85eaae6017aea : SUCCESS in 20m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c011e3c6ca984f2e9ba4bea3f1b107ba : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3086c8876c2f485eba9a898b41d09e8a : SUCCESS in 18m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5459962c24bf4a209183788b87a38e95 : SUCCESS in 19m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4e55c22062fc446c95561501efc12c94 : SUCCESS in 5m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ae62f19698f942a0a15765266d8edc2f : SUCCESS in 7m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba2ec9c563a74edeb81d874943f88cd0 : SUCCESS in 51m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d126083b66548a683f32d0c6b01e23d : SUCCESS in 55m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f047ecb5bd5f4da9817f735fa08d294c : SUCCESS in 55m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2a8f38083674c2c9130f630c49620de : SUCCESS in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02da1a36b31f4809b8bd4b0af958d438 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a10ea4d9ec347b896ea1ad29904293c : SUCCESS in 23m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07cf65d6bfaf43fe94cae940a2bbc317 : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e15fe86fb9f114ba26ea7ba2a766354e162d4ccb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 20:45:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5f3c6503490a964d6e35b55a86e6b7f087be5226","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 20:46:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8092a72fd3d245269f974117e914c9bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b4680211ac4da18bb17124ebe8fcd4 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/db4fe95d015e422a9f37fba3aedb9e46 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c7ab03e2d984d68b0d3330bb698242e16f18a30","revisions":{"ba21d77ecb66ee663841f87c483f69e104f01e0b":{"kind":"REWORK","_number":1,"created":"2024-07-23 17:49:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/924768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/924768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/924768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/924768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/924768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/924768/1"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-23 17:48:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-23 17:48:42.000000000","tz":-420},"subject":"Implement context manager protocol for logging mutexes","message":"Implement context manager protocol for logging mutexes\n\nNewer versions of python expect to be able to say `with self.lock:`\ndown in logging.\n\nSee https://github.com/python/cpython/commit/74723e11\n\nChange-Id: I30305566d12a1b8be4c8bde4b416b798322a1385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba21d77ecb66ee663841f87c483f69e104f01e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba21d77ecb66ee663841f87c483f69e104f01e0b"}]},"branch":"refs/heads/master"},"7c7ab03e2d984d68b0d3330bb698242e16f18a30":{"kind":"REWORK","_number":2,"created":"2024-07-24 16:44:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/924768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/924768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/924768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/924768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/924768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/924768/2"}}},"commit":{"parents":[{"commit":"7c12870068e283344f750b1a16d0b1f91cc9045c","subject":"Merge \"add object-count quota for accounts in middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c12870068e283344f750b1a16d0b1f91cc9045c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-23 17:48:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-24 16:44:30.000000000","tz":-420},"subject":"Implement context manager protocol for logging mutexes","message":"Implement context manager protocol for logging mutexes\n\nNewer versions of python expect to be able to say `with self.lock:`\ndown in logging.\n\nSee https://github.com/python/cpython/commit/74723e11\n\nChange-Id: I30305566d12a1b8be4c8bde4b416b798322a1385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c7ab03e2d984d68b0d3330bb698242e16f18a30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c7ab03e2d984d68b0d3330bb698242e16f18a30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
