)]}'
{"id":"openstack%2Fswift~924771","triplet_id":"openstack%2Fswift~master~I2b11487f8174e36b024c503d7e78de1904e0a281","project":"openstack/swift","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-25 12:00:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2b11487f8174e36b024c503d7e78de1904e0a281","subject":"Pass timeout as kwarg","status":"MERGED","created":"2024-07-23 18:01:15.000000000","updated":"2024-07-25 12:02:09.000000000","submitted":"2024-07-25 12:00:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924771-py313","meta_rev_id":"2b6e64ea6994b6eb9e703d197eac5690a7ed5a40","_number":924771,"virtual_id_number":924771,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-25 12:00:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-07-25 10:56:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-07-25 10:56:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 19:10:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-23 19:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 10:34:23.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"a1931dba8a571e5fec000ad981bb3cc948a50538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-23 18:01:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09a76501b958e6a195938427d44a8e26ea32628e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 19:10:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56b19f772fa24d69896c85fb88f6e8a4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3ce0dcdc535461cadc1bb26611fd277 : SUCCESS in 17m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6b13565279a4c03a507a82cf304f780 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b67a4c935084dc29d4bc2e9021adbb4 : SUCCESS in 25m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc5cdf5fe84047fdb2cd22c8f47677af : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":1},{"id":"228e1b4a737fbc3f83498fa5279a206cb148e3b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 19:14:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ba532a0f09486b88d8cc941e1b5a27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4334c0974d1a43b8885fc588eb13803d : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/f491bccaea2347029c3c8613cd041642 : SUCCESS in 47m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae2af9e1b35744638f561d44f953406d : SUCCESS in 58m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfea6fa32b014285b805ed4aa1269556 : SUCCESS in 47m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40163905c9ea4017bb646f3c248ea42c : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ede4ec4ef5a40daaa077e1a537f88db : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f052ba08db384aa5b0d9860c05fb8445 : SUCCESS in 24m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/799745d6a3e240ae9d33d180ad0b12e7 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9bb7bc598bef42fd82e3a6a913f9bde6 : SUCCESS in 19m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57463acced054e12a53c6c562b86722c : SUCCESS in 20m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0726b152eb0240089412f920f7c85d1c : SUCCESS in 19m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/42846ae4cbc0469b8deb36e7180f8826 : SUCCESS in 26m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5331f3c87be0457a8bb631f9d3de69c8 : SUCCESS in 27m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/54ef0a22fb44469b83a0e6226476c002 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/711fb61874e34859ab260cfb0f849424 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b41f6d3a94f5488e9fd90732122d461c : SUCCESS in 19m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6ff8a9e07e9043708c2e9e9d7d25e312 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/29584c7926c4438494d0faaf180948c7 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/56a94a2248444363b96877531476b0a6 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/004112f4e7e44855b53355331a2f515d : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0ec514469c24c12823c42cfda84743b : SUCCESS in 13m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b7d25e07bf04c4ab419075bbe6ddbd5 : SUCCESS in 6m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/abde16bda945415d8654d2d30c6621ff : SUCCESS in 52m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24c88185576a4e9a88119749dee66a87 : SUCCESS in 48m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0003bded779f40508356d0245b40f70d : SUCCESS in 53m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01cca73506ec4c9bb45414e973c767eb : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ca9a382a37a44848b308f3e7f9b7164 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0fc007328b574680aa6827942deb32f8 : SUCCESS in 23m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa14c41529bd4dd2b61c4c244d11e265 : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":1},{"id":"444b2191b7b48518fd68cd79022a79cb073096c4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-24 10:34:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0155dd32ba38761af291ae41e5802f238501a45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-24 16:45:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"31af56f392640b9971f6595466b68394362c93d1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-24 16:45:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f78c06f1650d152fa16fca1d423548bb32225e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 17:41:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/642e18032d7541cb9e5ecba95a3c8c60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b85e56aa3ef4d11ac38768cc89f16b7 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/9c213a83d4b64fcd87875c5869eccdf6 : SUCCESS in 55m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c12a66bf04e47c4adde08923e3bbf72 : SUCCESS in 41m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c37be0d4b2b4d2e88125edfab89b830 : SUCCESS in 46m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d699631a9174a6ba8a80074b7350501 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7ba1e240d104c39a94d1d819d80084d : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dcf17be5591547e9a45847887b7bce05 : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1823bdb945e4f638c25bfd6ff3ed74a : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a2b34bcb9ae34c29979702a6d8641584 : SUCCESS in 21m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81570dd6d1c0445bb67283ea4e63967e : SUCCESS in 17m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dd8bf1b9c9ab4148bf35a0e86f2cdbe6 : SUCCESS in 18m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2ead6fb961c04383984b31c48805097c : SUCCESS in 17m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a350619baf754f8fab62d6558ab8255d : SUCCESS in 31m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b25831760b14befac73c785019e425c : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b587f5e2299442a7a04cce136975e54b : SUCCESS in 16m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e90c2c170cea4c41a5cf50a6302a5086 : SUCCESS in 18m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1845144cd69d4143b440b2006e44f502 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c88c4d21482c4692a01f93c8fb00d125 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fbc353eb15b143008f919707155caaa3 : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a23f14d1b7b44217a4b6621fae7e597e : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a88e7e193c4c4631858ab42dd18cdf2b : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2dabed7d3db64737ab2e2985a435f89d : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c7f9cb9fc7643c4b3cc7a57b9dadf10 : SUCCESS in 49m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/441bdbd23d8e43b6885a894f56ad758f : SUCCESS in 48m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/415d2f160ac54ae0a73b8f8dbbcedb05 : SUCCESS in 36m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4207f19529ea4dce9263c852f52662da : SUCCESS in 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c57b2ac2b934f34bb8bc7df15788ae3 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6ff09b6b201c4b1eb49e5447e8a5b1fd : SUCCESS in 16m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b7a98f253424beb8e4c4e4fb0edba6e : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":2},{"id":"81ee5f9de024e50ef7ccd356ecca64c8895ea6ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 18:06:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd21cb0ab624f98b1b918c19c29cbda\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2260ed9afa0b4e0bbb7fea100675f964 : SUCCESS in 1h 19m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85d4577aa798403093104a311ba47210 : SUCCESS in 1h 10m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/767c4cb1edf744ed9f61fddca4a9b6eb : SUCCESS in 22m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/05e5931350d842d494f167bde96a6dd4 : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":2},{"id":"5adee986b5492b897934e16486178e9f8ca01984","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-25 10:56:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8a1c6d85712027312e21a953b34ef366441fcadc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 10:56:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60732f9f213b6f992cb38da8c189198ef9e48939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 12:00:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db0bdeab0bbe47f08f658b21d283c70b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa48a1ca46574cdb88e753662918c06f : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/4decf5994fa44d89a68e9fc96100ee6d : SUCCESS in 1h 02m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b6ff911f2ebb4b23ac122913c1991190 : SUCCESS in 49m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a66b6a80d7247b1b86b217b1fee4b9e : SUCCESS in 59m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e8e64682311c4a8ea764315553c10119 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/49fa7016bdf949eca8296687fac9eac3 : SUCCESS in 20m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74a50ed1c9024719856cc337b8f498f2 : SUCCESS in 20m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/49227543bf434b5f86ebbc272e5e469d : SUCCESS in 20m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/25842047436f49b4a0fa0ca63b25b7f3 : SUCCESS in 15m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1ca6fb20866c457c9b0bb3f3565372ae : SUCCESS in 29m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2e90d21f7f4432898a474a0e07a2c62 : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b8a5d3237484c778774ac6cd38e204e : SUCCESS in 20m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/105419945a3c458d9819146b6c90a16c : SUCCESS in 20m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc9a7909c8ba46258454e0fec89e70f4 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/91c5413ca9c7402fa4bd9a7cf189e738 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a919588c221c4747933a910cd5f943bb : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e15065dfc11e466586398c1d8bddf704 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7fee00edf3404bcf817218552bfae267 : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebcaf9e763a84f3498adc875c4c0990a : SUCCESS in 54m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c84d0985ea654bb298449ee9574f5953 : SUCCESS in 47m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36c88d2877914c1eb0cac78ebd1e210d : SUCCESS in 53m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bdf42e6f85b41cd9893ca7a6252c6b1 : SUCCESS in 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba5e9857d8ef4055a6c29c8e0bf14222 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a767f15103d4d91ab9d8ecaca8c5445 : SUCCESS in 25m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95a9b80bc7ac463cabcb6cfa4ce59741 : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ec4f8c3d304b99cc4c4a755e96ac07b2a349d942","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 12:00:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2b6e64ea6994b6eb9e703d197eac5690a7ed5a40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 12:02:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42d60e0c77274f0aa68343ce8db3664e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1aeadaa1921408b8a392735ef0ec067 : SUCCESS in 1m 14s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d998147c6c9545149ec2e677e077507b : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd7861313f113652df10bcd31ff5ce79c09007d8","revisions":{"f5d19d98da54cb5b44f742319006d1db62b436e5":{"kind":"REWORK","_number":1,"created":"2024-07-23 18:01:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/924771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/924771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/924771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/924771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/924771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/924771/1"}}},"commit":{"parents":[{"commit":"ba21d77ecb66ee663841f87c483f69e104f01e0b","subject":"Implement context manager protocol for logging mutexes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba21d77ecb66ee663841f87c483f69e104f01e0b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-23 18:00:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-23 18:00:22.000000000","tz":-420},"subject":"Pass timeout as kwarg","message":"Pass timeout as kwarg\n\nCleans up a deprecation warning:\n\n    DeprecationWarning: Passing more than 1 positional argument to\n    _sqlite3.Connection() is deprecated. Parameters \u0027timeout\u0027,\n    \u0027detect_types\u0027, \u0027isolation_level\u0027, \u0027check_same_thread\u0027, \u0027factory\u0027,\n    \u0027cached_statements\u0027 and \u0027uri\u0027 will become keyword-only parameters\n    in Python 3.15.\n\nChange-Id: I2b11487f8174e36b024c503d7e78de1904e0a281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5d19d98da54cb5b44f742319006d1db62b436e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5d19d98da54cb5b44f742319006d1db62b436e5"}]},"branch":"refs/heads/master"},"fd7861313f113652df10bcd31ff5ce79c09007d8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-24 16:45:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/924771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/924771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/924771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/924771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/924771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/924771/2"}}},"commit":{"parents":[{"commit":"ac5bec8889f131c6ba9a763f9cde6fcf8f88c514","subject":"Merge \"AUTHORS: update Zaitcev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac5bec8889f131c6ba9a763f9cde6fcf8f88c514"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-23 18:00:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-24 16:45:12.000000000","tz":-420},"subject":"Pass timeout as kwarg","message":"Pass timeout as kwarg\n\nCleans up a deprecation warning:\n\n    DeprecationWarning: Passing more than 1 positional argument to\n    _sqlite3.Connection() is deprecated. Parameters \u0027timeout\u0027,\n    \u0027detect_types\u0027, \u0027isolation_level\u0027, \u0027check_same_thread\u0027, \u0027factory\u0027,\n    \u0027cached_statements\u0027 and \u0027uri\u0027 will become keyword-only parameters\n    in Python 3.15.\n\nChange-Id: I2b11487f8174e36b024c503d7e78de1904e0a281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd7861313f113652df10bcd31ff5ce79c09007d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd7861313f113652df10bcd31ff5ce79c09007d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
