)]}'
{"id":"openstack%2Fswift~924780","triplet_id":"openstack%2Fswift~master~I86f10fcf4fd0bc5902c3dad49c6f3019fa159b64","project":"openstack/swift","branch":"master","topic":"zaitcev","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-07-23 23:40:02.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-23 23:40:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I86f10fcf4fd0bc5902c3dad49c6f3019fa159b64","subject":"AUTHORS: update Zaitcev","status":"MERGED","created":"2024-07-23 19:57:11.000000000","updated":"2024-07-23 23:41:40.000000000","submitted":"2024-07-23 23:40:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924780-zaitcev","meta_rev_id":"1b72baceafc253892bfde147230fd3319aaeabdb","_number":924780,"virtual_id_number":924780,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-23 23:40:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2024-07-23 21:26:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-23 21:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-23 21:26:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 20:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-23 20:45:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 21:26:21.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"48acfd4b81560083315c800316febe11bdfba809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-23 19:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d87876a1120d46d7692266737be161a5f2974c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 20:23:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbb7436986dc497bb71c367fb2343b7d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24b31ed4578c4b49bf4548b6c51be405 : SUCCESS in 19m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d3edc8b5f9749ba80b09764d0deaa1c : SUCCESS in 19m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4a1323157bb44f38863dd38c8572a63 : SUCCESS in 25m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfe70f13604f4f598bcfac7b28c59ede : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":1},{"id":"860b59e40619c2c0a531cb1b129b992d87a70a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 20:45:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c435eea82a284cb9b7d086606c78f047\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c82e48688fd490b97d25263e7238acd : SUCCESS in 8m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/593000c057624b5aa4ad61c9218cf4ab : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6bbd78cb1ea345c699c4701424b90cb2 : SUCCESS in 21m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80cf9a3a7015414fafa181eb24a7176f : SUCCESS in 21m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a0714724090248b58d61ef048552e372 : SUCCESS in 21m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0b1fa18ed4304c4d8a1453162c77180c : SUCCESS in 27m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b93bbb37cb7c4ed6bb76c6c2cda215ba : SUCCESS in 36m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/583d9a8b56f24f26b3a9d54eb800fe5a : SUCCESS in 25m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b77ed5efb1f4a39997fcfb088565e4e : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2d06778a03941ebab92e5e0995970bc0a906a0af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-23 21:26:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c13ddf11a718b8806eb2d6407e4f51f24775b75b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-23 21:26:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0bb5af6b4d3bedbbb2e59ff5e2158ca7d2ad7139","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-23 21:26:47.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58d8e2162eb2e750e60a3e50b0c4673b48d4f45e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 22:05:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e60afd6f99b41b9b0e2c4e61bd6317b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad3abdff2334f9b8915a6b0c106f392 : SUCCESS in 8m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0add172823aa4b49be02a9dc9d61c181 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ec446cb26f324c73abdceb47ef45757c : SUCCESS in 20m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a37dd90c11c4d5380db11fbe6f0448d : SUCCESS in 15m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c121e07e269e43e9ae2b9945311952e9 : SUCCESS in 22m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6f70b58c6677438ca66bd58c812ff349 : SUCCESS in 16m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/749a109b900d4aee9e6d4fbe0710bc35 : SUCCESS in 22m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fddcb151d6fc4b8a81e5f99f966f440a : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f892557971cb4f2ea7457289993fc29c : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"be90b923c5d5bc32008025ea948c12d9130cd9c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 22:06:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b018bf4313ef9d4861e8707c7b113145a1d27e53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 22:35:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98f97a9045b7467bb7c286899a1a2c8c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31cde18dad944c9686031e9f9c945a76 : SUCCESS in 19m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c94494f55024d888733715ccc67f58b : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d22810a75a0542e1962748dc25128bd8 : SUCCESS in 26m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/08d29d3fd53d4b298054d85cdcc38953 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a54517e4940576a534d87c1969921d31e79e955e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 23:40:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f8f5e0f18e44477ba49c6cb0334e4c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f49b85f4a70e4555b8048fb48904a958 : SUCCESS in 8m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/741288cbb90e404f89807525831a92af : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/df65328b41244db7bb250e062e00c44f : SUCCESS in 19m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d76ae6fe70a2421280dc514dd9923211 : SUCCESS in 12m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/77b473c5004745a89c7ed44f5f8d543f : SUCCESS in 17m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ae46a2ff8b1241bdb8c9e177dff7b357 : SUCCESS in 19m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e12ef3c375c14e42936bd32cf464f4b0 : SUCCESS in 19m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/908217c4492e4307a04c82f60045bcd5 : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3ed6470f0374d27a5a04d3596652688 : SUCCESS in 18m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/31d132542c4845889e935bf258356416 : SUCCESS in 19m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8cde1046ef1e44c48783d84e6cc0870b : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/38daf3e7f7254399b358fee2a8b04005 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8e7596f5baed487bbe12a45038fab5d7 : SUCCESS in 19m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7683e81744524f7fb93694593ff77d4f : SUCCESS in 6m 05s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cfac4bd6c2ca4fb287784b92cdeac657 : SUCCESS in 8m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2196c428fdd141aa97294e5f5dac3887 : SUCCESS in 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381329c312134db4b5c3da24506a3d2b : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"524291cfad596b55cb039b9deea8fe69c6c828b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 23:40:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1b72baceafc253892bfde147230fd3319aaeabdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 23:41:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61a8314c6df145b989fa2f2137cb41e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887859918840489db4cbcc80b219bfc6 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a3b762b33fe64e218c6d79bae02ad9b6 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dd86cecac1b1afe2255d008a521a2c6d9577d931","revisions":{"347a2400a24e5bc88ee766ddd556f28f5ba96b85":{"kind":"REWORK","_number":1,"created":"2024-07-23 19:57:11.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/80/924780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/924780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/924780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/924780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/924780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/924780/1"}}},"commit":{"parents":[{"commit":"533a601cbfa83ce1ad007b3bd83ab202602425e9","subject":"Merge \"swift-drive-audit: Work with ISO timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/533a601cbfa83ce1ad007b3bd83ab202602425e9"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2024-07-23 19:55:53.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2024-07-23 19:55:53.000000000","tz":-300},"subject":"AUTHORS: update Zaitcev","message":"AUTHORS: update Zaitcev\n\nUnfortunately, kotori had to be switched to a whitelist,\nso it\u0027s not appropriate for a public address list anymore.\n\nChange-Id: I86f10fcf4fd0bc5902c3dad49c6f3019fa159b64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/347a2400a24e5bc88ee766ddd556f28f5ba96b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/347a2400a24e5bc88ee766ddd556f28f5ba96b85"}]},"branch":"refs/heads/master"},"dd86cecac1b1afe2255d008a521a2c6d9577d931":{"kind":"REWORK","_number":2,"created":"2024-07-23 21:26:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/924780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/924780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/924780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/924780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/924780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/924780/2"}}},"commit":{"parents":[{"commit":"533a601cbfa83ce1ad007b3bd83ab202602425e9","subject":"Merge \"swift-drive-audit: Work with ISO timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/533a601cbfa83ce1ad007b3bd83ab202602425e9"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2024-07-23 19:55:53.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-23 21:25:42.000000000","tz":-420},"subject":"AUTHORS: update Zaitcev","message":"AUTHORS: update Zaitcev\n\nUnfortunately, kotori had to be switched to a whitelist,\nso it\u0027s not appropriate for a public address list anymore.\n\nChange-Id: I86f10fcf4fd0bc5902c3dad49c6f3019fa159b64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd86cecac1b1afe2255d008a521a2c6d9577d931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd86cecac1b1afe2255d008a521a2c6d9577d931"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
