)]}'
{"id":"openstack%2Fswift~924820","triplet_id":"openstack%2Fswift~master~I1aa96e97281421c69cb75ea24ce427c20c4c615e","project":"openstack/swift","branch":"master","topic":"py312","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-08-27 19:29:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1aa96e97281421c69cb75ea24ce427c20c4c615e","subject":"account info: Add --sync/-s option","status":"MERGED","created":"2024-07-24 06:21:54.000000000","updated":"2024-08-27 19:31:07.000000000","submitted":"2024-08-27 19:29:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924820-py312","meta_rev_id":"78183d6143a2ce94e994d800a021e076b09adbac","_number":924820,"virtual_id_number":924820,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-27 19:29:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-27 18:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-27 18:30:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-24 07:19:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-24 07:23:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-27 18:30:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"d10cb0bc669e4efd55740102136e74f32c5cbfdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-24 06:21:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d136d619506b1dfb38a2800f3e4eb0ccfb1732c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-24 06:24:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"837f9037d855b1e6f049cbb33406b799ea4b06a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 07:19:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318a3bbe90fc41d68330cfff3fd04d9f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f253e4cf26d434783a98c3917c207c3 : SUCCESS in 16m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/129e18eeb7cf491aa8070485959188b1 : SUCCESS in 22m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/27dd8479c9a34a3bad3961c42b4ab287 : SUCCESS in 21m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3daf1f3b96064b7daa9643a142207df3 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2125c17a4bb89620920814dd05e52ada6919bef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 07:23:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43987143d34b41b484ac2c66985d91a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6894db9a48724c0498f3b20555ffdcd9 : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/5197863d49304773a53004d07f4aaa9c : SUCCESS in 48m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dad35b24eb3d4539a5b8527928c5a092 : SUCCESS in 47m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/023467d8ea12418c87cb5322e8942b78 : SUCCESS in 48m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/664d782153ef4bf5a51f2c7b3dc66153 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc86de531cd1485aadfe5ec6418a47d7 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76fa8109394241fca8d6d4d55853545d : SUCCESS in 20m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43673d2c65294def97cb7fa2c7bbb37d : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4fbe702a4a7a4818ae420aab0af3f30c : SUCCESS in 19m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f94acde210f46479cb3713d77197f0c : SUCCESS in 18m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a5a231e84020438da8192c5534fc4111 : SUCCESS in 20m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5ddca521688143f28b11516c05eaa0f5 : SUCCESS in 14m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9b45a7509f924909b2e603bc5f538511 : SUCCESS in 33m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c5674df0ec3b4a0096b867d46c054ef3 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2ee2169d4ce949c2b7080624aa134f9b : SUCCESS in 18m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6b76bafa7cff4a499b634a54729b0ba4 : SUCCESS in 17m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cca64d5b889d4960aeda12e0b40a3104 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d9a28028c2114923aca7ca53793c2d85 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cd3eb47f70e34deeb78df3af7d3aea17 : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ae58550dc5e14f858f238cebd68e27fb : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6ef2bcb9fe943efa2a9b862de5e7e17 : SUCCESS in 12m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/442b11234a9e4100b0a780fcaf7c0219 : SUCCESS in 8m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/451566fa424b407fa1b0364c67d89508 : SUCCESS in 52m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6281144fc8594ef99d0b20e48612e7cf : SUCCESS in 47m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c47ed11e7332441daee5aecd4706afb7 : SUCCESS in 46m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1702fba51e7443af90868125792f3448 : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52f12089b6ed46aaa142433845b60136 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/899188c29c174ac0903d693b045cd083 : SUCCESS in 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1268f0b586fc499fb6fd957fbdb9b810 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5c870dcb9143bafb90b8c8b2165bbff083083e00","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-27 18:30:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee88e0ed7f9ccc11513f5ac69a3e1d8158057388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 18:31:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95fc7504d4177c951e05c40eae1a08176fc76186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:29:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29422bdb641441ea8e2e8f0cc0e0309d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6049f6c488f41a88b097b7ea093132e : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/ab65f7bd408e41b79ad8bc7e16208524 : SUCCESS in 53m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/496b834e98474fb3bbcb92f8650194c1 : SUCCESS in 53m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/757b9aec7e4a453f9ba6b661023d4ddc : SUCCESS in 55m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/95526cc94beb41a180fc21fbcb33ada9 : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dfc923120a1346eda796fe75bcaddcfa : SUCCESS in 19m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e7dd050224048f59cab332a560f6d8d : SUCCESS in 16m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b70f288d5f1d4121a1b0777b941f53d8 : SUCCESS in 23m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e454e5b729a84e739f6a6a7fc1ca4791 : SUCCESS in 23m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9bf27a0b4c0e4770907d574c69e470de : SUCCESS in 34m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/933a0da0cfae4d20ba0cec85837ee676 : SUCCESS in 23m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3444c9ba70c9442781a8f9900f0eb0bb : SUCCESS in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4a2b6125fcc421aac4de6f05a225459 : SUCCESS in 20m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b0ec8d3486794dc28637b08b773b356c : SUCCESS in 21m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/edeb9a9d7fcc44afbf74a675d6772bd8 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/04c87671c7624af695ec841efb4ad6ac : SUCCESS in 17m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6d66beef25464dfab92f00a204bb8eb1 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d739c424e7148a3a1be7309d3025e7c : SUCCESS in 6m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/05532795d77a4a729cd28fd4dd0666b6 : SUCCESS in 44m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eec1cb039210440c9df31df249e45134 : SUCCESS in 45m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/af6dfb10813f4784ae38d878e83afad3 : SUCCESS in 37m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1539bfd4acde4d64817a34c37d484fef : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7301a63e198c48868afe32df9f591b04 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8d1c4a5ea6ae4b8d828d9645ad65f203 : SUCCESS in 15m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e32242d12744415d891f1233d17eab17 : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3f07ad97bd89708609e02ded7208478898f2d1b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:29:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"78183d6143a2ce94e994d800a021e076b09adbac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:31:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8be08acb1d648899dd632cc30b32178\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9cc6a45de3a40a4b58dd88fed18b674 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1f309e067d834f94bc9521d5c57cafee : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78278c11a63d8abb697c37a0d990dc5b6e08396f","revisions":{"78278c11a63d8abb697c37a0d990dc5b6e08396f":{"kind":"REWORK","_number":1,"created":"2024-07-24 06:21:54.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/20/924820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/924820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/924820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/924820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/924820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/924820/1"}}},"commit":{"parents":[{"commit":"2c5dc64d2562030ba4976b1c88462e1ebf8717c4","subject":"Pull swift-*-info scripts into swift.cli.info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c5dc64d2562030ba4976b1c88462e1ebf8717c4"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-07-24 06:16:17.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-07-24 06:16:17.000000000","tz":600},"subject":"account info: Add --sync/-s option","message":"account info: Add --sync/-s option\n\nThe container-info cli already supports the --sync/-s option to view\nincoming and outgoing syncs from the db. Account works exactly the same\nway. And hey be might want to interrogate who an account has been\nsyncing with. So it should have the same option.\n\nThis patch adds the same option to account_main parser:\n\n  $ swift-account-info \u003cpath_to_account\u003e -s\n  Path: /a2\n    Account: a2\n    Deleted: False\n    Account Hash: c1e207026da71a9237c5d9e206b9ec59\n  Metadata:\n    Created at: 2024-07-24T06:08:28.760490 (1721801308.76049)\n    Put Timestamp: 2024-07-24T06:08:28.743650 (1721801308.74365)\n    Delete Timestamp: 1970-01-01T00:00:00.000000 (0)\n    Status Timestamp: 2024-07-24T06:08:28.743650 (1721801308.74365)\n    Container Count: 10\n    Object Count: 100\n    Bytes Used: 880\n    Chexor: 0673fe17a762662d74e8d3ad93e0785b\n    UUID: 705ac602-5e85-49ef-92a4-3082d40cad4f-sdb3\n  ...\n  Incoming Syncs:\n    Sync Point\tRemote ID                                \tUpdated At\n    10        \t285bbaae-90da-409a-9533-f0d113764e44-sdb4\t2024-07-24T06:10:34.000000 (1721801434)\n    11        \t80c45862-dcd3-4ffe-9659-91851276d244-sdb2\t2024-07-24T06:10:35.000000 (1721801435)\n  Outgoing Syncs:\n    Sync Point\tRemote ID\tUpdated At\n  ...\n\nChange-Id: I1aa96e97281421c69cb75ea24ce427c20c4c615e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78278c11a63d8abb697c37a0d990dc5b6e08396f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78278c11a63d8abb697c37a0d990dc5b6e08396f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
