)]}'
{"id":"openstack%2Fswift~925406","triplet_id":"openstack%2Fswift~master~I6663d1d469c9bffd6d5c431776639b25a0227b97","project":"openstack/swift","branch":"master","topic":"daemon-init","hashtags":[],"change_id":"I6663d1d469c9bffd6d5c431776639b25a0227b97","subject":"wip: use Daemon.__init__","status":"NEW","created":"2024-07-31 15:10:16.000000000","updated":"2024-07-31 17:49:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cbfd3bd3fe1a95bc0a74f60d9b7c31b91951e4b4","_number":925406,"virtual_id_number":925406,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-31 16:07:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-31 16:07:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f82386908506e46f44cbbcdbbcfc4c216a570709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-31 15:10:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"858e2b7e6d69779dadc362b4985c4af260cbb178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 16:07:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac382da521ee48208eff68acadb713dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e1e630de414751b5e15f8ec2e8775e : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/5825f0707e544deaa52430a8844ef1e2 : SUCCESS in 54m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db48b18161eb43cf895bba8083b0bfe3 : SUCCESS in 41m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d632bb91918f402082c1f42fa7f806cf : SUCCESS in 47m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b32cba261384c5e9f15de9134019bbe : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cee288f5b2f4cecb1c9227225ece5ae : SUCCESS in 21m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7154226650a441d953265479e05bde7 : SUCCESS in 20m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/459d64f37d8c42e68d4cc00c6c9de740 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7202757307a94173b448c493c7b95fd9 : SUCCESS in 20m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4a30ab860b34eedab228514a024c4d6 : SUCCESS in 16m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/920a8c8bcf374a0db884dd46864494ae : SUCCESS in 23m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e870d587f02e4f0a84eb42cc1c2476d1 : SUCCESS in 21m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8f3736bc427d44c7afdf87e4f9a0d89a : SUCCESS in 32m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f9a1ee235d5a47939b852760426cfee6 : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b944bff560f4d05a2dd6432aee2b222 : SUCCESS in 17m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dba53cbc17be44fe9b66cb1a087b2a09 : SUCCESS in 20m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/558f0c8bfab0420c9a52039547149290 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/742cdd382aed4af7bce711d18e11f7cc : SUCCESS in 17m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7f4187ed18fe4301baa773b8f4f0378e : SUCCESS in 18m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30bc875a00254be68a31db6909a83192 : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb42b162bf43476aaaba0a6287bd9573 : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8fdd2028b30848f6ba7d16e07a8a024b : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3815051b849541d79199c7223c7180f9 : SUCCESS in 51m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf47739a2e5a44ee98a4309fc27e2f42 : SUCCESS in 49m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6626c7bd1444958b8d3858a158de8f3 : SUCCESS in 42m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c9cc68147544ddd976412ee5c77a517 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eeba3fdd3f54bb49c458fb44f092de4 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e5591a40da6445e7ac834a3de48a0fe8 : SUCCESS in 25m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c3badd9ceb74673804ac11e9b819201 : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":1},{"id":"cbfd3bd3fe1a95bc0a74f60d9b7c31b91951e4b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 17:49:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a90f69e803f4ddbbae2c10f3b0f6a48\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78e8622b1e454b66b2b4511867b0766a : TIMED_OUT in 1h 31m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3665e89c1a224dbea066ca1edc1c21c5 : SUCCESS in 1h 01m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/20cdb157733e4cc79cf08621fe590096 : TIMED_OUT in 1h 30m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/63f74892445e44d899cb954efca930f3 : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d27fb5b41f99032ed51464e5b14760c573bd5d6c","revisions":{"d27fb5b41f99032ed51464e5b14760c573bd5d6c":{"kind":"REWORK","_number":1,"created":"2024-07-31 15:10:16.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/06/925406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/925406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/925406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/925406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/925406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/925406/1"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-31 15:09:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-31 15:09:08.000000000","tz":-300},"subject":"wip: use Daemon.__init__","message":"wip: use Daemon.__init__\n\nRelated-Change-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\nChange-Id: I6663d1d469c9bffd6d5c431776639b25a0227b97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d27fb5b41f99032ed51464e5b14760c573bd5d6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d27fb5b41f99032ed51464e5b14760c573bd5d6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
