)]}'
{"id":"openstack%2Fswift~925550","triplet_id":"openstack%2Fswift~master~I4ff77ca6a94a6b9613d93082cd36a23929c94236","project":"openstack/swift","branch":"master","topic":"servererror","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-08-23 19:24:14.000000000","reason":"\u003cGERRIT_ACCOUNT_35790\u003e replied on the change","reason_account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2024-08-27 17:51:23.000000000","reason":"A robot voted negatively on a label"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-08-22 20:23:55.000000000","reason":"\u003cGERRIT_ACCOUNT_35790\u003e replied on the change","reason_account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4ff77ca6a94a6b9613d93082cd36a23929c94236","subject":"Server Error listing s3api bucket - Write test with xml escape character - Resolve 500 error with encoding_type\u003durl","status":"NEW","created":"2024-08-01 23:55:00.000000000","updated":"2024-09-10 21:57:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":36,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"074c581a66df1aad963f613b40122876854b717a","_number":925550,"virtual_id_number":925550,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-10 19:09:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-02 00:01:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-19 20:38:51.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"},{"updated":"2024-08-21 23:44:13.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"f655b473f44c2824be4f49aedecbbd8fdcdd1fe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-01 23:55:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d78cf2380ae9ec688fd63d89db234e94580fa3d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 00:01:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7ad344bd39b4735a6e3ca43ce274c89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2532d5058ce4119b042f101b9caf7f6 : SUCCESS in 6m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0c31702324e47a1aa5f9ec5629d9ae8 : SUCCESS in 4m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a4867e7b95b44feda8a2f043f565d7d5 : FAILURE in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170aa58de138420bb905920168d22635 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f34225d8fd274d352ca617c89362d2a8581b5452","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-02 00:08:54.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b392bade64d063e0abb95cb82d2a3b9cad2b8ff2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 01:18:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3ecd30849c4f638fe3cacf6f74016c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d11609c4ba654e95989259ba1576b046 : SUCCESS in 59m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6158210ba40740e0b54790e1be580145 : SUCCESS in 56m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/78e992444c2e427fa36f43fa3bd7e639 : FAILURE in 1h 16m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc476697dda8410d9c508edfbd643bbf : SUCCESS in 1h 21m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1b2acd0b7181db8793fa2bef6bc90d466fbf11be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-13 14:43:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d6956919c259e28a65b2b498003f73b94a4e0b89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-13 15:22:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa64b385a0a94dec090e0d78ba593f90d8a84603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 16:12:57.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7ac0242d89544378893db65500d573f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a14b691c9f22446c864b19d5aaac1e3f : SUCCESS in 7m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce548dbd30b44b4bb4a0901fcc2dd2cf : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9877f1d445d4d79b5420aef61997c4f : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf7a4557c6de4346947f08ff7a9d7dc7 : SUCCESS in 23m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/696d69ce38dc4829a79c650f4003ccb7 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ef4b52224624e48bd772c580dfece23 : FAILURE in 14m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3cf15448c1a4440a9c06e1c5383043a7 : FAILURE in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/972cb1d83eb84ee58aeb17549a1b0d3b : FAILURE in 12m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/63d7c165a6c14b72ba6efa20200075f4 : FAILURE in 13m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a9735670031746678b3348ab8a9991fd : FAILURE in 23m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de49574ec6c74bcab51251ae056178cf : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/348b4b79ecdc4581936cd7dec5f8deb0 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b0929b7156b54ea7a713bd1cbb15686a : SUCCESS in 16m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/716c8c8538a64ac39daf277fbe4a08af : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/49dc71cf78c44144bc1cc59d6cbd80ae : SUCCESS in 17m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0fcf79c56c3f4faa95b4a5de521467b1 : SUCCESS in 17m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9596d6e0b634b22831557303428b5ea : FAILURE in 6m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a73e5fcd593445d8a0dfa29829b74b46 : SUCCESS in 37m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54f197810ad648188ef297edfd3d9dad : SUCCESS in 48m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/719e4d3abbf54d299cfd1dca546cfe79 : FAILURE in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/438e936e111f4779b3857d4311b24094 : FAILURE in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a13fe7c9b32454fa8cbc8ac20744f9e : SUCCESS in 18m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c09d3d083995dcd32e997c4a1187a3091895618b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 16:57:19.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6204065d379c477f85259b7209c0d913\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/025ac0744e014eb78402b05ac05e39a0 : TIMED_OUT in 1h 30m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52263423d0b947048c202bd89c2038df : TIMED_OUT in 1h 32m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fb5044a8fee41669538d7518d846730 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8c112e66e904942af0e0c41ac1f5fb8 : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":3},{"id":"7dbe05fa96789ab8ea11f5566ba6e724247e8895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-15 22:41:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0b44d2347a8a86d1cee9f93e98778a53ad4fc0a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 23:27:22.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e706913ef9b74aad8867d353c7afaa8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/895dc0e14a014c26b1fb3a02341fa2e4 : SUCCESS in 9m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b09427365e4d4a82a2faba79a17c0a33 : SUCCESS in 16m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be8065cf4da74084bc7d9654bd4b3bb8 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53685b20af57434ea80b71b04f5bc2e6 : SUCCESS in 17m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03cf3e58894743c9b756c1e657fc2214 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/656d89f0c63043198487ec58ba030515 : SUCCESS in 17m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f283ce1243d54ae1bbf13c4153d37938 : SUCCESS in 18m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/75fc51eaebc147ddaef7453ade03c5bf : SUCCESS in 15m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e04e62250fe64f6c95505ce7db93ef85 : SUCCESS in 19m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c1f4829ad09f426e866a122b9e35fea0 : SUCCESS in 32m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e9037a9cb9a468e80c2a54cfe3808f9 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a327395ffa764a18af032ba0f8eeef00 : SUCCESS in 19m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89c1d9d6a06546fc9ffc386304e18b6b : SUCCESS in 18m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/65bdccd7a0d64ad3826846f83a92ff0b : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/87bd4f073a1c44869044d1c3b1d19430 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cb191026588b459297e640e4a6d4fcc9 : SUCCESS in 16m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d6ab09bd5c748d09e74c2108f04b5aa : FAILURE in 6m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d401c5b77c640848d42ff03202ccd64 : SUCCESS in 35m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c72936a31817444aa4d722de294a77a0 : SUCCESS in 45m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b04c1c89dbe489fb079420d705f8c46 : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51f8789fedb429f8c62ad34466da119 : FAILURE in 3m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5e068a3757b4e369e72c79a5b457d71 : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":4},{"id":"02306939e931012ecd5ae17073375cbe0beeac25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 23:36:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4006c4683833420a8198e06800fe226e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f2a0d31065f413e9e5afc6bc7c9afd3 : SUCCESS in 53m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/521241241ed64683a197720150dd78c2 : SUCCESS in 53m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a0574e2518c4fcca03e80877e4adc6c : SUCCESS in 39m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1731a9417204923beb6269ad435ccfd : SUCCESS in 48m 13s","accounts_in_message":[],"_revision_number":4},{"id":"6ecbd8d52aef96a113c8139e08ca29f74f6c70ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-19 14:19:05.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"04c6254211c512954b4f38196b732ae3e1270c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 15:07:43.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e18fac989c74dbfbab116ee4cf6926b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7db812a85d454cd7b1eba49d696495c3 : SUCCESS in 7m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75cfc7d35149479d895169fb64fa2435 : SUCCESS in 22m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e689303db9e4a86bfc5192b1e8b8de5 : SUCCESS in 23m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d23c916735d14be39d7207f803fc21d9 : SUCCESS in 19m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f4cec38a2754c8886b0bb5fa439c8fb : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33393b4aff284d239de9274d9e3d9ce1 : SUCCESS in 26m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/190bc4ea90624177bba2bb1f103bae66 : FAILURE in 21m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a20c8f5f7c644302bdb9bf2e19a4d7f7 : FAILURE in 15m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/58374e0406b44baa907440a358b3c065 : FAILURE in 20m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/99cd851724074eb99cb37ea3e790e7c4 : FAILURE in 18m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad32753510a44535b5b98b87f99be18b : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f4463359a7d4eb49587f0bf7fb84b2b : SUCCESS in 18m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0a87f80b265b4ee79e93f24d9588fcb7 : SUCCESS in 15m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f0ab0c28c7a34ee080187ea8f7fc7700 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/959ec804a4904ed48b54fbd7b314f14c : SUCCESS in 18m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a8e8147999584ddf92b64c81daad17f6 : SUCCESS in 17m 08s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d87cab852a934af19bb9e8aa716d595a : FAILURE in 7m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ed0ab5d019c4e599cd73341bfb834e6 : SUCCESS in 37m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a661e3780ec5439ca68f52e322a1a5e2 : SUCCESS in 46m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b10decfcc29c4e2eaf10cfe12f940e84 : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cda00e0564b445c48584cf3b03fad59d : FAILURE in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a7aac82bc204b259f802f6b51728c7b : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":5},{"id":"05c2a48cd179af365a40ea3de7db8b167a8a4e69","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-19 15:37:05.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f24b95ed599234a25fb88c95c53b6d9832db5410","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 16:19:56.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbbfb3cf93fe493e835ca39d3140f546\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/347ae1a2cac24078acfad9fc1f50d7b2 : TIMED_OUT in 1h 31m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68ec16d13e9b4ad9b0c187d444402134 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/626e67ea89f44458b12a5d7591777196 : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/63d7935a8c954a8db800c2960bcc7f05 : TIMED_OUT in 1h 31m 03s","accounts_in_message":[],"_revision_number":5},{"id":"55dd57d8a7aa1fbef1feb81a639f135a6e1a7075","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-19 20:38:51.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5a2a22ddb708c8d8c54247d4650408af27a81d26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-20 16:38:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a1ce80e2a33e836feca624fa6dc46af25942c7a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 17:40:06.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a2323cfbb7b4671be0ad34847edbd9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb1725ac3c64a10be1740536460060b : SUCCESS in 9m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e2403f49b6649a0bdd9a756f0c3a0d0 : SUCCESS in 17m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3162964b16a146d4a4bc00cf6b1ddc07 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d22bbb46e664b9c8e6fe26d27cecea0 : SUCCESS in 23m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb03b2125782497fa5e9e417f8131556 : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84b209a2fd5a498faa8582a020f7cc37 : SUCCESS in 21m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7369aaa0c6814a0281032ecde7d2e9a4 : FAILURE in 19m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/164511bd5f1544dab687bcab4a695c47 : FAILURE in 16m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f8f5c4ca1bb94a17a4d59a9600de0581 : FAILURE in 16m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cc83fd6fc4574cdb8acf16a4c9b376ba : FAILURE in 21m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/593769a3b8394eda8faa36fcd76800cb : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/58bf7f720b304118aa1c259f19934156 : SUCCESS in 17m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7c6f16c3281c4e888e2da02bb8bb7b66 : SUCCESS in 16m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/847c9d993a3e4b939d732763119916b0 : SUCCESS in 24m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9e28ed810555472bb7f9ea1fca021f42 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0f29ca7e30e94138b43dfecc5ded25cf : SUCCESS in 18m 29s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad4c779b1ec94ee0bd209381034f7314 : FAILURE in 7m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1679e838847e418fb81de5101e24c7ec : SUCCESS in 32m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/390ee6fb48594c8aa6350300547ffbf1 : SUCCESS in 54m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7416d659f02c40f3bebe8f6fac525e2b : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f4365ef296444669432e7251a036915 : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0678912257f94c35aec1e1576b351aa5 : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":6},{"id":"a12d3b3e13c80029d5bb1827f6f9dfa645f50c7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 18:21:14.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48f4ef8b0f7f40ea9b93aae24036f52f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85c897a3c42d4fe7b3fccf7c2dbf357d : FAILURE in 1h 21m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b5430fcca7447719fce0f2096f2d6c6 : FAILURE in 53m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/17f247bdda7640a896cdafc3731d4a5e : SUCCESS in 52m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4a3376678eb40f5a8fe8eb8dc67e2a2 : TIMED_OUT in 1h 31m 08s","accounts_in_message":[],"_revision_number":6},{"id":"a40024d17a97a7b08c75f4bae756428bd8048a5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-21 20:15:56.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"4642ca4de8ba78c32dc0b85aeb582fce4540e7f8","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-21 20:20:57.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5e108d3ede7ad80424c845748ae2bd69e4a07582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 21:30:02.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f35aa01b7b54ed78bb5b1053faabbb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c96015a02944d7f9ff35e5ed642f53a : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/9b5ce70a0dee408784338e14d9d71460 : SUCCESS in 1h 08m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f52780aa6872483d8f58b598347e6684 : SUCCESS in 46m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef701d3014a54cd2a752a09b3ea5f6d9 : SUCCESS in 1h 04m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77e63f0dac5941ffa221383f9804fada : FAILURE in 24m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/386a2e8ed4834e8aa5675a5c076589f7 : FAILURE in 23m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/557897235dd44c06938d82d338257b8a : FAILURE in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f3449a8b2c74735b4299fa2d66bdbc6 : RETRY_LIMIT in 2m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6feea8927b354d68a4c6919afdb4d36d : SUCCESS in 16m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c971e5b1cc6045f5bc3e82c09fe8ca68 : FAILURE in 15m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/920acdce565c4b02827c98c25304d082 : FAILURE in 19m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ce6bc43d16734cf2801e78f088ca6fb1 : FAILURE in 15m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/99a5dd3fe74a4de984a945c9ea2bce17 : FAILURE in 23m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4c578a3f902464bbe176f0cfafb262f : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0cb95cdd11d43909a1b5136564dcdd1 : SUCCESS in 19m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a77a18cdbf1c4bbb889a2f6d6fbb7934 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/79ba0474610d48688ad29d1daaf6b56d : FAILURE in 18m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/98df4495603249caa103a968ce14c8e4 : FAILURE in 16m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9b99b084b5a94e5dad42ee28f48aa7f7 : FAILURE in 17m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2cd641bdf9d409e82c6fd0ce40a29ab : SUCCESS in 7m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bdb405ad4c8e476484686a2050d555d5 : FAILURE in 13m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9429b43cfb5d4542878f7e89bcf6fa99 : FAILURE in 5m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee19a9b5062c494ca587621c24d589a6 : FAILURE in 55m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e895fb6bdb3a4f569719bcc8c1f9e79c : FAILURE in 34m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5beb8b7f14014845a38421588d75aeae : FAILURE in 37m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38226ba9085f420fad4fbfced4713070 : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0246f05263a641f1b9387d94c56abe42 : FAILURE in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/194341a741a94e538719877e115d6745 : SUCCESS in 18m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/719264646e964e53bd80e5cdff441d35 : SUCCESS in 51m 01s","accounts_in_message":[],"_revision_number":7},{"id":"54b5c31a4ee256af4bf7be9f9113f82afef519e7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-21 23:44:13.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"218e98264c243bbcd3d236c9e273451c9cf193bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 10:57:27.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a15c02c48534f94beeec772dc16d6c6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6890cf5a05c74ddfa2a095d1def4bee3 : TIMED_OUT in 1h 30m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6f9b903c169455981899b06e091d6f4 : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/075eb9d8f398471ebbe03db37aadfc79 : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e07283a58a444c748e489ac104bf34a7 : TIMED_OUT in 1h 31m 06s","accounts_in_message":[],"_revision_number":7},{"id":"3bf0025b94f0d007e1a0f3967a2fcc53698ec058","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-22 20:04:51.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"b7b6e63ce0acabcbea12c7ccb97d94e9be7636f2","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-22 20:23:55.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d1fa1beee7777d18d295fc15919cc808beb86f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 21:31:56.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d97a1d23f814c1a8683fb73e495109b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd22957092140e2aae0968cc1478bdc : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8d7b1c78d09f469b8c81330472494a72 : SUCCESS in 51m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e2cd30a3a2d4459988b4a8556c9f183e : SUCCESS in 38m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6a4a849f17945688dd746ba0603674e : SUCCESS in 51m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57e6f117435b4261911fbe60a1b27c7b : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17c173989cd74cdb815a9d317619cf4c : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b22b4d114dc54914a43eddbc5d1b6e4c : SUCCESS in 21m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/57d8c73053454ba9948e40859dfac263 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a18080635d224b618e8b173e8e2e67ad : SUCCESS in 23m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f06314a734a4072b880d7e771b53b46 : SUCCESS in 19m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5b856951a0e142c09deb24c7386fb534 : SUCCESS in 20m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/03d66aca58024a8a97243ea02f328fc7 : SUCCESS in 21m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/74560426ac234c5f8303ecd3e3e8996f : SUCCESS in 38m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e7f41d2548c54ee1b95fc99f3ee79b78 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d41577e6684a4292913130eaaacc8991 : SUCCESS in 21m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cbe14a0cc12b4c578e46835848327d26 : SUCCESS in 18m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6f79479419d744429dc0ceac408f0637 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/becb491b1a2d4dc49790aefc45095a09 : SUCCESS in 18m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d98306b5b7d47cbb7e0d721ce36856f : SUCCESS in 17m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c7bec37d558e471e8222c710b00f6461 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/18cc475301384e3d95fe3c1ab40d4603 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/28d3e44ef2824b96a2c39764806d12ae : FAILURE in 9m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/99e4e9061f434c9c98fb84d8fcab8c45 : SUCCESS in 1h 08m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62487c74291a465a98e84b60b9b4a299 : SUCCESS in 51m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/424dcfa78c1e4668adfa1f17e8a6a42d : SUCCESS in 50m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df38ccab70eb48d5a602b4c82e81e16e : SUCCESS in 23m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b48a8854d9074eaa8573b962fd1af02a : FAILURE in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a80d38cd01044bba87c9680ea34ce43 : SUCCESS in 23m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7282fe98fc5d4a78a7b7e6cb295ff3fa : SUCCESS in 51m 43s","accounts_in_message":[],"_revision_number":8},{"id":"f1f66aed4a90100fcece414d2c6a2e3a0d5133c5","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-22 21:58:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4713bb33347c1023d5cdbf15cc9256e8c496a0d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 06:37:55.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f822b46bb10a416a880c208afad47ec1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08498d75ea5246789abb384318816d06 : SUCCESS in 48m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e96104a56504bc2a8c8e31289c7f571 : TIMED_OUT in 1h 31m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ab325b045b94410bad9e9cd7410941e : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/065f0da499b24b6b99bc646df0f7101a : TIMED_OUT in 1h 30m 48s","accounts_in_message":[],"_revision_number":8},{"id":"981157d8141e6438a9d716b83aa8fd86d1e788b4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-23 16:19:00.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3c163588b90910b7b117b3e51eded6c3d0ee3513","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-23 19:24:14.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5c0b453a4dee7261fdeb0042b1c691902f8dcb65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-27 15:30:02.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d80b67d628d46fa7f9f3373bb3cf6092dbfb6308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 17:51:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0917c6251d0244ce84caf3843d59cc61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2af5599663b44bb86f934cc0d133c1b : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/1485049cb862445384a2369eae2047ca : SUCCESS in 1h 02m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c451a090b0194f53a94478fd433daf8d : SUCCESS in 52m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f05fefb1c3b542d8a8591b3abe51ba39 : SUCCESS in 58m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9466bd1bc2043b89efd6411d046ed17 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/722fbadc4bb745d5bed03b36cdad8ebc : SUCCESS in 21m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba19cdff31e14797904f68fa98730aa1 : SUCCESS in 28m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1af409d45daf469ab255fa242c8a95f5 : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/367a3f477ec240dd81e13785408685c0 : SUCCESS in 20m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a712519e24ad499eb9edcb64dac30365 : POST_FAILURE in 20m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/19837a3a020f40cc994ff1a2d2501bfb : FAILURE in 19m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/64953c7947b642debdb94a4ce7fa466a : FAILURE in 20m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9cd88516d15140559a7059c285c4d90a : FAILURE in 26m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62c92c67b5b940109e477ad734199b9b : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4c8d1b678ac4c678827c37c92d7a35d : SUCCESS in 21m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/54543c2bd4ed4bdcb4887d2fba440a33 : SUCCESS in 20m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/590fbdb970e1474ca2d86702c78510ee : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/34563f51355e4c03b06db2aac45a431e : SUCCESS in 17m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9b9daacf95ca420e96c8f749179d6c0b : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/136a25df6abb4d82bfc711078e40186d : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a963d20467434f2882b315a73020f184 : SUCCESS in 22m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4efe53432c6d47f5b58a135a79fef619 : FAILURE in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/52259b68412249d882ce560426d903a0 : SUCCESS in 1h 05m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d5c3affb4744a808e3932415d8ef440 : SUCCESS in 48m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/306cfde1758a4188abbd5c1571bb7860 : SUCCESS in 53m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7defcdb1e2aa4cb8aa46f46151ea4615 : SUCCESS in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5becd0cc1cba48bda3796835d40dd85f : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/855378413b924c278228030fef107642 : SUCCESS in 17m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cc4cedf0c8c49cd83c9679fe246749e : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":9},{"id":"c27c79f55d525da27f380b60c027bdb06ac48c2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 20:58:32.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d78543ad916492aad72af2abf075c4f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29de3e8141c3470798e337c16f0e1df9 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7c82c1ff6994081a62de29f779034af : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbaaefed8fb740b1b8094914a61c7541 : FAILURE in 1h 29m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/915cf996f6a943dd9d8a417c17a3bc0d : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":9},{"id":"8d249fc79fc09398577a3033180e7bd77079f97c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-09-10 17:50:47.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"75d0731314bd9eb619ddd918923f9688124107b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 19:09:07.000000000","message":"Patch Set 10: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/823e8de8a951422b8395fe0171433506\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb3308e508404c3fbcffdfd6461afa38 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/20fee5e046f94e73bbc3f2beb89be3b2 : SUCCESS in 1h 08m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6a5fe185fdac40d18d909f92f604cedc : SUCCESS in 52m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8cad5576f2447b894863bda59fe073c : SUCCESS in 51m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f03ab3fa93bf478e8a084b54004fb9dd : SUCCESS in 22m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c383b3c19cf43d58ccc9c9077414c96 : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a2e87bd5e7646e0a778181d35b53789 : SUCCESS in 21m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84839ce001bf46c1ba3b6f7be5e90994 : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7868ddcd7ea5458a891b59ce7ac7e554 : SUCCESS in 15m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/866024e0fd314d6aaa5255730da05844 : FAILURE in 19m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/65a9cdcce7e24a1ea6f80ec960933ea3 : FAILURE in 23m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/27963c997bbc481397dab49b27566b2b : FAILURE in 18m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ca254380d999412ca37766bf6d63d0b6 : FAILURE in 19m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cdb91bab1df04a8aa5d41eb0820cbf6e : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3ead5c3d16914dc89052de06359735f0 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c29fe9534aa2482eb039055b8f020c46 : SUCCESS in 19m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/71edfa6aad7843f6bc34cc7bbe5b138f : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e1efd047786f489aa1cdcd69c4ce25ad : SUCCESS in 11m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2ccb163a0c274cab8b6680e0a1e9016e : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0751fc8252bc4fa180b44bb87291d7c0 : SUCCESS in 4m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a65a5d52990247e5ab108558175384ff : SUCCESS in 12m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c9e1e6f674c84bd29b7757b4d1725146 : FAILURE in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b1e0eb4b5334f5db7e82d016266630a : SUCCESS in 55m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a503de5af84a4365a34ccd60c5522ec6 : SUCCESS in 49m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0789d15c0f874b43ad5ce3aea19a3cb2 : SUCCESS in 42m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59a264b85b3a4668abd41c278132e066 : SUCCESS in 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/149589d6db764170876fc0326e5ea5e4 : FAILURE in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/72b355a95c13499d9573c8d3f36173be : SUCCESS in 20m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60a7c3f418964a9c92a000f464c2cd85 : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":10},{"id":"074c581a66df1aad963f613b40122876854b717a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 21:57:59.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2cade6b4e52488e8ea0cb983373ad8a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e38955c493b48b19e9e0183965e56ac : TIMED_OUT in 1h 31m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/644abfeba87548d88deb08dfd4b83a6d : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f5539219b1048d9b9da1eed9948b756 : TIMED_OUT in 1h 31m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d452a8685d244becb0b993be1e990d9c : TIMED_OUT in 1h 31m 02s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"8acff48b9a09f56831b186ad4239a99409e168aa","revisions":{"7a5448ab85f6f508547adbadd37fde15b55f07ec":{"kind":"REWORK","_number":1,"created":"2024-08-01 23:55:00.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/50/925550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/1"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"subject":"s3api: test xml escape chars","message":"s3api: test xml escape chars\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a5448ab85f6f508547adbadd37fde15b55f07ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a5448ab85f6f508547adbadd37fde15b55f07ec"}]},"branch":"refs/heads/master"},"f75f7604d759c507c1fa6b896041dfb98b7763e5":{"kind":"REWORK","_number":2,"created":"2024-08-13 14:43:23.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/50/925550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/2"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-08-13 14:43:01.000000000","tz":-240},"subject":"s3api: test xml escape chars","message":"s3api: test xml escape chars\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f75f7604d759c507c1fa6b896041dfb98b7763e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f75f7604d759c507c1fa6b896041dfb98b7763e5"}]},"branch":"refs/heads/master"},"cc2ee309f12da412746adbdf5faca8a8beca8ad4":{"kind":"REWORK","_number":3,"created":"2024-08-13 15:22:58.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/50/925550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/3"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-08-13 15:22:34.000000000","tz":-240},"subject":"s3api: test xml escape chars","message":"s3api: test xml escape chars\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc2ee309f12da412746adbdf5faca8a8beca8ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc2ee309f12da412746adbdf5faca8a8beca8ad4"}]},"branch":"refs/heads/master"},"2b9b7ff05e375d5e11ce91be8d91b1fba22d989e":{"kind":"REWORK","_number":4,"created":"2024-08-15 22:41:35.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/50/925550/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/4"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-08-15 22:36:48.000000000","tz":-240},"subject":"s3api: test xml escape chars","message":"s3api: test xml escape chars\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b9b7ff05e375d5e11ce91be8d91b1fba22d989e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b9b7ff05e375d5e11ce91be8d91b1fba22d989e"}]},"branch":"refs/heads/master"},"e0e3b3d6738e377f39deed87f3b43569436414b8":{"kind":"REWORK","_number":5,"created":"2024-08-19 14:19:05.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/50/925550/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/5"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-08-19 14:18:22.000000000","tz":-240},"subject":"s3api: test xml escape chars","message":"s3api: test xml escape chars\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0e3b3d6738e377f39deed87f3b43569436414b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0e3b3d6738e377f39deed87f3b43569436414b8"}]},"branch":"refs/heads/master"},"412d6704583ba8cae4dd89614ffbf4a58879cc54":{"kind":"REWORK","_number":6,"created":"2024-08-20 16:38:44.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/50/925550/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/6"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-08-20 16:37:12.000000000","tz":-240},"subject":"s3api: test xml escape chars","message":"s3api: test xml escape chars\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/412d6704583ba8cae4dd89614ffbf4a58879cc54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/412d6704583ba8cae4dd89614ffbf4a58879cc54"}]},"branch":"refs/heads/master"},"d25062d3beb84477f4b224d6eab078bd646d18d0":{"kind":"REWORK","_number":7,"created":"2024-08-21 20:15:56.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/50/925550/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/7"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-08-21 20:15:40.000000000","tz":-240},"subject":"s3api: test xml escape chars","message":"s3api: test xml escape chars\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d25062d3beb84477f4b224d6eab078bd646d18d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d25062d3beb84477f4b224d6eab078bd646d18d0"}]},"branch":"refs/heads/master"},"44f7b0770d999655f001a66bbc6551f208171201":{"kind":"REWORK","_number":8,"created":"2024-08-22 20:04:51.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/50/925550/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/8"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-08-22 20:04:29.000000000","tz":-240},"subject":"s3api: test xml escape chars","message":"s3api: test xml escape chars\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44f7b0770d999655f001a66bbc6551f208171201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44f7b0770d999655f001a66bbc6551f208171201"}]},"branch":"refs/heads/master"},"490ca88f8f42aa02916c068177f88836df2dab7b":{"kind":"REWORK","_number":9,"created":"2024-08-27 15:30:02.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/50/925550/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/9"}}},"commit":{"parents":[{"commit":"682c71afe4b1c253651856593441df8fbccbdcbb","subject":"Merge \"Implement context manager protocol for logging mutexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/682c71afe4b1c253651856593441df8fbccbdcbb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-08-27 15:24:45.000000000","tz":-240},"subject":"Server Error listing s3api bucket - Write test with xml escape character - Resolve 500 error with encoding_type\u003durl","message":"Server Error listing s3api bucket\n- Write test with xml escape character\n- Resolve 500 error with encoding_type\u003durl\n\nCo-Author: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/490ca88f8f42aa02916c068177f88836df2dab7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/490ca88f8f42aa02916c068177f88836df2dab7b"}]},"branch":"refs/heads/master"},"8acff48b9a09f56831b186ad4239a99409e168aa":{"kind":"REWORK","_number":10,"created":"2024-09-10 17:50:47.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/50/925550/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/925550/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/925550/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/925550/10"}}},"commit":{"parents":[{"commit":"8efb3338726d6f167722bc5c4b0b3690f9f6a4ae","subject":"Merge \"diskfile: Treat EUCLEAN like ENODATA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8efb3338726d6f167722bc5c4b0b3690f9f6a4ae"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-01 23:54:31.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-09-10 17:45:54.000000000","tz":-240},"subject":"Server Error listing s3api bucket - Write test with xml escape character - Resolve 500 error with encoding_type\u003durl","message":"Server Error listing s3api bucket\n- Write test with xml escape character\n- Resolve 500 error with encoding_type\u003durl\n\nCo-Author: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n\nChange-Id: I4ff77ca6a94a6b9613d93082cd36a23929c94236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8acff48b9a09f56831b186ad4239a99409e168aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8acff48b9a09f56831b186ad4239a99409e168aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
