)]}'
{"id":"openstack%2Fswift~925575","triplet_id":"openstack%2Fswift~master~Ia0041c4eedc33a2675890d52cbbc33b9190f1665","project":"openstack/swift","branch":"master","topic":"expirer_task_metrics","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-08-06 18:25:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0041c4eedc33a2675890d52cbbc33b9190f1665","subject":"expirer stats: add test coverage for \u0027errors\u0027 stat","status":"MERGED","created":"2024-08-02 10:50:47.000000000","updated":"2024-08-06 18:26:16.000000000","submitted":"2024-08-06 18:25:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925575-expirer_task_metrics","meta_rev_id":"d9010bad7873794a590632e927641dfb738f4f61","_number":925575,"virtual_id_number":925575,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-06 18:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-08-02 13:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-08-02 13:23:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-02 11:49:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-02 13:23:42.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-08-06 16:27:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"21639e73ed8ba6254184693c77793db2938d60d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-02 10:50:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfcacce84e536fea264bf5c990d7621a28836d60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 11:49:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/460e8df3c8ab4e39abfcd1411cd2fd04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f780e231b543788df191c85ccd762e : SUCCESS in 9m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb9d1aa541d94d56939f83dbe0f6d143 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e316623d0f8940988a84395a3512f695 : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/691db223138a40e090044562cac98906 : SUCCESS in 23m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f838603f8cb4fb3a4a7269bf89b4502 : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/90d9028bf2b64779b251b18b73cd56b1 : SUCCESS in 21m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9047c6e32f6d4bb88e0038b38ebf06cf : SUCCESS in 16m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/18d3f4dec3544f97be6d60488c9f5862 : SUCCESS in 18m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/439b437dbef1425f86cea15817d268bb : SUCCESS in 19m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a4eea335d2c8453988c357e0f61e1e9e : SUCCESS in 24m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3b122714557b490a8472e545ed734f10 : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b78d1065e3f748529d55053e9e0b1bc5 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/459a61a9ba474c7fb43ccad40ea8d654 : SUCCESS in 20m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fd73a176645b4fae8207926352cb8705 : POST_FAILURE in 17m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/513d7fc173384243914cb961af09a3e3 : SUCCESS in 19m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0eeefd8bda0b4231b0b4a7d0c5ba2b68 : SUCCESS in 15m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7dcac8ff75174736814b8e26233a86cf : SUCCESS in 35m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d116e5f93534afb97bd07e6b45cef13 : SUCCESS in 55m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b09a8f8cbf54af787bd6132e93c4ffa : SUCCESS in 21m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d88a235f9bc346d6b307b14df728683b : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/70da4211e46b42bfa8477847df250617 : SUCCESS in 18m 42s","accounts_in_message":[],"_revision_number":1},{"id":"221d75c0031127ed58c454cf1252affd9eeab7df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 12:01:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bfc899f791a4cc5aac04f87cd311298\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d368a1b3da14486c972a30d565bbe435 : SUCCESS in 1h 08m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68952faaee40487aae3e0a3c5fbb19f4 : SUCCESS in 48m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/79df0d45fdc44029b101f5bc29ffd960 : FAILURE in 54m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/67931166d93549ed88400364d3abf935 : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ac6658920dc8280d2299501f2bede4309b58c954","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-02 13:23:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e5ddc2a7ffcfdf7724cd7bfe2e25eac14e7b2a05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 14:22:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/822dc0703509452ab9b36a13be3bbf82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edaebea75abb41809da572f706d59544 : SUCCESS in 10m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2ed9a190f1047b4ad2ebcb7982d47aa : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cd2048c24144cc9aa4791604b11db08 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7638f4dbb3c94ea0a630fea114cfe4d1 : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/50f1f8f68642412688be38b9472b8b6f : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/95fca42ecfa64c87979e120d9bda2842 : SUCCESS in 22m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/daae8801d0b942809887e4be5d37e232 : SUCCESS in 18m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d89b4774deb745469ad22c98eb105ada : SUCCESS in 18m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e95612ba131d4405ab433e0dbfe6279d : SUCCESS in 20m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/45a2d68bf60444d8a7eaddf274c4d369 : SUCCESS in 25m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/652ec18197f245c6acbd1d878e8ca915 : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/36fe7d389f9440a890c8944167aee1a6 : SUCCESS in 20m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6cc31a7a624d486583bb2cf79a1b25a7 : SUCCESS in 28m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aee080c7edf44e6b85203d485b9440ff : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ced25732244848fb9bcc645960a9d0fb : SUCCESS in 19m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3f5894064f1741bca6852099a1659f43 : SUCCESS in 18m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a042470ca6f54fc4bda6a9f99e723fee : SUCCESS in 47m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb22be1d155548f4b59eb8e44d7c41f1 : SUCCESS in 46m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78874d890ae44f8ea8a043947f8248bc : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8693a837f6142298a8668f6835e6b3e : SUCCESS in 6m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b65945eeaa1f4b0dbfde7f938c58d958 : SUCCESS in 17m 24s","accounts_in_message":[],"_revision_number":1},{"id":"cef8bb7993ba1560bad3ebf21ec88ee07db3b20e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 14:51:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f23de8efe9b3ed3e0d7c8b7d642161f66dd9ccf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 17:18:04.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/925439 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"52502db97bda77a23c597d63a42ca0068276557c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-06 16:27:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"828db1cc53de486e8199fca49a5300553b2f1ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 17:25:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c806008b7924dbeab93cfa6f2ec3936\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/079cad44a1974ec9b9ef6a360ed79572 : SUCCESS in 9m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7eb8ce5b121547e0bf7d387bd5181d03 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd2577105a064201a09475ffda42abc6 : SUCCESS in 21m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9196dad4e43d47e582acb7c9a62ac12d : SUCCESS in 23m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/483d4230d57e4fd590d9a8d70e87ba4c : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4360a637d8944c508421abc8d5676fb1 : SUCCESS in 21m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a04c6b1478c47689c11463b4ed927ec : SUCCESS in 17m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03980a092d9745eebdc22fb4d855a80c : SUCCESS in 19m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/740d629be3ac43cc8d8f7dcc1e63603a : SUCCESS in 20m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5568b874e2df42aca80ddb423b2523c9 : SUCCESS in 28m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e203c0aa60b4dca8488a24822644b1d : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/be68113496c14da0ac887919324d890d : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bdcd9465c4384118be25e0c3e8d6b3c9 : SUCCESS in 21m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3c0c9c7fecab49a5be1daa753e4884e7 : SUCCESS in 20m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/14457df4f70441f7a2baa0a09b55fa30 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/89a0d30391ae40ba866f00b9e7c3b3b5 : SUCCESS in 18m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4dea7b70ea15465094942cedce270e6d : SUCCESS in 52m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f8cceeff5c19492ba0fc9784c2cd057e : SUCCESS in 39m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d82ef5d5cc04f31ac79ab1dd9c7444f : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0586ba8df4864f44812aacb93c94629b : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9539e32f111b479a96a287a7691257d3 : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ce318480beb5fef98003fd0bb7f2167671e5c0fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 17:26:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d7b951da041b06c36ce6b90058bbd5aab2352b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 17:51:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d219ec89753548f7a3054ddf687c860a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d3847d18fd74c048279ea0ecdda3b58 : SUCCESS in 1h 19m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47e9fc7c1b8d4148b33b46b1e659b759 : SUCCESS in 1h 13m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0f0dae814d34a45873c64b33317d87f : SUCCESS in 1h 08m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a58a5332f63f4191b0ce442875cde2f1 : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e0a28c852946037969a327921491e8087275f5fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 18:25:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92186482b7bc4832864b282720e09ecc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c64289566b7d455dbbd14c35ad41ec5b : SUCCESS in 8m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/43c48442a1874a66ab9b9d70f4d806db : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d7d116806851412ebc023b85a52641e0 : SUCCESS in 20m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52389d6f95a347cdb81dbd0e4b1cd047 : SUCCESS in 23m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/827f3caa402d4c83a89f01b1d5f4a205 : SUCCESS in 20m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0640149883c24c25923d55a9cdb1a824 : SUCCESS in 17m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/faad96dcf4474c88a35dfa5256400737 : SUCCESS in 24m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/198aafe844a0496d9a8e78754651bc82 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59ba99beaae64fc992963dbc2f3279aa : SUCCESS in 20m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e5bcbfb45684f19ac9c68d08b011696 : SUCCESS in 22m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8563536495cf485790e28be5c6e8d55c : SUCCESS in 19m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2acec7ca2768409f8713d5581192b467 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8c7da9f885d34aafbecb2b23540d14d4 : SUCCESS in 21m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e25b55769cdb4f34959e241f720535b0 : SUCCESS in 5m 21s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/98a9a576fa384503a3a9dcc9cb189ba9 : SUCCESS in 9m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6dafda137024d6a9336c06e1348c113 : SUCCESS in 40m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/132cb4469b9f45f5b1608afed1b9215a : SUCCESS in 51m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed375833e3d34ee8acedde8c4208236d : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711d126b530048968bf820ef338fe0a6 : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/efc0ec7a9b974deeb400de5857993946 : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5687878a456bce8d6bb1195d1e18d5150f730e82","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 18:25:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d9010bad7873794a590632e927641dfb738f4f61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 18:26:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c99834a9941341d9ab575608b74fcf85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd6c5b8c5a6645259fc7927eb31ce657 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f682a0604129497e8735cc044baea28c : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d14c7c830ba92dfa9bcb59efee367c66001479b","revisions":{"9d14c7c830ba92dfa9bcb59efee367c66001479b":{"kind":"REWORK","_number":1,"created":"2024-08-02 10:50:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/925575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/925575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/925575/1"}}},"commit":{"parents":[{"commit":"b400a1fdb3ae609d5f86d4d0f1b590640337b8f0","subject":"Expirer: add delete task iteration related metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b400a1fdb3ae609d5f86d4d0f1b590640337b8f0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-02 10:31:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-02 10:50:12.000000000","tz":60},"subject":"expirer stats: add test coverage for \u0027errors\u0027 stat","message":"expirer stats: add test coverage for \u0027errors\u0027 stat\n\nDrive-by: simplify patching of test expirer instance.\nDrive-by: fix comment spelling.\n\nRelated-Change: Ib151c498ba325f39570e963e5b7948080ffcd3d6\nChange-Id: Ia0041c4eedc33a2675890d52cbbc33b9190f1665\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d14c7c830ba92dfa9bcb59efee367c66001479b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d14c7c830ba92dfa9bcb59efee367c66001479b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
