)]}'
{"id":"openstack%2Fswift~925689","triplet_id":"openstack%2Fswift~master~I061576d02ed1efe42aeb70cb9c9cc1a4f0ec6f51","project":"openstack/swift","branch":"master","topic":"db_state_in_async","hashtags":[],"change_id":"I061576d02ed1efe42aeb70cb9c9cc1a4f0ec6f51","subject":"sq? more unit test coverage for obj updates","status":"NEW","created":"2024-08-05 12:00:24.000000000","updated":"2024-08-05 13:23:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff5552251bf60fb2e47debd75bc4bffec63b1d2f","_number":925689,"virtual_id_number":925689,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-05 12:44:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-05 12:44:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be70a66152068c0651aac7035f7d9a6bf672cffb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-05 12:00:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6dc755de71de985a6d48bc24f76ac2c15848143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 12:44:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3f88c8a20a546e092bce17ffa54d7a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dad7a50b7bd408f9e5488290b0ff4de : SUCCESS in 8m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5fe2283cd75b4425a56fb16bffe0346e : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a1898feeb244895b1490c79e5b8a864 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/849e8084155b4903acc9a32a9f610b5b : SUCCESS in 22m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8cfd25951744ca8bb4c64453af0c9a3 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b89cbfceb8a42eab33fafb6d535c035 : SUCCESS in 20m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1b0c884f2c7c4dde8a0c4272204fcfcb : SUCCESS in 15m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bbce8fcf73c948db8a55f24f4b97311b : SUCCESS in 20m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1f25d273443d478483b7c00ab8aca121 : SUCCESS in 15m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b7c6d8e7162247c7bef3c6ecb22329ee : SUCCESS in 22m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29e8871bcdf641a09e3ab9730d4f4099 : SUCCESS in 20m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e861d792da8c41c0ad31927a15598f9f : SUCCESS in 14m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f88947b589c84211bb47aa945285874a : SUCCESS in 21m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/462947da80334a9ab8d407d24bb13eac : SUCCESS in 19m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/21f1b84d62a649a584dd17b8ccd05bdc : SUCCESS in 19m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3175458ac46c47038e08c1764f895f3b : SUCCESS in 20m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6135444e5444c90bf3cdd14b5727f82 : SUCCESS in 40m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6926dbe75fe74a2a813120b636eac7c1 : SUCCESS in 41m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/274e1c5c60284af89ed2205164990d3b : SUCCESS in 21m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5232560a7a43b5bc3e095807d01adf : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/427de893e78e4b32befda47650175af8 : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ff5552251bf60fb2e47debd75bc4bffec63b1d2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 13:23:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2117ac79f9f4521802baf322ae75385\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2f8a1bd68ac482e8403689a971f9805 : SUCCESS in 42m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e307422a20f45978bcc41f61c3f4f37 : SUCCESS in 1h 13m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffd5f652f000433091f7cfa883799656 : FAILURE in 53m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a9ec9e9c49b404b96c9d534f3d16f90 : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5b8d47a262ccb24810bc00f0e5f412290f05dde","revisions":{"c5b8d47a262ccb24810bc00f0e5f412290f05dde":{"kind":"REWORK","_number":1,"created":"2024-08-05 12:00:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/89/925689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/925689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/925689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/925689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/925689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/925689/1"}}},"commit":{"parents":[{"commit":"14696fdb2954e2a37bda4773645655c62e75201d","subject":"obj: skip to async_pending when root is busy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14696fdb2954e2a37bda4773645655c62e75201d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-05 11:59:49.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-05 11:59:49.000000000","tz":60},"subject":"sq? more unit test coverage for obj updates","message":"sq? more unit test coverage for obj updates\n\nChange-Id: I061576d02ed1efe42aeb70cb9c9cc1a4f0ec6f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5b8d47a262ccb24810bc00f0e5f412290f05dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5b8d47a262ccb24810bc00f0e5f412290f05dde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
