)]}'
{"id":"openstack%2Fswift~925762","triplet_id":"openstack%2Fswift~master~Idccf107370f7a72b77818f4848f83c3f46f6e3cd","project":"openstack/swift","branch":"master","topic":"log_prefix","hashtags":[],"change_id":"Idccf107370f7a72b77818f4848f83c3f46f6e3cd","subject":"remove unused log_headers option from docs","status":"NEW","created":"2024-08-06 11:42:54.000000000","updated":"2024-08-06 13:16:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b01417ad82dd91cdaa4a242000758efe64f54df","_number":925762,"virtual_id_number":925762,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-06 12:56:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-06 12:56:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d6967a9b54de77397b4c9fb4b7a223f2db28422","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-06 11:42:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b583d928765cdee8e94cb3dbbd9779df4c851e8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 12:56:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aa5855627934948b917c922234070cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dff513a54384bfe9f43a18fccac3ff4 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/6379e5062abb4cfbbe1b78653b7dcfc2 : SUCCESS in 1h 00m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9963d950a4a4496699f754eb7081c104 : SUCCESS in 46m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23e74f7e81e140c4a1a7faba8b109650 : SUCCESS in 1h 00m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35b439e477e146ecb2b72aebadcafbaa : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c077f41ec52146788ea28d7ca9ac5ec4 : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/125f89c14b6b4c3f98c6e72465e2c3ec : SUCCESS in 19m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/81c8ee94fae84dcca1cb599c92db46b3 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e32fcd65886d4c48942b360404c4b199 : SUCCESS in 19m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eee5c94095e7415e8cf5ef6456f1a2cf : SUCCESS in 21m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/75488100865b4659a0b0b58eddd177d4 : SUCCESS in 17m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/96c77be187784ea0a969db940f5dabb0 : SUCCESS in 26m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8887266d8c4b40f7b5c03dcd457fe6fa : SUCCESS in 26m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ea33324a3af94102911b7f6c776fe7d2 : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b15639b170b49d4a9b284808e2bce47 : SUCCESS in 20m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a6d4162de9e4a3589fcf05e182ab9bf : SUCCESS in 19m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5c8d7e1d1c0e45488710e85309df5639 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/64d3429d9e704123aec3d0792c253983 : SUCCESS in 20m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c78f0d7b23074d95803eea438b70ee80 : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b536423129d4429b401f89c9c994fe6 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ff1787f686d4123acb4235cd1395c00 : SUCCESS in 13m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c18c8cf3e6d948ab8970975d04e3c6ec : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8517b81e07e4a3baedb91cd3e11243d : SUCCESS in 52m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f556bc216a2e4ce3b382d2d09ba06fa1 : SUCCESS in 47m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e1407cb33c14b2f9b9803e88a186470 : SUCCESS in 48m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a4f9c29eb8d4a24845072ea439322ef : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3342bd62b99b4e07a51c8d19f4d5ccfb : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f683853fa1044dcdacd155bbdbe872ec : SUCCESS in 16m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb461f394cbf4c46b1847108d16267b9 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8b01417ad82dd91cdaa4a242000758efe64f54df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 13:16:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34bae3e2818f4fecbb33155c47abcd70\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/154afee2580143b09db4284cf4103d82 : TIMED_OUT in 1h 31m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/847fba93d2144b65ba9bf07d1be97ff2 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d8e58c231b04eb4b95e176093fac70f : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b949a76b10540ea9df6c21c8816e165 : TIMED_OUT in 1h 31m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"834d8c218ed4f869058a6ee6f1efa3461e13adec","revisions":{"834d8c218ed4f869058a6ee6f1efa3461e13adec":{"kind":"REWORK","_number":1,"created":"2024-08-06 11:42:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/925762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/925762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/925762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/925762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/925762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/925762/1"}}},"commit":{"parents":[{"commit":"d6f6b5ddcc38493340a873c38da447b0c79b377e","subject":"statsd: deprecate log_ prefix for options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6f6b5ddcc38493340a873c38da447b0c79b377e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-06 11:16:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-06 11:42:22.000000000","tz":60},"subject":"remove unused log_headers option from docs","message":"remove unused log_headers option from docs\n\nSample confs and docs show a \u0027log_headers\u0027 option for the [DEFAULT]\nconf section and middlewares. However, the option is only supported\nfor proxy logging middleware.\n\nChange-Id: Idccf107370f7a72b77818f4848f83c3f46f6e3cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/834d8c218ed4f869058a6ee6f1efa3461e13adec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/834d8c218ed4f869058a6ee6f1efa3461e13adec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
