)]}'
{"id":"openstack%2Fswift~925825","triplet_id":"openstack%2Fswift~master~Ieb5207c550bff81b3c67d882153e2ec73d4d19b9","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-04-10 00:30:13.000000000","reason":"Change was abandoned"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-08-18 22:41:12.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-10 00:30:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ieb5207c550bff81b3c67d882153e2ec73d4d19b9","subject":"Add helpers for adding/popping parameters","status":"ABANDONED","created":"2024-08-06 22:55:39.000000000","updated":"2025-04-10 00:30:13.000000000","total_comment_count":19,"unresolved_comment_count":17,"has_review_started":true,"meta_rev_id":"da493a394a62ea08d44105a754d351ee66691037","_number":925825,"virtual_id_number":925825,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-09 18:15:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":-1,"date":"2024-08-18 22:41:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-07 00:00:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-07 06:47:20.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-08-07 21:23:08.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"28e08142b5e4e1771ae9a04d9eb2dfe755986e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-06 22:55:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9640629faaf041dad0a5bb5ba0bf8b3e441d8a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 00:00:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/214926dabf084914b832560f130a8acf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7739eb40bf2941498ea41ac84efba0a2 : SUCCESS in 7m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/882fb8b1f163464093f20e30c4b5a831 : SUCCESS in 1h 04m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de3fa86db83b468e8caa549482368d55 : SUCCESS in 47m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/150d1f27539148b8a8859f40650d7e3c : SUCCESS in 1h 01m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93c2b300e8b643108d17c38ae2702c60 : FAILURE in 26m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d8884cccf9bd446eab4f901f7a982092 : FAILURE in 24m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15c8c8cee9604af688811de7cb3d296f : FAILURE in 25m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4024dad48daa498cab8070b2d2220ed6 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55575e3abc874d39b15c478509385189 : FAILURE in 15m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6d32d297c84d48ab9965d0f95017a147 : SUCCESS in 16m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a65f0ed39a0d4a1896df108dff905c88 : SUCCESS in 16m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a4660f892642439c9455b210b3d69aa2 : SUCCESS in 14m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/95663128a409437a967c006621b1b9f4 : SUCCESS in 18m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/035bfc825dd64e6096f6b0a9c809192c : FAILURE in 21m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ab9146b93b94027be7f7934968b27d8 : FAILURE in 23m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/43fc08ff428a4136ad4be05216a6b4ab : FAILURE in 24m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0cf07d4c88164b269abdaf6cda681adf : FAILURE in 20m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b17cf949d7234954b21261c25bff9873 : FAILURE in 21m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/32feec37a32847a39bf24cf7967db6bf : FAILURE in 23m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2995a27e40f422ea5c1c0b21da4ceae : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eadf90c01ebb494eb7d0fa2f4b2e734b : SUCCESS in 13m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6dd4dc3c89194349b3936ce843acfafb : SUCCESS in 5m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/70b649bdcca04a0b8ad83726645bf8b4 : SUCCESS in 49m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64fa4fe83c034800a7cbbc677756faef : FAILURE in 44m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2dc27f2e3a7048f389943c968c4012f0 : FAILURE in 46m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/033cb494dbdb4ee59d45fae39764b475 : FAILURE in 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9cc8d2fffe544e589c6bb208e03416b : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cfa15253fd1a4bb5ad1bba38d53ed0f0 : SUCCESS in 18m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccd300848e5048b6822ad0f167cfcd37 : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":1},{"id":"fb0b817470f2a25cfa6d522157c28cca1bef81d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 00:28:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f813b2a959d64cc38c1769c05c2c649f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27a5b28ab3da415a99058f17de895375 : SUCCESS in 1h 09m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a82da5ffa68b42748d9b292fba9e60ae : SUCCESS in 1h 05m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/41415e84f98c437a9d1185ebdc287298 : FAILURE in 1h 17m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/01f6994df33e4885baf8c64ac768e854 : TIMED_OUT in 1h 30m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6b1d4a311e0dba6f1eca19903d18210bb5e497a9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-07 06:47:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"41331e9d1c000ee91d2c57a73301b91ddf8049e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-08 23:20:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"948e5e10922dcd725227da1b761025a716f82d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 00:16:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2afeb74c39394a5182daeac5820d2448\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d8e8bff7ae4258ada36f913162d524 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/63e612a669224430bd1e036e2b8a0471 : SUCCESS in 48m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd12ba19caf84196a82795c22ee77e03 : SUCCESS in 38m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28d5ef6848f44809b8b7846cd90caec3 : SUCCESS in 50m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/981fc5e3074b48b09f74592103b0d754 : FAILURE in 15m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26f0efb509244c91973f18d2214110d5 : FAILURE in 24m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36b985d6fcf546e99fe9c542077fd113 : FAILURE in 25m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/45ea3688b8314828bd875ee892d96c26 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24e0e76ac3aa427191c38baeb82b51d8 : FAILURE in 24m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52f0d51e9c6d40a8ac9bb03c90fe1bd2 : SUCCESS in 19m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2b38e10dfda24b0ebb8d652004603c83 : SUCCESS in 16m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/65133ea9d4ca4bc0833607de6ad9fe70 : SUCCESS in 18m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f63ff878a85c418a8a28c7ae78f9b818 : SUCCESS in 30m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba65db6b997344e1a134bfd273be9570 : FAILURE in 17m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/17700931d0be4429b22a44fbcef1b688 : FAILURE in 19m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/54b260a9a93f4543a8f307d340113aa8 : FAILURE in 19m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/898ebb50d6f74beb9e967697ec722daa : FAILURE in 17m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ef5357e6619436aa6bae55002116018 : FAILURE in 17m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/070b8e7ca69945cb9ceabc259ae79a45 : FAILURE in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d3ceee4a0c86434f8f1f24111eb52e25 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78546ab7804e41b7a1cd30081c5b317c : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad89845f0e5f4dd7ad52d36b6be2a8d5 : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aadc3b27bf64f3090701ed32f4c8739 : SUCCESS in 51m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a65f18771fd42b6a022d3a52279e331 : FAILURE in 31m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b650dd875eaf439ea9eb8b5f82b45169 : FAILURE in 31m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e38bf0b5e87045bcbe11aca6a1e6e2f2 : FAILURE in 22m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f20750d211f4dd6b26be0d9ba0d4bfb : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c65a9c7119194b88b5141fe864ffce9f : SUCCESS in 19m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa6006179e984dbbb817c79a93836e79 : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":2},{"id":"daf513dab06c3798a064f9e8dc9959d97c9ba23c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 01:08:17.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ae6b5cff28e448a83855f7863ee6625\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d2454e1d6ad4b2ca23661ac6e62b848 : TIMED_OUT in 1h 30m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00c8439f8df44cf4a31f0e4ac7f48c24 : TIMED_OUT in 1h 30m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a02148930d2648d9a02a07d4f2e5f394 : FAILURE in 1h 20m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac0095ca7269410ab11def233f5009a5 : TIMED_OUT in 1h 31m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f6e6d272e961bfecf5d9b51763dc5049dc44c988","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-09 17:05:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"59a2f45cb78c5b12bab0ef185de6367a8fd9056a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-09 17:11:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b8ee7e9f3037fb4c1bc38d0f84aced9e7f0c1bcd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-09 17:35:29.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6fbe34e7058e98f60ad220d2f69e3a91c13d4c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 18:15:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97d6a1ddfb1042b9abd6173c75b8d077\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b07797e8a0440f799e1146b2ef76143 : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/c920752d5ec14fa890ead2bac2df931e : SUCCESS in 1h 03m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d38bc1661d4472da22e0c636d7d6b4a : SUCCESS in 41m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97d31155c63f4525acce00152b66746d : SUCCESS in 57m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d19684b10da14efaa9bd6260e835d9b0 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed7687ea85e748dda11c38b6069034cd : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46cc00435bbb450e947684e1ef31d99b : SUCCESS in 22m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/896305531e724136bc62095dcc84ca55 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a83760a74fd1435db58142c71b595d38 : SUCCESS in 15m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/040ee8816b8542dfb9044632216f0f8e : SUCCESS in 17m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/45395b22210044679411351c6458851a : SUCCESS in 18m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a4d31cb12bfa49348b8c0680ee07cc30 : SUCCESS in 15m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d6296f166fff467cbebc44001e440e0d : SUCCESS in 19m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/79b328f66c25432c8aa3f73cfcc2c122 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ece476106292464fb8e19476733d6c34 : SUCCESS in 18m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/23839424e1c5409db2ffed03fe889f7f : SUCCESS in 20m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eb265efd89eb40af913ec54076b46990 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d8e16baa31be4f4ab95d9c34c7888d0b : SUCCESS in 19m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3ea9f50068cd490db33c1f9a8444e56a : SUCCESS in 19m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/693487aea1904a9c926ac89b4034323f : SUCCESS in 5m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/810cd6b4cf5a470ab0811df65410e849 : SUCCESS in 14m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7172cd2cf3d84937b54b752878db84c8 : SUCCESS in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/45a5c92d9852489c9ba850ebe02a7621 : SUCCESS in 55m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bba86fa7b4a94c4f8861f68b75231287 : SUCCESS in 53m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92ed4a240f8145608c3a486202713a17 : SUCCESS in 50m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7671f2a661ff4ceca1e2c5e4adcf89c1 : SUCCESS in 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae0568e749e4478882ff3327d224d469 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/70f9e4973bac419f8b82465de3da6ac1 : SUCCESS in 17m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/604b7552eeb0497b86eba4c766daa8a6 : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":4},{"id":"3ebc460e0d993516f5c9bb8849db747bfb2c0885","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 20:22:24.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a657ce9d83bd4779801c432512b06201\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94f4fb14c065431db27ffaf69dbd5640 : SUCCESS in 1h 09m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a608e71b33547dd9267fd2ed9e239ac : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb3cc3b10c034581b328a3143b5fc8b7 : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3999708a6ae245a2abd66df1a073cb2a : TIMED_OUT in 1h 30m 50s","accounts_in_message":[],"_revision_number":4},{"id":"3ad4c9972e332d935df04b990cf652b32219d5ad","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-08-18 22:41:12.000000000","message":"Patch Set 4: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":4},{"id":"da493a394a62ea08d44105a754d351ee66691037","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-10 00:30:13.000000000","message":"Abandoned\n\nAlternate fix merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ed9aa86bcee17ddc651b2f43468ca4591beeee0d","revisions":{"d9e6b0ceb368ea7b291f3721e27b77a8e1f8ed23":{"kind":"REWORK","_number":1,"created":"2024-08-06 22:55:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/925825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/925825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/925825/1"}}},"commit":{"parents":[{"commit":"b400a1fdb3ae609d5f86d4d0f1b590640337b8f0","subject":"Expirer: add delete task iteration related metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b400a1fdb3ae609d5f86d4d0f1b590640337b8f0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-29 20:43:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-06 22:53:09.000000000","tz":-420},"subject":"Add helpers for adding/popping parameters","message":"Add helpers for adding/popping parameters\n\nThere are a bunch of places where we want something similar to HTTP\u0027s\nheader parameters, but we have enough control over the header, parameter\nnames and parameter values that we can avoid dealing with quoted values\nand the like.\n\nAdd two new helpers in swift.common.utils and use them where sensible:\n\n   append_header_parameter(header: str, name: str, value: str) -\u003e str\n   extract_header_parameter(header: str, name: str)\n      -\u003e tuple[str, Optional[str]]\n\nThis avoids us needing to reimplement the appending-all-the-other-params\nlogic all over the place.\n\nAdditionally, add two more helpers for dealing with multiple parameters\n(such as in the symlink middleware):\n\n   append_header_parameters(header: str, params: dict[str, str]) -\u003e str\n   extract_header_parameters(header: str, *names: str)\n      -\u003e tuple[str, dict[str, Optional[str]]]\n\nThese are just nice wrappers for the first two functions.\n\nAs a side effect, stop using cgi.parse_header; with py311 we started\nseeing warnings like\n\n   DeprecationWarning: \u0027cgi\u0027 is deprecated and slated for removal\n   in Python 3.13\n\nand sure enough, it\u0027s gone in py313.\n\nChange-Id: Ieb5207c550bff81b3c67d882153e2ec73d4d19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9e6b0ceb368ea7b291f3721e27b77a8e1f8ed23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9e6b0ceb368ea7b291f3721e27b77a8e1f8ed23"}]},"branch":"refs/heads/master"},"7f992d40c4d597380563b54cfba5445238e101dd":{"kind":"REWORK","_number":2,"created":"2024-08-08 23:20:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/925825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/925825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/925825/2"}}},"commit":{"parents":[{"commit":"6a5a681227f21c4663259cf3f63942ffe44091be","subject":"tests: don\u0027t mutate and re-use Response headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a5a681227f21c4663259cf3f63942ffe44091be"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-29 20:43:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-08 23:18:27.000000000","tz":-420},"subject":"Add helpers for adding/popping parameters","message":"Add helpers for adding/popping parameters\n\nThere are a bunch of places where we want something similar to HTTP\u0027s\nheader parameters, but we have enough control over the header, parameter\nnames and parameter values that we can avoid dealing with quoted values\nand the like.\n\nAdd two new helpers in swift.common.utils and use them where sensible:\n\n   append_header_parameter(header: str, name: str, value: str) -\u003e str\n   extract_header_parameter(header: str, name: str)\n      -\u003e tuple[str, Optional[str]]\n\nThis avoids us needing to reimplement the appending-all-the-other-params\nlogic all over the place.\n\nAdditionally, add two more helpers for dealing with multiple parameters\n(such as in the symlink middleware):\n\n   append_header_parameters(header: str, params: dict[str, str]) -\u003e str\n   extract_header_parameters(header: str, *names: str)\n      -\u003e tuple[str, dict[str, Optional[str]]]\n\nThese are just nice wrappers for the first two functions.\n\nAs a side effect, stop using cgi.parse_header; with py311 we started\nseeing warnings like\n\n   DeprecationWarning: \u0027cgi\u0027 is deprecated and slated for removal\n   in Python 3.13\n\nand sure enough, it\u0027s gone in py313.\n\nChange-Id: Ieb5207c550bff81b3c67d882153e2ec73d4d19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f992d40c4d597380563b54cfba5445238e101dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f992d40c4d597380563b54cfba5445238e101dd"}]},"branch":"refs/heads/master"},"1555215ee7f4943e80c61d6ea258bf25d45bab87":{"kind":"REWORK","_number":3,"created":"2024-08-09 17:05:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/925825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/925825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/925825/3"}}},"commit":{"parents":[{"commit":"6a5a681227f21c4663259cf3f63942ffe44091be","subject":"tests: don\u0027t mutate and re-use Response headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a5a681227f21c4663259cf3f63942ffe44091be"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-29 20:43:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-09 17:04:47.000000000","tz":-420},"subject":"Add helpers for adding/popping parameters","message":"Add helpers for adding/popping parameters\n\nThere are a bunch of places where we want something similar to HTTP\u0027s\nheader parameters, but we have enough control over the header, parameter\nnames and parameter values that we can avoid dealing with quoted values\nand the like.\n\nAdd two new helpers in swift.common.utils and use them where sensible:\n\n   append_header_parameter(header: str, name: str, value: str) -\u003e str\n   extract_header_parameter(header: str, name: str)\n      -\u003e tuple[str, Optional[str]]\n\nThis avoids us needing to reimplement the appending-all-the-other-params\nlogic all over the place.\n\nAdditionally, add two more helpers for dealing with multiple parameters\n(such as in the symlink middleware):\n\n   append_header_parameters(header: str, params: dict[str, str]) -\u003e str\n   extract_header_parameters(header: str, *names: str)\n      -\u003e tuple[str, dict[str, Optional[str]]]\n\nThese are just nice wrappers for the first two functions.\n\nAs a side effect, stop using cgi.parse_header; with py311 we started\nseeing warnings like\n\n   DeprecationWarning: \u0027cgi\u0027 is deprecated and slated for removal\n   in Python 3.13\n\nand sure enough, it\u0027s gone in py313.\n\nChange-Id: Ieb5207c550bff81b3c67d882153e2ec73d4d19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1555215ee7f4943e80c61d6ea258bf25d45bab87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1555215ee7f4943e80c61d6ea258bf25d45bab87"}]},"branch":"refs/heads/master"},"ed9aa86bcee17ddc651b2f43468ca4591beeee0d":{"kind":"REWORK","_number":4,"created":"2024-08-09 17:11:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/925825/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/925825/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/925825/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/925825/4"}}},"commit":{"parents":[{"commit":"6a5a681227f21c4663259cf3f63942ffe44091be","subject":"tests: don\u0027t mutate and re-use Response headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a5a681227f21c4663259cf3f63942ffe44091be"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-29 20:43:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-09 17:11:09.000000000","tz":-420},"subject":"Add helpers for adding/popping parameters","message":"Add helpers for adding/popping parameters\n\nThere are a bunch of places where we want something similar to HTTP\u0027s\nheader parameters, but we have enough control over the header, parameter\nnames and parameter values that we can avoid dealing with quoted values\nand the like.\n\nAdd two new helpers in swift.common.utils and use them where sensible:\n\n   append_header_parameter(header: str, name: str, value: str) -\u003e str\n   extract_header_parameter(header: str, name: str)\n      -\u003e tuple[str, Optional[str]]\n\nThis avoids us needing to reimplement the appending-all-the-other-params\nlogic all over the place.\n\nAdditionally, add two more helpers for dealing with multiple parameters\n(such as in the symlink middleware):\n\n   append_header_parameters(header: str, params: dict[str, str]) -\u003e str\n   extract_header_parameters(header: str, *names: str)\n      -\u003e tuple[str, dict[str, Optional[str]]]\n\nThese are just nice wrappers for the first two functions.\n\nAs a side effect, stop using cgi.parse_header; with py311 we started\nseeing warnings like\n\n   DeprecationWarning: \u0027cgi\u0027 is deprecated and slated for removal\n   in Python 3.13\n\nand sure enough, it\u0027s gone in py313.\n\nChange-Id: Ieb5207c550bff81b3c67d882153e2ec73d4d19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed9aa86bcee17ddc651b2f43468ca4591beeee0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed9aa86bcee17ddc651b2f43468ca4591beeee0d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
