)]}'
{"id":"openstack%2Fswift~925870","triplet_id":"openstack%2Fswift~unmaintained%2Fxena~I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1","project":"openstack/swift","branch":"unmaintained/xena","topic":"zuul-config-error-swift","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-08-07 14:40:23.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1","subject":"CI: Move off CentOS 8","status":"MERGED","created":"2024-08-07 12:51:25.000000000","updated":"2024-08-07 16:39:23.000000000","submitted":"2024-08-07 16:36:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925870-zuul-config-error-swift","meta_rev_id":"df2b1e60e9d522d6d1c897bdcc84e8fb8ae9fa3e","_number":925870,"virtual_id_number":925870,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-07 16:36:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-08-07 14:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-08-07 14:40:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-07 12:51:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-08-07 12:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61cfc1f764dbca11e68280a6f4f7502b07577970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-07 12:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e19f0cbf5fa41e277c7420c54099bf323b25f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:51:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch unmaintained/xena.  The\nproblem was:\n\n  Job swift-tox-py39-centos-9-stream not defined\n\nThe problem appears in the the \"opendev.org/openstack/swift\" project stanza:\n\n  project:\n      vars:\n        ensure_tox_version: \u0027\u003c4\u0027\n      templates:\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - swift-jobs-arm64\n  ...\n\n  in \"openstack/swift/.zuul.yaml@unmaintained/xena\", line 533","accounts_in_message":[],"_revision_number":1},{"id":"c337617229c299566ab75646b605b1205dbe1496","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:51:55.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch unmaintained/xena.  The\nproblem was:\n\n  Job swift-tox-py39-centos-9-stream not defined\n\nThe problem appears in the the \"opendev.org/openstack/swift\" project stanza:\n\n  project:\n      vars:\n        ensure_tox_version: \u0027\u003c4\u0027\n      templates:\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - swift-jobs-arm64\n  ...\n\n  in \"openstack/swift/.zuul.yaml@unmaintained/xena\", line 533","accounts_in_message":[],"_revision_number":1},{"id":"04f7b9461f7738a3afeb988cf7f08640a86c32aa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-07 12:52:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0614774806aa447d165804fc3a070b638155c72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-07 13:12:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1bd9e8d1ea487c16715502fae81ead66dae7ae85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:12:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7af0f9014dda422783a1b40f2c0ed19f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3956f9a2b4ae4579bf8ec8086ca9dadf : SUCCESS in 7m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7776f783ccc4cfdaf61acf20dc9e66c : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7403e0a43db436d93b12a0bbd909c8b : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89e95b860ce34c268e651ff740f605eb : SUCCESS in 25m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9be7539a2ff648d4ad669e6320c02de5 : FAILURE in 3m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/88f1f788a5cd43e6951449eba8d76a0c : FAILURE in 3m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/867dab2b01eb40d39f53c7a2ea22f28b : SUCCESS in 20m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6a6c843fb457497bb8334be351a6135f : SUCCESS in 18m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1baa8fb694124b67af08d5b615c590dd : SUCCESS in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1833ec7f23114c3b8f0249092571bb8a : SUCCESS in 18m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c354997604945c5a0a38ac32dcf2e35 : SUCCESS in 21m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b9217b4e999b4778844b8670b2bcb8c2 : SUCCESS in 23m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/47ed315d7b214ea98d4bc8b02877d4fe : SUCCESS in 21m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5297c1819fae4ef6aec7e43e9abbf306 : SUCCESS in 22m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fa2054f670054879b39096bf614be524 : SUCCESS in 21m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f45495566848457ea2fd73c00bca68f4 : SUCCESS in 18m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3d43b1f622104eaebadc6396ad9ae63a : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/79aca8a8384c49e08639bc50c6dc182c : SUCCESS in 6m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/28e3f4ae1e214a3c84c0df9db0fb97cc : SUCCESS in 31m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92aa20206ae848568aa3a8f3fca1dffa : SUCCESS in 30m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39edf09a9f6a4baea82b659996fdcb1b : SUCCESS in 16m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22fec16c0b524e34a08269efc5074f82 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f9af6dcd94a84b44ac50319114e88ba9 : FAILURE in 6m 26s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/215dcb1da9e7483091c1a43d3b5ee159 : SUCCESS in 47m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bdd3e419bac43eebb5b2b77a890018e : SUCCESS in 54m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/04ef78b87ba44820b7915f7b6016d863 : SUCCESS in 49m 16s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0c6a5a78dfe46d8b26a6ef2f8069b28 : SUCCESS in 52m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f454d4d8ab20c9c01280693cd1d838993239ae5c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-07 14:40:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a737abbe7f418bdc417c6dfdc9938c6d8f226afa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:41:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"883dae0077df1aa45722c0bd0ad98de4949e592b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 16:16:01.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e3f27ddb6324af789d13230bcd04a28\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67c0b5485cf54289aebb92a343b4ef88 : SUCCESS in 1h 03m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2977ffbacbc4afb9109f0814d7e06f5 : TIMED_OUT in 1h 30m 54s\n- swift-probetests-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/4a48a0bb2a624df384a345b866bdf1e8 : TIMED_OUT in 3h 01m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/967834612f184caca43bc3015cee2d9e : SUCCESS in 1h 16m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/da35a75bbdf4405ba9a7faa34af6ccc0 : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d6469ffe70d28432b8319dbb93d2ac5bc813d0fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 16:36:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c721e4417d14c59a5f50b23be7341cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ad8d46d75d4c9cb3c8e400820148ed : SUCCESS in 9m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2dc8f8337f044de8badd3028a8481874 : POST_FAILURE in 6m 25s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b002a341b303487698508af5922c36fa : POST_FAILURE in 6m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6223f47757004670b1b414a7b9cd681a : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e5eb09216e7b439881ea4f79ca6cfd32 : SUCCESS in 13m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b520048de7df40a4accb445eaf0c2147 : SUCCESS in 16m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0ee18857721f494a88e2baab17aa77b6 : SUCCESS in 17m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dc7220ba72b5480bab2686552d4ce239 : SUCCESS in 14m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3393907885147bd8e20b9e738ee3b43 : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11dbde7ee7b042e1bb3b525bde8cf530 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dce2863dcb464c81b1c2de5c0c5c4892 : SUCCESS in 23m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7da6425f711e4f75a6b9aa51b268a4aa : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e879bdeee2e1486b8b5e45bc7c658677 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7930aec738e5499f80ea255d10d56120 : SUCCESS in 23m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d246f9b9e3d643558512cdea2a0adf43 : SUCCESS in 5m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/53c95ba263584b8bb8cd4b263a5780bf : SUCCESS in 57m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f01f0505efe4996a95aa4ca2d2a9474 : SUCCESS in 26m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6fd2ace98118480f83ba02a608747a4a : SUCCESS in 30m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d53181e199cf428e9288f2b4427590a5 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a56164616bf4da69a5f13f523cf25c2 : SUCCESS in 5m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/187a7805ee644167b152469ce8ba361b : SUCCESS in 45m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bb13e0b39ec4a9ba48f932ca9b0326f : SUCCESS in 53m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63c27702ce8349e48e7c4ef94c1d9a73 : SUCCESS in 46m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eef63425f5bb8dd5b4f0799a8b61deb4839bd693","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 16:36:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"df2b1e60e9d522d6d1c897bdcc84e8fb8ae9fa3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 16:39:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16dc960077534ffa8f89e44030e3982c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac7eb94ef534a1499cee9e06b548eb3 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/06a2de97ef7648d596676262cec481b4 : FAILURE in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/fe7e35bcd187456a972b43a1b28a7b75 : FAILURE in 25s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eadb2acde6849aa1e78cfb0a12a9396ff5ccd40b","revisions":{"5aa9e730947ce0edec3a8f5171ff24baf067d2c0":{"kind":"REWORK","_number":1,"created":"2024-08-07 12:51:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/70/925870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/925870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/925870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/925870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/925870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/925870/1"}}},"commit":{"parents":[{"commit":"8c4f9526ac62ee23e60f714d9bb806589db6d8a2","subject":"Clean up centos-7 based jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c4f9526ac62ee23e60f714d9bb806589db6d8a2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-04 15:58:19.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-08-07 12:50:01.000000000","tz":120},"subject":"CI: Move off CentOS 8","message":"CI: Move off CentOS 8\n\nThis is a combination of 3 commits.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move probe tests to centos 9 stream\n\nPin selenium to 3.x for now, until we can run down the issues with 4.x\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move off CentOS 8\n\nRemove swift-tox-py36-centos-8-stream job entirely.\n\nMove the following jobs to CentOS 9:\n\n  - swift-tox-func-s3api-ceph-s3tests-tempauth\n  - swift-tox-func-s3api-tests-tempauth\n  - swift-multinode-rolling-upgrade, as well as the other rolling\n    upgrade jobs\n\nRemove the swift-multinode-rolling-upgrade-victoria job, as py39\nsupport (required for CentOS 9) was not added until wallaby.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Migrate CentOS Stream 8 FIPS job to CentOS Stream 9\n(cherry picked from commit 6fd031055c951dc84a8f80252985b80edb5a6971)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nConflicts:\n  .zuul.yaml\n\nNOTE(elod.illes): this patch is backported to unmaintained/xena as\nCentOS Stream 8 is EOL and centos-8-stream nodeset was removed from\nZuul and it gives a config error: \u0027The nodeset \"centos-8-stream\" was\nnot found.\u0027 Official Xena runtimes [1] only contains CentOS Stream 8\nand python 3.6 \u0026 3.8.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/xena.html\n\nRelated-Change: I596415d17f77f48a6e8a63a61b734a8ca0865847\nChange-Id: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\n(cherry picked from commit 2032fb214eeb41cbb30a0c71525d0db9e4899760)\n(cherry picked from commit 9fa1b7edec54f855ae559709982a22d4aa57be79)\n(cherry picked from commit 1a18d3c16bbec95da54aa6b2a8c3e1469181d799)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5aa9e730947ce0edec3a8f5171ff24baf067d2c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5aa9e730947ce0edec3a8f5171ff24baf067d2c0"}]},"branch":"refs/heads/unmaintained/xena"},"eadb2acde6849aa1e78cfb0a12a9396ff5ccd40b":{"kind":"REWORK","_number":2,"created":"2024-08-07 13:12:01.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/70/925870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/925870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/925870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/925870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/925870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/925870/2"}}},"commit":{"parents":[{"commit":"8c4f9526ac62ee23e60f714d9bb806589db6d8a2","subject":"Clean up centos-7 based jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c4f9526ac62ee23e60f714d9bb806589db6d8a2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-04 15:58:19.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-08-07 13:11:54.000000000","tz":120},"subject":"CI: Move off CentOS 8","message":"CI: Move off CentOS 8\n\nThis is a combination of 3 commits.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move probe tests to centos 9 stream\n\nPin selenium to 3.x for now, until we can run down the issues with 4.x\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move off CentOS 8\n\nRemove swift-tox-py36-centos-8-stream job entirely.\n\nMove the following jobs to CentOS 9:\n\n  - swift-tox-func-s3api-ceph-s3tests-tempauth\n  - swift-tox-func-s3api-tests-tempauth\n  - swift-multinode-rolling-upgrade, as well as the other rolling\n    upgrade jobs\n\nRemove the swift-multinode-rolling-upgrade-victoria job, as py39\nsupport (required for CentOS 9) was not added until wallaby.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Migrate CentOS Stream 8 FIPS job to CentOS Stream 9\n(cherry picked from commit 6fd031055c951dc84a8f80252985b80edb5a6971)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nConflicts:\n  .zuul.yaml\n\nNOTE(elod.illes): this patch is backported to unmaintained/xena as\nCentOS Stream 8 is EOL and centos-8-stream nodeset was removed from\nZuul and it gives a config error: \u0027The nodeset \"centos-8-stream\" was\nnot found.\u0027 Official Xena runtimes [1] only contains CentOS Stream 8\nand python 3.6 \u0026 3.8.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/xena.html\n\nRelated-Change: I596415d17f77f48a6e8a63a61b734a8ca0865847\nChange-Id: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\n(cherry picked from commit 2032fb214eeb41cbb30a0c71525d0db9e4899760)\n(cherry picked from commit 9fa1b7edec54f855ae559709982a22d4aa57be79)\n(cherry picked from commit 1a18d3c16bbec95da54aa6b2a8c3e1469181d799)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eadb2acde6849aa1e78cfb0a12a9396ff5ccd40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eadb2acde6849aa1e78cfb0a12a9396ff5ccd40b"}]},"branch":"refs/heads/unmaintained/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
