)]}'
{"id":"openstack%2Fswift~925875","triplet_id":"openstack%2Fswift~feature%2Fmpu~I37e23422ff435004c29accc17e4335313dfe6d71","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-04 18:11:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I37e23422ff435004c29accc17e4335313dfe6d71","subject":"Merge branch \u0027master\u0027 into feature/mpu","status":"MERGED","created":"2024-08-07 13:54:07.000000000","updated":"2024-09-04 20:11:17.000000000","submitted":"2024-09-04 18:11:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"925875-merge-master","meta_rev_id":"f9f5713bb67af10d901a52a4b76959c5c5b683ee","_number":925875,"virtual_id_number":925875,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-04 18:11:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-04 16:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-04 16:37:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-07 14:46:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-07 16:21:46.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-09-04 16:37:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"4ba60ea41acbc85788d5ba9ce0425d301701db7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-07 13:54:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1282e3e7dc1f82c8995c0419fe9d2f442a41b0d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:46:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cd7cd08d4ce4ef3b9a9bb48d035ade5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/65bf37bb004d4644a4b28c26821058b8 : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d775f304b7d4696967377585ad285a1 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4821ca83ef334875b5b7d540ffc9bc58 : SUCCESS in 5m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07f0a82b66ad44f38f7d10d1dbc2794b : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6c8f575b234404d842c74067f2fc4f9 : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3d49d658c56496eb70f7e9af60b8cc4 : SUCCESS in 20m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f5574da8a954f0ca73ad40b197455e8 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4dcb28f8a2dc43a393b1fc7200d27ead : FAILURE in 18m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/590827d0e9994fb6ba435e578bdc9158 : FAILURE in 18m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/baf6e0efe6134dc4ae2b87bf43c90def : FAILURE in 19m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8bab24e1fbd34dc29e5fa98d76900633 : FAILURE in 21m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e5b3900c98fa451e95114e169a05ebdf : FAILURE in 29m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a9c5ad50e1c495c92a79e47e47d18e0 : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a39476f736bf430593f207ed6673ca52 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0debb06117cc4b0cb20564a8f8e1e787 : SUCCESS in 19m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6db03633d25d486cb81a414bb69d9325 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1f15df67c4a54a3ea8ce9d00eaee2965 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e6335d5dbfef474496cdfbcc07091143 : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d11e2213361495b97f9a7ccda1210e3 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7356595413614b8889e4d7fad3e06641 : SUCCESS in 15m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5202ef35345c42339adc97c8752f720c : SUCCESS in 6m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ad4efa6fc9c488eb5e683e29d832a8f : SUCCESS in 48m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4b11d5575bd4029943c8b0e66b2c546 : FAILURE in 22m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87de8b9c1d57442fa685d0a613b6db62 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c8b5593b6aaf4c2c82c000c8353c4169 : SUCCESS in 19m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5c56872ef6b4b1c47f22397404f464485a630583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-07 15:05:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fac18bceab793b3b18b4fad289035c273a482d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 16:03:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f801d3fd8e4f4af1bcd3823014a7fb30\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/12d0be7ed3184119b2faeb7d02830f63 : SUCCESS in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fbeff7f4da249e6ae7db9f05713097a : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0ea2a8dea924d0e9348c50a37032951 : SUCCESS in 4m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eef6c11c17de4bbca7992f62f52daa5b : SUCCESS in 21m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5a9ab13af264c1c8bcfa8d3050c809c : SUCCESS in 21m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0dbd3d62651489fb520c16b78b8216d : SUCCESS in 22m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/632c2b66d7a94d948764d0a748b2f3b4 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/387477594bfe41cb99b9cc4f43da175d : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c12204f52aa497baa54441ca7f6c745 : SUCCESS in 19m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f10d5f54f96645f6962bd6d35f51f667 : SUCCESS in 17m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e24eca4863124a9aac3b30c8ee5abb92 : SUCCESS in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bd4f37de4be542d4b93020035d684496 : SUCCESS in 24m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/628a409b16d24d16978da02f561dc87c : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/821dd2c1e2754d9c8caea00faab66aea : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf80d0768e4040e1aaf2a7285a8e5311 : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/804509540b1a43748a3ecd811c54b84e : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dbb3a2dfea2e451188598117dc68efd1 : SUCCESS in 19m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/14031c5d505f4cdf87f00ea1684e4e02 : SUCCESS in 20m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e73eb2e3ed724203ab262022545b4036 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84ce9528f2424b6bacf466da6e71f1b2 : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1e6cd9413ded41cfae0ad91a36fc624f : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c63be3a9f3294831a6d227f17d9122e3 : SUCCESS in 52m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76707c4d48de46139202b904a90fa778 : SUCCESS in 22m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b278cbaa82ea4ee0bf40ac330d969def : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3567ff126ff0467787caf28876a51f4a : SUCCESS in 15m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0c7721b9691c2469144f5f9a310b2ce45051267e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-07 16:21:46.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4c594d856e1081c9d6a809f0268cd6ddace8205e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 19:17:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f641333abf5348de89d12f26195a8173\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8f1f08f6f40407baaa4f19782130cc1 : SUCCESS in 57m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce76640ce32a47eea2b9fcf64725ef3b : SUCCESS in 1h 16m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7453ee0f60a47f692c40ba889edd212 : SUCCESS in 1h 08m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/02cfc0db22f54f009c57881218bae692 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":2},{"id":"5c936840761348fc8bda1ceb04281dcc69ac56a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-08 09:31:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a020018b7589e9531093b205895f494fab3641fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 10:30:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5c6e5296dee40f7a2e64e693a5e64d3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e76445396bc434db70eebe5ad58c1a5 : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/564df7e675f4437bb40778751d526fd3 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/212ebb9bdd444f4db480f62baf94cb4d : SUCCESS in 5m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad9a5d55934c4e0c89e451269c394d45 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6ed6a6751be4179af5eaca37d406fda : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec8a7ae1d50040c98835d465f7fb990c : SUCCESS in 19m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6519dc1a831d4e6caf250665d4c97e90 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70249a2f86ea4b1d9405e844d6fa3941 : SUCCESS in 18m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80c50ed981cc4badb5656c5ad4290aae : SUCCESS in 14m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4519f996240d4b5e986ad93379b30b50 : SUCCESS in 19m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8c11969b6b2b45b4b9281ec01601c8ba : SUCCESS in 18m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c3f80a3bfa8d45cabd17b01eb7ba33f2 : SUCCESS in 24m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ee73ee634b3447d9ecfd08c2ec11f4a : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d198959f22274bfca7277cc07c9cb534 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0584a8e51c1e4d1db17e1340d09a8943 : SUCCESS in 19m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3c2b4664e7ec4d3285842af480101593 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5199b130db514e07911474430c148254 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/adae0cf4a7114728abe9b7e0166b60b1 : SUCCESS in 19m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/42ec413fa6234b5e864ab854abcbc8b2 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1af61383eb44b16bea3f14973b099bb : SUCCESS in 13m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c933911f4e4e436881957c40ca36e197 : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1de0b5bdb314f0887211e2db0115a57 : SUCCESS in 58m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/672b50e6d61e4d2aba3b8723af8e49bf : SUCCESS in 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5b0f0bd8834139b1afada24974348b : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bdc94fb44b30469d8a096a2cbf718f87 : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":3},{"id":"f7f0e37b49beba1137ecda1e2008cf13e972c51f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 11:21:54.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/836e5166a3854f1c8e8009745188ef03\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b30f2eede11745b08c15b4acf4168d5f : SUCCESS in 1h 02m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc6ba90037a84d59860b14f1903fa7d6 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/420a26713be1481e87caeb22db491c6a : TIMED_OUT in 1h 30m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b94e654139f04a6b8a0569022043f812 : SUCCESS in 1h 19m 14s","accounts_in_message":[],"_revision_number":3},{"id":"7010ac0fa8f269393fe11ad28d1dcdd979163892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-04 15:52:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"678b52798fde7e2699b0c5a1d68ac2835f52e987","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-04 16:37:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"42b644adece7ff462659b7bed3891a5ecec19e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 16:49:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eedca73dbe840e3847167987a7af2ae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5b3df0e88474a2992cd2b37cc7d8226 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dc59e76f00b4130aa4696c7f5b9b042 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a00a984267543faa528f94b75b1adf4 : SUCCESS in 6m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/697a4d84931b447bab2b312c8ca7a600 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f82efcc12e3642848f9e403bd5da1883 : SUCCESS in 24m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b461be4f1f184a45b218b3c4ee7ef504 : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/abcdfcbd4de247c2b0c739b3eec2a758 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ecaeea1ed974c3bbbd05e7d0b61aef3 : SUCCESS in 17m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d68721e31bff44e28da50dd12816996f : SUCCESS in 19m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/821d548117c849f58c7b7986aaf1f587 : SUCCESS in 23m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c5d2e4d09ed9444497098e8a9047e94f : SUCCESS in 20m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f8b07e50332f48f4aba6e3ba982808c9 : SUCCESS in 28m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9d6834ada1f1498394625312cf469603 : SUCCESS in 19m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29c3a2f4d553411597878ae5d872666b : SUCCESS in 21m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b094739d6b0d4ae6b3bf702f99112e9c : SUCCESS in 20m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ade6378a963b4468962fab0808383e53 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7503c6886d9341d6a72bfda605faabac : SUCCESS in 16m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d1beb97970a249eb882497e9b12ea4a4 : SUCCESS in 27m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e98bda0a44944bde90b33608502dff23 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dddafab02502479f8c97367d4cf0ccc6 : SUCCESS in 16m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8405d8a113044ea0b49107c1b2db5b93 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0196669b220147a3ac8a9d8f456bb862 : SUCCESS in 51m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc0c86065bd54653bc04660b325abd01 : SUCCESS in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e2786910ae49cb962623a45a34dceb : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e32d2a5f7ea4187aa8cc7fed2eddc49 : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":4},{"id":"f019f46e00c25249f3273428cffa7c0633127404","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 16:49:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"eeee58f257a7e3156b47c0d570e4964b087474cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:11:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a757def17844cfb936a1c7c197b3f19\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac2134348c70489e869ff2fcde8ad44f : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d46fd16a05412ea989cd9a76d8f783 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b7874f7e9d3409eaaab83401b361a4e : SUCCESS in 6m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/85527cb24bc64792a4fd71e51a3853d0 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d2480e0eb677463ead8a51103a17f747 : SUCCESS in 19m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/545f4973383e49ff9112c1e5c66bbc09 : SUCCESS in 17m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/da4492eaea594d5fa89323bd824bd4b1 : SUCCESS in 21m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ec9c5fc23bb94407b018fbf894ccda7a : SUCCESS in 18m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ac97982b8d044efe827740407c341c65 : SUCCESS in 20m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c6dbc7d7f074787b838a2231c48ff8a : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c8e832d96d64ef69cb6fca8aa4847d7 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa5f7d00ca394ae998fcf870db794980 : SUCCESS in 19m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/66ba4048188942939ff1f068657e4aa9 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/759d47b093e8488ebecd99884c7a158e : SUCCESS in 17m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ac080b6b6e11450f98ffe52ebd04d6e6 : SUCCESS in 18m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b35449f6c4c474f9279e700d995c2d3 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/48ee2c7f6d6c48739d621cf5beb3f269 : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1d3e36851e240d2b53288557d8e2a22 : SUCCESS in 52m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da05783ff1d946379ffd828dd21ee583 : SUCCESS in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53658fe6ac544d4aa4c17594018f67e7 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71f712d9c5974d4ab6dce313af63f5e4 : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":4},{"id":"c99947491932391fde51bd8f19fd5856692988e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:11:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"96ac8da2da38e299bdca6a7ca565a289c8221b8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:15:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2743460ed0c440fb39b54f9e267527e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae4070b529e4261b94ada6aa3d6bcaf : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba1f5fb0526b4f6f9a1f818c62e0186f : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c8dce03345ff485ca655c9f7059d7ccc : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f9f5713bb67af10d901a52a4b76959c5c5b683ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 20:11:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1fffd968a841488539cffc8a7c68ba\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c18d85c8496f4a62a6021c11d56df60c : SUCCESS in 59m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/155de716053e4841a31fcb318c576e3b : SUCCESS in 57m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/72c6b420da414ccfbde3252b3af5558e : SUCCESS in 58m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f637beef64e0449dacefb5f65a5d6608 : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"31b3823148bbda17d97a34b1fd7dd5cb389a91bb","revisions":{"525b2a386ff04cd39d3b3be7ce671a7e55864af6":{"kind":"REWORK","_number":1,"created":"2024-08-07 13:54:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/925875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/925875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/925875/1"}}},"commit":{"parents":[{"commit":"51fd293f1f947f6972e4fa6751f94f750bbd86f2","subject":"Merge remote-tracking branch origin/master into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51fd293f1f947f6972e4fa6751f94f750bbd86f2"}]},{"commit":"9d14c7c830ba92dfa9bcb59efee367c66001479b","subject":"expirer stats: add test coverage for \u0027errors\u0027 stat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d14c7c830ba92dfa9bcb59efee367c66001479b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-07 13:53:51.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-07 13:53:51.000000000","tz":60},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: I37e23422ff435004c29accc17e4335313dfe6d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/525b2a386ff04cd39d3b3be7ce671a7e55864af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/525b2a386ff04cd39d3b3be7ce671a7e55864af6"}]},"branch":"refs/heads/feature/mpu"},"ee61b971c7e851ff1ace1af7ff6b0187151b4f59":{"kind":"REWORK","_number":2,"created":"2024-08-07 15:05:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/925875/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/925875/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/925875/2"}}},"commit":{"parents":[{"commit":"51fd293f1f947f6972e4fa6751f94f750bbd86f2","subject":"Merge remote-tracking branch origin/master into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51fd293f1f947f6972e4fa6751f94f750bbd86f2"}]},{"commit":"9d14c7c830ba92dfa9bcb59efee367c66001479b","subject":"expirer stats: add test coverage for \u0027errors\u0027 stat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d14c7c830ba92dfa9bcb59efee367c66001479b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-07 13:53:51.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-07 15:04:57.000000000","tz":60},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: I37e23422ff435004c29accc17e4335313dfe6d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee61b971c7e851ff1ace1af7ff6b0187151b4f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee61b971c7e851ff1ace1af7ff6b0187151b4f59"}]},"branch":"refs/heads/feature/mpu"},"13ac8f3641066886c4ba56200c81b92f2d7ce365":{"kind":"REWORK","_number":3,"created":"2024-08-08 09:31:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/925875/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/925875/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/925875/3"}}},"commit":{"parents":[{"commit":"51fd293f1f947f6972e4fa6751f94f750bbd86f2","subject":"Merge remote-tracking branch origin/master into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51fd293f1f947f6972e4fa6751f94f750bbd86f2"}]},{"commit":"6a5a681227f21c4663259cf3f63942ffe44091be","subject":"tests: don\u0027t mutate and re-use Response headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a5a681227f21c4663259cf3f63942ffe44091be"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-08 09:29:59.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-08 09:30:57.000000000","tz":60},"subject":"Merge master into feature/mpu","message":"Merge master into feature/mpu\n\nChange-Id: I37e23422ff435004c29accc17e4335313dfe6d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13ac8f3641066886c4ba56200c81b92f2d7ce365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13ac8f3641066886c4ba56200c81b92f2d7ce365"}]},"branch":"refs/heads/feature/mpu"},"31b3823148bbda17d97a34b1fd7dd5cb389a91bb":{"kind":"REWORK","_number":4,"created":"2024-09-04 15:52:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/925875/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/925875/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/925875/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/925875/4"}}},"commit":{"parents":[{"commit":"51fd293f1f947f6972e4fa6751f94f750bbd86f2","subject":"Merge remote-tracking branch origin/master into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51fd293f1f947f6972e4fa6751f94f750bbd86f2"}]},{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-04 15:51:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-04 15:51:41.000000000","tz":-420},"subject":"Merge branch \u0027master\u0027 into feature/mpu","message":"Merge branch \u0027master\u0027 into feature/mpu\n\nChange-Id: I37e23422ff435004c29accc17e4335313dfe6d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31b3823148bbda17d97a34b1fd7dd5cb389a91bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31b3823148bbda17d97a34b1fd7dd5cb389a91bb"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
