)]}'
{"id":"openstack%2Fswift~925910","triplet_id":"openstack%2Fswift~master~I4d32146f4d09ed2354cb4c757555952dca109263","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-08-07 19:34:59.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4d32146f4d09ed2354cb4c757555952dca109263","subject":"tests: Don\u0027t actually monkey-patch when testing the updater","status":"NEW","created":"2024-08-07 17:46:53.000000000","updated":"2024-08-07 21:19:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6fb4b0aa72c7a6d8cdb233d19f53cca8fb530ce0","_number":925910,"virtual_id_number":925910,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-07 18:39:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-07 18:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-07 19:34:59.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"}],"messages":[{"id":"6d0df109c1c0fdf12ea286a241c179e471f919db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-07 17:46:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd452c3ef0b401a133c97763c96b9ebf7b6bb8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 18:39:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9511ba9ecdcd4a07b68b967d72ef5932\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c9ba433e1654ba099bf1112ca017b0e : SUCCESS in 9m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2eb96304ccc94c4ea9bcf9bfff7ae4b5 : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/798ecff2228148b9a87f15bf2e4f4e61 : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15c717d8a8d34b7b92b7e003d558742b : SUCCESS in 21m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8d5c1501e7941a3869f6e4988813f7a : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3a1e02c78d24033a49556da8ce1ca3c : SUCCESS in 20m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb13abde81cf4c31b21cc7289b0bf820 : SUCCESS in 20m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/07e69c9175da4069abfcf1d76c4db576 : SUCCESS in 16m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d23612cfa02941c9891402f4b5de2c82 : SUCCESS in 18m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/158381cec470426e8a2757717c3b0c7f : SUCCESS in 25m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/937bc2b03b834f6db30314a2e15145ef : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/190bb0c0a2ca48a08a283782a2cea064 : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e65b752aa2bc4bbdb014bf5c6b3240f5 : SUCCESS in 24m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ca58cfdfe88349ff946a837f4219a2bd : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/09173c6869fb42ae92fa95def4374ba8 : SUCCESS in 21m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4c37fcf82c104c4e9c9580138e346c39 : SUCCESS in 19m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2cf771b9669747189dadc78190413b2d : SUCCESS in 47m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/784357ba3ad44359b65f45e4f236fc15 : SUCCESS in 44m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cda009ae09e4d23ad2e61fddbbe4a35 : SUCCESS in 27m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f0c7d9865c44b78843c19b34446720 : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4bfae7190a494b019e8239741021561e : SUCCESS in 17m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a17d5a78827aed38d42332476c784b409e814d5c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-08-07 19:34:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6fb4b0aa72c7a6d8cdb233d19f53cca8fb530ce0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 21:19:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2320bbfde6504da0b26857a9384425ca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f3d3ab0950b42599a0335490040c406 : TIMED_OUT in 1h 30m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/914ff8ff69304e8289d4fe527ab57346 : SUCCESS in 1h 02m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/53032995c2954bcc8d339ec666b91a11 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6015a46039249afa02f0d063be31f11 : TIMED_OUT in 1h 30m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2e79950ac6b4146472f8a1578772332458c4e1c","revisions":{"a2e79950ac6b4146472f8a1578772332458c4e1c":{"kind":"REWORK","_number":1,"created":"2024-08-07 17:46:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/925910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/925910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/925910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/925910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/925910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/925910/1"}}},"commit":{"parents":[{"commit":"9d14c7c830ba92dfa9bcb59efee367c66001479b","subject":"expirer stats: add test coverage for \u0027errors\u0027 stat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d14c7c830ba92dfa9bcb59efee367c66001479b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-07 17:24:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-07 17:46:24.000000000","tz":-420},"subject":"tests: Don\u0027t actually monkey-patch when testing the updater","message":"tests: Don\u0027t actually monkey-patch when testing the updater\n\nI saw some weird test delays that only seemed to manifest when running\nthe full suite of tests; digging in more, it seemed that roughly a\nminute was being added to TestContainerUpdater::test_error_in_process\nin test/unit/container/test_updater.py because of a gc.collect() call\nduring monkey-patching.\n\nChange-Id: I4d32146f4d09ed2354cb4c757555952dca109263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2e79950ac6b4146472f8a1578772332458c4e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2e79950ac6b4146472f8a1578772332458c4e1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
