)]}'
{"id":"openstack%2Fswift~926131","triplet_id":"openstack%2Fswift~master~I666bd84edff4a9ca6f38c450d953f5421be5e15a","project":"openstack/swift","branch":"master","topic":"log_prefix","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-08-14 14:34:22.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-09-02 15:59:42.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I666bd84edff4a9ca6f38c450d953f5421be5e15a","subject":"proxy_logging: deprecate [access_]log_statsd_valid_http_methods option","status":"NEW","created":"2024-08-12 11:14:36.000000000","updated":"2024-09-02 17:19:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"094c3e766bb25ed3822a931deb5271b17f1cb675","_number":926131,"virtual_id_number":926131,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-02 17:19:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-12 12:23:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-14 14:34:22.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"9c369e085966955ab8a40f8e6abb8835e81e289c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-12 11:14:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afffe2f607a03dd81d066850878d2c1598efa8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 12:23:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e6ab6218d544f2183c12cdcbe2cdfe8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8982669a9c49208e3b046334d21c3d : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/3311fc0b37d84bf1a54936c4089b3b4b : SUCCESS in 57m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c186f10b91549e681579e3b85479628 : SUCCESS in 55m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/700973f70a7b4df2aaa045a4b8c9c31f : SUCCESS in 49m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7db743d6930c4b39a87421570f06ce4a : SUCCESS in 22m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb8697df5a5a47cfbc046c263ca29c19 : SUCCESS in 22m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c0e93c1ad3747a09b82b8f548032364 : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2923895516fa419096433b7a51815605 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d4ed8d114e8f491bbd8e5ae3df3b4e7e : SUCCESS in 19m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7304083a844b4283aa3d12dc45b70cc5 : SUCCESS in 17m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/958379cafa3f4c0787e0fe09db07cd00 : SUCCESS in 14m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a4923293e3fe4b16b293e6d16440e718 : SUCCESS in 17m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/43bdeec0d63546438e863d8a61592104 : SUCCESS in 18m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/66b4686e5bbd4021b01a5df2bf507976 : SUCCESS in 20m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/12918e1f95254db7951c97631724dbcf : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4630e494d6644a3b91756586a33effa7 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9394c2a5182e4c6db0b16af13e3f0de2 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d2c96fcf048746d6a5406439499fc0a4 : SUCCESS in 18m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/13b97122714247ef8c9bdf84b7bd56be : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/feb6d1dd3f3d4649bf9f5eff3a273803 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a3e68a47051473f87deff28ef621c8b : SUCCESS in 13m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e9174eaf843c4d9bac78d5dff980ec72 : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab45fcbce02843ea8179e5b3c64584ba : SUCCESS in 46m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e84a144bb9ac492b9806cd45bb27caa7 : SUCCESS in 40m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5b47ed0302c413298702d1deec1ba78 : SUCCESS in 45m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63d0b93045384353a288b36c3fad6a60 : SUCCESS in 21m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee457909f2884f05b91e7da43afe4bf1 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e46f0c9fd6547399ebb263d1b5493e8 : SUCCESS in 17m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e43544dc6274237b63653f55821ded6 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b5dd21d2c4374812bcb5f87003477205a56f2a1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 14:09:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/674525a4cf674587878c4de3ca7ea96c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6edb6b5cfc3f4078a159e1c029034bfe : SUCCESS in 1h 19m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1476d6c8abb14e97875c0ff867260d48 : SUCCESS in 1h 21m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9eed3d3b66c04ec1a3b9db8fc99676da : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/857516238b92470d8d6fe889d02aac6e : TIMED_OUT in 1h 30m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8f1ed8a13bd7692449c1d2d622384f03c49fed84","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-14 14:34:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"190e7abb57c30545fda551f1612af9f6bcf7ee37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-02 15:59:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dbab41d83865432d18cec5d2134f84d34f382d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-02 16:01:11.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8ffe97b10b2f00af9c57817af272f14c59b5dd0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 17:17:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2dbe37cf1de48999b61fdb4310ef5ca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b62729924f04bd194e1ee9ff99ab715 : SUCCESS in 52m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7db9ce9781db43269e3d359613aa95b2 : SUCCESS in 57m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7991678485e448bbc7122546dcf6487 : SUCCESS in 1h 14m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f94bdfb7a7d4e38a358dd802ffae591 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":3},{"id":"094c3e766bb25ed3822a931deb5271b17f1cb675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 17:19:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5c9118f438e4d87a1549ab2962f8be8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01106a1ae6ed4717b4964c98c48b5cca : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/a981451f77764cbd973c3ee795a967b4 : SUCCESS in 56m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3121db2e31b64f2493d762f601d9aef8 : SUCCESS in 37m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/904ef618d3b74bd2800154b87eaaa639 : SUCCESS in 1h 04m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6d73012d85d4aa3a8760ff1e769c5fa : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e74de39c9454218859ebc043bdee3db : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aade2772e68e45c6902d23e47114a895 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c30cd782bc1f454d833f685465c8b594 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a71095c8497496cac206ecfb939e563 : SUCCESS in 16m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9e0779b32ef41a7afa1c6b645c534cb : SUCCESS in 18m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7159c9a2a4494e4cae10dfe22051af67 : SUCCESS in 18m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/484dba5306ed4c9c8a79e984d289979a : SUCCESS in 26m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/23f10735570148a2b9ebad0f0c1b9816 : SUCCESS in 29m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a39951c03bb04a11baf6316924841db2 : SUCCESS in 19m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc837c8aa8af4535a3a241f41e9f1a88 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/970a04d6552d420b925df3788bdc76a6 : SUCCESS in 18m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/27ed445f09eb4b2abefeea160b829ca2 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cf525493620e475bacc70f35613a4b4b : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bcc8e233a6964d51b2e688527e28de7b : SUCCESS in 16m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad698b2028cc4728ad830131f54660d7 : SUCCESS in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/552e85bbebf94f8286e812b91aba88bd : SUCCESS in 15m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/24b1ead24051454c9c0f8982ab74649f : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/87564ba7f38e4ae0a4c80113069ae904 : SUCCESS in 40m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b6a3be3d29a402c9e4deadf87d3ce51 : SUCCESS in 48m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/384b1569be704537833fe820c7020cb8 : SUCCESS in 51m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4300f5058d84376956afac720e090e9 : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b30894f7be4d64beef31ab058e18b5 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d165c2a5b5ef4726a3ca5a781ede3401 : SUCCESS in 16m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88eda54fdc7240b685b6b5ae36b43bb9 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3f9c6ad66c196e3646857a49e064af2a1118bff4","revisions":{"153640144d683fb40ed58e66904bcfb99f4c7406":{"kind":"REWORK","_number":1,"created":"2024-08-12 11:14:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/926131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/926131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/926131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/926131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/926131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/926131/1"}}},"commit":{"parents":[{"commit":"969efd29b7ac91e5fff1388b4a5d7958467e0597","subject":"statsd: deprecate log_ prefix for options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/969efd29b7ac91e5fff1388b4a5d7958467e0597"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-12 11:04:40.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-12 11:14:07.000000000","tz":60},"subject":"proxy_logging: deprecate [access_]log_statsd_valid_http_methods option","message":"proxy_logging: deprecate [access_]log_statsd_valid_http_methods option\n\nThe Related-Change deprecated the \u0027[access_]log_\u0027 prefix for proxy\nlogging\u0027s statsd client options. Although the\n\u0027statsd_valid_http_methods\u0027 option is not a statsd client option, it\nis logical for it\u0027s naming to be consistent. This patch therefore\ndeprecates the \u0027log_statsd_valid_http_methods\u0027 and\n\u0027access_log_statsd_valid_http_methods\u0027 config options and replaces\nthem with \u0027statsd_valid_http_methods\u0027.\n\nA ValueError will be raised when instantiating ProxyLoggingMiddleware\nif both the supported \u0027statsd_valid_http_methods\u0027 option and one or both\nof the deprecated options \u0027[access_]log_statsd_valid_http_methods\u0027 is\nconfigured.\n\nRelated-Change: I49d40d10c75443def6476ed8c8d4cc4473291484\nChange-Id: I666bd84edff4a9ca6f38c450d953f5421be5e15a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/153640144d683fb40ed58e66904bcfb99f4c7406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/153640144d683fb40ed58e66904bcfb99f4c7406"}]},"branch":"refs/heads/master"},"d20e82587311eb2a28c310a03763de2dd694958a":{"kind":"REWORK","_number":2,"created":"2024-09-02 15:59:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/926131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/926131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/926131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/926131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/926131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/926131/2"}}},"commit":{"parents":[{"commit":"f73daf6e706b77387e1b490e99fe4525f58c8efc","subject":"statsd: deprecate log_ prefix for options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f73daf6e706b77387e1b490e99fe4525f58c8efc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-12 11:04:40.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-02 15:50:23.000000000","tz":60},"subject":"proxy_logging: deprecate [access_]log_statsd_valid_http_methods option","message":"proxy_logging: deprecate [access_]log_statsd_valid_http_methods option\n\nThe Related-Change deprecated the \u0027[access_]log_\u0027 prefix for proxy\nlogging\u0027s statsd client options. Although the\n\u0027statsd_valid_http_methods\u0027 option is not a statsd client option, it\nis logical for it\u0027s naming to be consistent. This patch therefore\ndeprecates the \u0027log_statsd_valid_http_methods\u0027 and\n\u0027access_log_statsd_valid_http_methods\u0027 config options and replaces\nthem with \u0027statsd_valid_http_methods\u0027.\n\nRelated-Change: I49d40d10c75443def6476ed8c8d4cc4473291484\nChange-Id: I666bd84edff4a9ca6f38c450d953f5421be5e15a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d20e82587311eb2a28c310a03763de2dd694958a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d20e82587311eb2a28c310a03763de2dd694958a"}]},"branch":"refs/heads/master"},"3f9c6ad66c196e3646857a49e064af2a1118bff4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-09-02 16:01:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/926131/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/926131/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/926131/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/926131/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/926131/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/926131/3"}}},"commit":{"parents":[{"commit":"3453d4ee6e260e78161a730dab1f052397977338","subject":"statsd: deprecate log_ prefix for options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3453d4ee6e260e78161a730dab1f052397977338"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-12 11:04:40.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-02 16:01:02.000000000","tz":60},"subject":"proxy_logging: deprecate [access_]log_statsd_valid_http_methods option","message":"proxy_logging: deprecate [access_]log_statsd_valid_http_methods option\n\nThe Related-Change deprecated the \u0027[access_]log_\u0027 prefix for proxy\nlogging\u0027s statsd client options. Although the\n\u0027statsd_valid_http_methods\u0027 option is not a statsd client option, it\nis logical for it\u0027s naming to be consistent. This patch therefore\ndeprecates the \u0027log_statsd_valid_http_methods\u0027 and\n\u0027access_log_statsd_valid_http_methods\u0027 config options and replaces\nthem with \u0027statsd_valid_http_methods\u0027.\n\nRelated-Change: I49d40d10c75443def6476ed8c8d4cc4473291484\nChange-Id: I666bd84edff4a9ca6f38c450d953f5421be5e15a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f9c6ad66c196e3646857a49e064af2a1118bff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f9c6ad66c196e3646857a49e064af2a1118bff4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
