)]}'
{"id":"openstack%2Fswift~926174","triplet_id":"openstack%2Fswift~master~Ie393075f79447627714692e3f01bb53e967a71e8","project":"openstack/swift","branch":"master","topic":"fix_fakestatsdclient","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-08-13 18:19:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie393075f79447627714692e3f01bb53e967a71e8","subject":"FakeStatsdClient: Stop issuing DNS calls for host.","status":"MERGED","created":"2024-08-13 07:21:45.000000000","updated":"2024-08-13 18:20:47.000000000","submitted":"2024-08-13 18:19:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"926174-fix_fakestatsdclient","meta_rev_id":"fd0b95933766807fa45a1b21af06797fd9a0573c","_number":926174,"virtual_id_number":926174,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-13 18:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-08-13 08:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-13 17:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-13 17:16:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-13 08:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-13 08:36:35.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-08-13 17:16:34.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f780517a68d2985268d01526f7f025d1be7b86b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-13 07:21:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0960be6883adf01314656d7cfb3dfb4ef927328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 08:12:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cad09254ceb462cb0101b6d0e655276\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d7da69cf3e4f73b1dfa68d4a3395b2 : SUCCESS in 7m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24a8cdf0bf324626bb33c8ef06e664fb : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd00a2f779f743f4b85a33dd768a98ff : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c894aee46cb45e388f167b2b322e850 : SUCCESS in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6df335fc1f74c6ea0bb309c3600bcbe : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e238b792e274219bd6f5573cfab7c2d : SUCCESS in 19m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c30c9343b5de4525825d9b64fede95d8 : SUCCESS in 17m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ea190a3335b844308e79c1a442a5a4a5 : SUCCESS in 20m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ffcfd8c679d7416796207c5d72167c81 : SUCCESS in 20m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c0e29579ed77429897c0c30c5bb64ae2 : SUCCESS in 25m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/034c56ed593c4beeb04cf285e0b28a49 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae72ae69a8984f54bebddbaf0fd4e431 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ae3168b48ac41b38ab956335d3b6b34 : SUCCESS in 17m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/274cdfffa275479791c2454a4e0213cd : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b7b762618eab4825ae73f7ba4ab2dd37 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ee01da42be654170a29dc01019f3b097 : SUCCESS in 21m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/761fd3cfda8a4f5494160b82e183b8df : SUCCESS in 5m 52s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e1b81cdb4c3541ed829e92c5d4253219 : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1e7f81181ff4d97af7943a294b0eef1 : SUCCESS in 44m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f67a30d6f5b84d6da6d02ab82eb9e805 : SUCCESS in 46m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c077f9ba9f042b49705a6d8b8beb193 : SUCCESS in 36m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef00991d4a7e4d71b9054d6587009641 : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e72b6ebbadb4d1ea974ef7d13351b81 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/370fd0f6dd774015858d21c05a4190a1 : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"37d43881809d1c38c3839d2e6755b45de2f3bf1e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-13 08:36:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f74247c7398fc2e3c5b4e9e98e872ec0b83e3980","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 08:55:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/506671e41979466abccc9cbf939fcb27\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4307afb4d4842278a9ec3907008dd8f : SUCCESS in 52m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec52c26d78d741128583e8e5eef0741e : SUCCESS in 1h 07m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d233dc5f6e0f4ef4bd524a315c0cbc2f : SUCCESS in 1h 01m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1b0d88b1f7e4dbb8f33f1a43b747429 : TIMED_OUT in 1h 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5d91757dd6688b1195d7ea4694132e67f4c5778f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-13 17:16:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e75b1a3d6e3cd6dd7d9474497ae2d24209d0281b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:17:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eeb54446dcfd1cfda580eaabe156f133b393acb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 18:19:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab5739eb5174415197173497a6064a15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e372fea79e41f08bb37543f1c70e3a : SUCCESS in 8m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/17306f1da1d9412380a3a79ef4be46cf : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/99f5456f5bed46eaa58efe68e8c47b91 : SUCCESS in 20m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dfa089986b90479091f5f7a8d9031ecc : SUCCESS in 20m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/44d80b89efc245f6856df981847af094 : SUCCESS in 19m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5e8093cbe82141878f8604a4748c5740 : SUCCESS in 20m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/576dd38b136e489da6be86f103e03d82 : SUCCESS in 21m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b978789a50684aec9f9adf2ac7f1f8af : SUCCESS in 21m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a2e9dfeeb1054533a1b7042978bec272 : SUCCESS in 19m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6145b742c58349f18118e762d4959a64 : SUCCESS in 21m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bf7515caa753405abad45f7df5f9e543 : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/53a87c1f30034920881b3c9bcd608020 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e16241ba9fce4972b3e0927b75f0646a : SUCCESS in 18m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4e14dda50db143ea8904384b4fff2da2 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db42a66c8f4a4ed9b284ab4f03c0b053 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6add170c4b54c558428c442282af388 : SUCCESS in 53m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e1381e30a311463281853f00e3cb444d : SUCCESS in 52m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8f2aad3135424296aa2effe5368bb20f : SUCCESS in 57m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e42a8b531564882b1373f9d21f3f007 : SUCCESS in 19m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3967046cbb884602a7f7bc48ffbec5e0 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/135a271ed8bc4b108c07907b205645b1 : SUCCESS in 21m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1d88351f21cc3ef47c2420853a9aa9b8a3408079","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 18:19:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd0b95933766807fa45a1b21af06797fd9a0573c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 18:20:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c97af87921be4edbb08b1116d5c5754b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02f733dd56f147ff8d3ed6170cc7c717 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6cc407733bc040118c263bdb6907b556 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6b73612d1dcd0bace2361338e6ae82eb591282b","revisions":{"e6b73612d1dcd0bace2361338e6ae82eb591282b":{"kind":"REWORK","_number":1,"created":"2024-08-13 07:21:45.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/74/926174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/926174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/926174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/926174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/926174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/926174/1"}}},"commit":{"parents":[{"commit":"1d5e65ce6a9cb61874fce6c86af6a8628eba9a8a","subject":"Merge \"Move utils/logs.py related tests out of test_utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5e65ce6a9cb61874fce6c86af6a8628eba9a8a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-08-13 07:04:40.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-08-13 07:04:40.000000000","tz":600},"subject":"FakeStatsdClient: Stop issuing DNS calls for host.","message":"FakeStatsdClient: Stop issuing DNS calls for host.\n\nRunning unittests on my home server, even though is a beast, would be\npainfully slow. Running them on my laptop ran much faster. I tried\neverything and couldn\u0027t figure it out.\nChris, co-author, dug in and after we ran out of options he fired up his\npacket capture and realised most tests were issuing DNS queries for\n`host.`. On my home server, it runs ipv6 dual stack, so these calls\nwould need to timeout before a test can continue.\n\nOn finding this `host.` Ben, another co-author, dug into the code and\nfound the only reference we have to `host` is in FakeStatsdClient.\n\nSure enough as I dug a little further it turns out our FakeStatsdClient\nused to override a StatsdClient function `_determine_sock_family(self,\nhost, port)` to stop actually creating a real socket.\nHowever, at some point the StatsdClient was refactored and that method\nwas renamed and changed. Which leads to a DNS lookup every time we\ncreate a debug_logger as it brings up the socket.\nAs you can imagine, this happens alot!\n\nThis patch overrides the new function that handles to socket creation\n`_set_sock_family_and_target(self, host, port)`. Which eliminates the\nDNS call all together.\n\nCo-Authored-By: Ben Formosa \u003cbformosa@nvidia.com\u003e\nCo-Authored-By: Chris Smart \u003ccsmart@nvidia.com\u003e\nChange-Id: Ie393075f79447627714692e3f01bb53e967a71e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b73612d1dcd0bace2361338e6ae82eb591282b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b73612d1dcd0bace2361338e6ae82eb591282b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
