)]}'
{"id":"openstack%2Fswift~926179","triplet_id":"openstack%2Fswift~master~I4aab4fedab337e197ab29deee1892935178a0e07","project":"openstack/swift","branch":"master","topic":"state_ts_deltas","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-08-13 10:01:38.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4aab4fedab337e197ab29deee1892935178a0e07","subject":"Sharding: Add minimum_active_age to shard compaction","status":"NEW","created":"2024-08-13 08:45:38.000000000","updated":"2024-10-18 00:48:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c089d8bdd2363cd8451ead07d51917cdc3aaa477","_number":926179,"virtual_id_number":926179,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-18 00:48:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-13 10:01:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f45880b213bff89fb6ee2b7b9d47eba8b4a82e8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-13 08:45:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03fb4921e13d2c7fd666d6ca7c7596ee98285618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 10:01:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4cf59dc4e47417ea00c0f3d0a2520b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e00a022705d49b387fe552047222609 : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/40c4a98ed248405dbe2e15bde0ac9204 : SUCCESS in 52m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4065411260d4416ba9dafdecd15f42c7 : SUCCESS in 38m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/552970f0551343a780685cc8cd8929f2 : SUCCESS in 1h 09m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72770ef8ca034429ac4619036cd74bdc : SUCCESS in 22m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48bb691c3d87458083d3d0b4711207ac : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91ac1f15030444dbb924b9b4dd5728e8 : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ebbe986154a2448f8c3de1ea9b09b97a : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c70ff86730e34ff68caa7e98c80aebd9 : SUCCESS in 18m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/568a7527db7c40518ec53e655948e6a5 : SUCCESS in 18m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/db6b88c14c934bae92831d3dbfb28431 : SUCCESS in 20m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/71afd4a4087449c2a8cab19f79f543ae : SUCCESS in 21m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0f06b8be52e84f4fa42759c536292df1 : SUCCESS in 27m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e2f9b2024de44081ae6e594948d503d7 : SUCCESS in 20m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4ad3d734e5354c73a3d046dcfc819585 : SUCCESS in 22m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8ddc614946384cfc929e85e1af97e189 : SUCCESS in 21m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8b7a9e7e2b0f49fd9b9dafaff6abfffa : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d6fa458717834c8a9029fc6c81a26dca : SUCCESS in 18m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dc4de9c6803f4c558bd11ebfdd6d0339 : SUCCESS in 17m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93a15fd0eeb742848de6576b901382ca : SUCCESS in 9m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3df899d77ab244bd945dc175c030ac65 : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/86690d838e7d4a45bf2576c48b7e5334 : SUCCESS in 6m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cde39c87982b4efab1f37b9f4cb78a55 : FAILURE in 46m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ecf147b03ea1439b9403933d24fa0ffc : SUCCESS in 45m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c547441b7d54faba9fde81e38d87506 : SUCCESS in 45m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38a2317079044349b8ccd6850684f543 : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae634725dbfe4a2cb64c3973c1328664 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f4c96bd1fd6c4f2cafc951d287b26789 : SUCCESS in 21m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/359b5f85ecac4442918868287eafe85a : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":1},{"id":"64bbff14e6a509364f3c45c72fd0c070aa158d73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 10:19:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae85dd4c6bc4419b94174c97e735c687\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56c17694c7b340a0a60902790f806e5d : SUCCESS in 1h 19m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f4ec4e5b7944aa584ae781090ca4950 : TIMED_OUT in 1h 30m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ab634225c86424c862ac58eb1f99674 : TIMED_OUT in 1h 31m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a24cae2663a84d5fb97b86d123b241a3 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"64a4b065a81ac5adf44c35332bed712f80a17599","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-15 04:18:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7ba8580c86db7f9bee44d2122d8f085e0995a02f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 05:05:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e480056de63444558263c73a25482512\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51477dd7920941238d8f7543098d0faf : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/af35f4e93ec34802afeb14e924ba63db : SUCCESS in 44m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6d380b9ff60471b890581dcfabd2425 : SUCCESS in 39m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cdcfd065a0f46119c461eda6919c1a6 : SUCCESS in 44m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad44ea38172d4546910809b08499f407 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2682a6afa8a94ede9b65b4484006de30 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46078b6a308443fa914582dcb9c80f73 : SUCCESS in 17m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a9a9cbb8c10148caa34e943bed229db4 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ddc5d9418e64fb4b83d81372ed1a1d4 : SUCCESS in 18m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d84ce7c7dc1e45e4ba0403570ae58307 : SUCCESS in 16m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9e2c16a4bd884827bee7ade368913613 : SUCCESS in 19m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c5dff3c81d654e66a6ebebee1c26628b : SUCCESS in 17m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4c332decf8d34724ac554d50fcbb1ed2 : SUCCESS in 29m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/15bcb4ef2f0b4698bde946307b9bf106 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a550cd4a18e7434da1f9bcb7eeae6f82 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9264a940887b46e083cc0ed51e7963bb : SUCCESS in 17m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/522b87fd97384223a558693a96bcb051 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f12482f140b0486d81e0c1155adbde1d : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f6d1750f0d144d3e9e2216b375fa3581 : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64174ad9c8d744719a2961ef2738273d : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58eba8bf1b6a4139ae38e3366f52aa57 : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b20a0929db2b4fc898626736e2372424 : SUCCESS in 5m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/28190099eda44de49690aeea4bbb0192 : FAILURE in 45m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/666e641484b341a89679cd54cf801561 : SUCCESS in 36m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/343575b9b47445b1b0b79f1709377940 : SUCCESS in 33m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e820141bd4ae491b9506332a6cd2824f : SUCCESS in 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f32101dea54595bca5b902f3aa06f4 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36da9c7f50d84a9cb0b8fa08cda02406 : SUCCESS in 15m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3cc5babd127406d8eb8e43b7a3f3821 : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5df69720a58229815bd2c302002af6ddf5bd7ba6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 05:51:44.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fbd344933294d2d96a25f1ab954a468\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec8af8ec560c44e8ab85c6738bc8952d : TIMED_OUT in 1h 30m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b54dcc7ff2640c8b630481af589c2be : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f9321b151704156a3e2ebe2f9202eeb : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a553541350ec4ea0a877008d07dc204d : TIMED_OUT in 1h 30m 51s","accounts_in_message":[],"_revision_number":2},{"id":"cb435996c632a96bb72a2d4653037ad082a50056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 22:35:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_7233\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"_revision_number":3},{"id":"6ee2369703540747186f382ff11c00d0bd88f372","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 00:20:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8af2d01c0396470a91dc7bebe4da2056\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10e04991786541b0ac8dcd13c4c35e8e : SUCCESS in 1h 43m 09s","accounts_in_message":[],"_revision_number":3},{"id":"c089d8bdd2363cd8451ead07d51917cdc3aaa477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 00:48:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c999ade5b8c04a49bb4ce4c8e2b472ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2103ec5c908544d6abec6fa37e23adf6 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/9a9bec16929d4b32b38809f09e571d36 : SUCCESS in 39m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37e34e6187cd427db79e96793d521633 : SUCCESS in 1h 02m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d41938802354aa58621a997008c704e : SUCCESS in 38m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7341e7ca02494fa9bcb2f3a9f72efd76 : SUCCESS in 58m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73039b2fb5b1420d9578008aea3b412c : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e27b7d17965d43d3a283aec48a07fc15 : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63f5f500b6dc499086e91cd3fae0c9de : SUCCESS in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0642e95cb8e412f885bd5830cfd2c90 : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4d13f6684ad049f4ade7f22c22a27e6a : SUCCESS in 20m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/beef1631c15d40fb936c07660169bc07 : SUCCESS in 9m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8776ef2fdd144a24abc7bd480e7ac7f8 : SUCCESS in 23m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e8304be1793e4bc09c36d27ac1b1670b : SUCCESS in 11m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3287566f518146ac85d2343040f950e9 : SUCCESS in 36m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/109f6ed2889c4d2aa8c4544ec0511078 : SUCCESS in 10m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/46e20b1c1688467785811551a573d354 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfac71ca63334210b23574d728ce62d2 : SUCCESS in 10m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d53ceb7203f345cebacf1c6b401cd7ed : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e5e21a6bb14c41e58f28b1b777052439 : SUCCESS in 8m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3d76678115da413fa676eb5651a55a79 : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fbf9e65c0ef470cadfeb61c23cdc888 : SUCCESS in 4m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d3b02b5f5c414f88a5989f90bd1cff71 : SUCCESS in 16m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/67efca9ca1fd42f6ac8e8fe951ae2c8f : SUCCESS in 4m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/23d55a4080944fecadb56325e15b404b : TIMED_OUT in 2h 00m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/560e16aaead140fe92ccca176b5188eb : SUCCESS in 31m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bde65eb1151148e6963e3e35ec0c4f91 : SUCCESS in 52m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7824955727a421ca5b47f3135f12c9c : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c34e22bcb141cc8fc4a78b4fd88c08 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8b6197e742b469798123a0df02c91b8 : SUCCESS in 12m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e75e556cf5144608919d8d2a97dbecb : SUCCESS in 33m 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8a86a6380d4cf97dfa3d6efc9b782fefe7d0672c","revisions":{"ca54e524adccc9df9a11b0a0f8429dd3f0b38273":{"kind":"REWORK","_number":1,"created":"2024-08-13 08:45:38.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/79/926179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/926179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/926179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/926179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/926179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/926179/1"}}},"commit":{"parents":[{"commit":"c527f148c096c695cf2c42810337d2250c001570","subject":"ShardRange: track last state change timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c527f148c096c695cf2c42810337d2250c001570"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-08-13 08:29:16.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-08-13 08:45:25.000000000","tz":600},"subject":"Sharding: Add minimum_active_age to shard compaction","message":"Sharding: Add minimum_active_age to shard compaction\n\nNow that we can track when a ShardRange state change happens, we can\navoid the early active problem. Put simply, an ACTIVE shard is only a\ncandidate for compaction if they are a certain age.\nThis is to give a chance for other replicas to catch up.\n\nThere is now a new option, `minimum_active_age`, which indicates, in\nseconds, how old an ACTIVE ShardRange needs to be before its a candidate\nfor compaction. The default value is 1 week, making it 0 would be the\ncurrent behaviour.\n\nChange-Id: I4aab4fedab337e197ab29deee1892935178a0e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca54e524adccc9df9a11b0a0f8429dd3f0b38273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca54e524adccc9df9a11b0a0f8429dd3f0b38273"}]},"branch":"refs/heads/master"},"4a2caa4dcdcb626d73f9e47eef80fdecc29c1ed5":{"kind":"REWORK","_number":2,"created":"2024-08-15 04:18:15.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/79/926179/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/926179/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/926179/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/926179/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/926179/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/926179/2"}}},"commit":{"parents":[{"commit":"fe14d7e0eb211471cc3af09ae8617a8bdb336a3d","subject":"ShardRange: track last state change timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe14d7e0eb211471cc3af09ae8617a8bdb336a3d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-08-13 08:29:16.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-08-15 04:17:43.000000000","tz":600},"subject":"Sharding: Add minimum_active_age to shard compaction","message":"Sharding: Add minimum_active_age to shard compaction\n\nNow that we can track when a ShardRange state change happens, we can\navoid the early active problem. Put simply, an ACTIVE shard is only a\ncandidate for compaction if they are a certain age.\nThis is to give a chance for other replicas to catch up.\n\nThere is now a new option, `minimum_active_age`, which indicates, in\nseconds, how old an ACTIVE ShardRange needs to be before its a candidate\nfor compaction. The default value is 1 week, making it 0 would be the\ncurrent behaviour.\n\nChange-Id: I4aab4fedab337e197ab29deee1892935178a0e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a2caa4dcdcb626d73f9e47eef80fdecc29c1ed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a2caa4dcdcb626d73f9e47eef80fdecc29c1ed5"}]},"branch":"refs/heads/master"},"8a86a6380d4cf97dfa3d6efc9b782fefe7d0672c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-17 22:35:59.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/79/926179/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/926179/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/926179/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/926179/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/926179/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/926179/3"}}},"commit":{"parents":[{"commit":"f2e42cc00c9c29d0231d40c327955d283b3eae0f","subject":"ShardRange: track last state change timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2e42cc00c9c29d0231d40c327955d283b3eae0f"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-08-13 08:29:16.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-10-17 22:35:59.000000000","tz":0},"subject":"Sharding: Add minimum_active_age to shard compaction","message":"Sharding: Add minimum_active_age to shard compaction\n\nNow that we can track when a ShardRange state change happens, we can\navoid the early active problem. Put simply, an ACTIVE shard is only a\ncandidate for compaction if they are a certain age.\nThis is to give a chance for other replicas to catch up.\n\nThere is now a new option, `minimum_active_age`, which indicates, in\nseconds, how old an ACTIVE ShardRange needs to be before its a candidate\nfor compaction. The default value is 1 week, making it 0 would be the\ncurrent behaviour.\n\nChange-Id: I4aab4fedab337e197ab29deee1892935178a0e07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a86a6380d4cf97dfa3d6efc9b782fefe7d0672c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a86a6380d4cf97dfa3d6efc9b782fefe7d0672c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
