)]}'
{"id":"openstack%2Fswift~926220","triplet_id":"openstack%2Fswift~master~I6dda5f2e44a6fca8a0705d682a3323ca39e75268","project":"openstack/swift","branch":"master","topic":"p-post-timeout-404","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-08-14 15:17:18.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6dda5f2e44a6fca8a0705d682a3323ca39e75268","subject":"WIP object-server POST: return x-backend-timestamp in 404","status":"NEW","created":"2024-08-13 16:00:46.000000000","updated":"2024-08-24 03:34:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cc2311aa89eaa64263bf858bc98ee0d466cd83b4","_number":926220,"virtual_id_number":926220,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-23 16:30:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-13 17:07:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"437cb74b4c6db71ebe011b732410d338d5848239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-13 16:00:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8934b36f90aaa4cec1dcd5e623234138931b776b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:07:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/000b16344d1f4ddca22468c288f07682\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b104ce17f13047319e37d1a7312f1bf2 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/e853c4399b484e05b6966b2d4aecebc0 : SUCCESS in 53m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33ec341f53e640b5a0414e8975de8e71 : SUCCESS in 40m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f513c776380844539b031644e3e89009 : SUCCESS in 59m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b556b4b4769443cea0c21ad5ac32f08c : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6fb69b3aea0429dac86a0a28ffc8c16 : SUCCESS in 22m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c657aae987c4fae914e6f491c8c7317 : SUCCESS in 22m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47fb81e4ef9b45b288b20a4d41e49564 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7fe07979eaf24e90b6b80dab696b957d : SUCCESS in 21m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/410938f2985045e5a3f5068a8c4db65d : SUCCESS in 18m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b57c8f66bdd74388a6989cf8c2898ad8 : SUCCESS in 19m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/325cd6141cb043c8aff4073cd2674179 : SUCCESS in 20m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c5d402fd83eb44b2aaa8f28ee005fbf5 : SUCCESS in 27m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7950cf25c35249afb0184d0d16ab63ce : SUCCESS in 21m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/35037d60e8434c2da58ee85ca125f657 : SUCCESS in 19m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/175b7478b3e64113b7288b25c411dd08 : SUCCESS in 19m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dd5c1208d9704be180dbc34d825a7154 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fad232f6a8b0446f89d64bb3173d5314 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f263e4a802854475b6ff680b996d21f8 : SUCCESS in 19m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85930d6d15e34edcb094c32dbdd41111 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/086d4e8d92774a6c843b44bf6e1b6ac3 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b454dfa139474686a614dc76c77384b6 : SUCCESS in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/94d655d1e55b4b2daf1b71f38acb0f95 : SUCCESS in 48m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bad296d5bb64866a15e7fdc1aa38f8b : SUCCESS in 48m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b509a2bdf88422995aada723d79a1a5 : SUCCESS in 39m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3aefb40660943c298ab2b79d4ae28c7 : SUCCESS in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56843dba29bb45aab676e4cf1d63231d : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0917ac15a33d4d03a7de8c4b9986c65e : SUCCESS in 17m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4784770621c4a9eb20ccb6494f1222d : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"490b37b58338311cf7d7b9c0d85f34c8a6a964b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:59:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd11b928a55a4d6c819494daece04761\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37924e5b29c04f4ea42227a6c3f559d9 : TIMED_OUT in 1h 31m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9f573ddd37d48a787f39707e8960a3e : SUCCESS in 1h 09m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7427b04d70d24cdebfe05c82723f788d : TIMED_OUT in 1h 31m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c4bcfbbf45b4329b5329172ebdbbafa : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"873cec58733d1a42c86e1fd04f13974bb4c96d88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-14 14:11:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1cc5c6c9d11ab071e798d8924e88042c3dda29ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 15:17:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21a4360e5eec4474832cf37fee4e9b9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414830aa27d945dd8e4548bde2832920 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/f61e89b68ea94e76b000417bf5d00d83 : SUCCESS in 59m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0158452309e474893b5eb887880665a : SUCCESS in 48m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e29211498d64b5f82c3e3ad6334c3c7 : SUCCESS in 53m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1d3db886f9442b5b677ee9906b2d126 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7fa4e8a93d041329ca7a5f9fd984e62 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59cb2f649b3943a1b9371349357914f8 : SUCCESS in 19m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0fc166e8d3ff4536b43b6cd35fee18f4 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02b92a684a2e4599aabf794d986f00ff : FAILURE in 20m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eec1a3151ff141379cb5480117a4d535 : FAILURE in 17m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cca183c9ecaa4991820ae9e51ce70184 : FAILURE in 18m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e1651b79202f47e89505de4b0cc8874f : FAILURE in 19m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bfce78197981418083a49f0bb36280ee : FAILURE in 28m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/871aa25880c64b96af2f92682d87c2a7 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3889c065f95e42578aebbba6a0f1362a : SUCCESS in 20m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65a98166e23a4d62937c12f491638f81 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/959b5b9ca62245b5ab76cc1e56b43f48 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/83d5e9c2f79f491095211f764a08be9c : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d76e1492798441f9a2057d29cfb0b7ee : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7379d42700c4049ad474d977d4c9d08 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/782bbc3eeb774628b1c6d59f57e3f5da : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/65efc83ac5c247bc80cd986377780c1b : SUCCESS in 6m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/075f856d326e4b2a92086c0980268dbe : SUCCESS in 54m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3de7a51a915044d3bbeb9f4c98e31912 : SUCCESS in 59m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed49c9612060444aa9b5e53131cb3712 : SUCCESS in 53m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a746cc3f61f44d98a6523b23c540d8cc : FAILURE in 25m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e58de5fd5d43528cafe80a1e324188 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49f4778342564a108608343cf9c65e4b : SUCCESS in 24m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/403fb4a1ee7e41e59d945e0efac0923b : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":2},{"id":"08d59d9d63510b6bf97abeca8d6d73659add9f4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 15:47:38.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98ca637abe0741aaa45aa744475ceae7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2187ab171d2e483694def00dcb4cd16a : FAILURE in 1h 04m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d72cbf6b0e342c989750706179d7053 : FAILURE in 1h 18m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff652ed2b444446c828d0afb8b45e5f2 : SUCCESS in 1h 24m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f98792e58e64f6b8e5cb1feadfbb809 : TIMED_OUT in 1h 31m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d2ddbe7c2bfa3234c17c2f45f7f6f2f8122f6ed4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-14 16:42:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"92a33e82dce5184b18da8063b8b6e40b319ff26c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 17:42:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/951557ec759d43269c729c30cc99bdbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7fcbe78be24d8196753c59867fe956 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/f36ac9351e164251916c68642fbd7ffc : SUCCESS in 55m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7aa3d7ef71604de39efa4c475c5b6b32 : SUCCESS in 46m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8980c0b10f7a41a59dac83d816573ad2 : SUCCESS in 49m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/404bdfae7fbb4073a95d13d68323ef67 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e7606e152f44e03b79df2522b73907a : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/708bc70c2f4442c28b995b44af02f0f3 : SUCCESS in 21m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0280302ceb314222950d2f4ec0e06573 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e0dcce824cc4d9abeb3f3594cce4192 : SUCCESS in 18m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03421997cc3d4425ab3f1027fa271307 : SUCCESS in 18m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a5962d5beac24d3188d7fa3ad45471b8 : SUCCESS in 19m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1288365f145e4882a45cd20496e52535 : SUCCESS in 20m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f4429bebfa6647de91f25d3372c8f319 : SUCCESS in 28m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7cfeb40e7513461c95b564cc58d4d1f5 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1d8f006f78324232ba2a71b386b341d2 : SUCCESS in 15m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c1bb9a7a16864fb2b0af88c2f73a1454 : SUCCESS in 24m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c0774937b3264ecc9b72a08a6355cfec : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4d758edd872e466a9ffde2af12d17f7a : SUCCESS in 15m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fe042e3ddc0e423cbd3f8f4c821b7b34 : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eb653e4910f04a609d23529237fa0daa : SUCCESS in 5m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89465fea1bb840d3a78f27a456a93539 : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/53552e554675479598930f64eee987d9 : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/da47275d7e00484abd23431384e10e6d : SUCCESS in 44m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d08490ca6c54a92bd438e3082424e4d : SUCCESS in 38m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/87b7ab51267e4013a8d00664ee65ff84 : SUCCESS in 52m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/177f69585a9840d1a96afd9b91362d85 : SUCCESS in 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0373e3531f62416db099be295532fddf : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92082bf03c13494b99804de05869ef77 : SUCCESS in 15m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04cd555e142f44bd85d0e290fc9d6d57 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":3},{"id":"c1f4391618b147297e6e53979102fa8bd0535a08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 19:22:04.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607444b32eb942828b4996b7a8f9bb37\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5addf1ea769b4df5b4148c50a7f6c74d : SUCCESS in 1h 14m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74d1eeb1058649eaaf4d88e62bbb1512 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d52c4745efe04aa6a91b4bf0f286daa4 : TIMED_OUT in 1h 30m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab9cd035b6bd486fa64b60b024a97636 : TIMED_OUT in 1h 31m 11s","accounts_in_message":[],"_revision_number":3},{"id":"097da2f2c4503c3ed79bdc12b2a8f02f47ad29b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-21 15:28:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"21768586be27597a2674305621d3a563346240ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 16:51:13.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/597e9aed362d4f1e8183f5f67ea2dc70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e494ef2072461b852b86b55e192149 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/71f382a43a804e49bee2b90c70d21016 : SUCCESS in 1h 04m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3943ec7faabc4eae8e5ee9d69e7e399a : SUCCESS in 1h 03m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bc57d5a49324bb0b5912225301bd0c2 : SUCCESS in 1h 13m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9304fa2293f04451be5374ff4932145d : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ace1fbfa6e140a98cde375b9d030578 : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc6dd7950fa24c86a5c377ece7d6de98 : SUCCESS in 19m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/24c65f98097742cca96d81cc7f863f29 : RETRY_LIMIT in 2m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5b02e06cac0345e8b4bd1922dd7e14e3 : SUCCESS in 24m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6e4f4037972841ef80dc605845bb0166 : SUCCESS in 19m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/24116a5b8ce348a189af939a25d917b6 : SUCCESS in 20m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/16b106cbf15c46d7976bf91d164d50f9 : SUCCESS in 26m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/24c04123495643aaae1c6e89273fa742 : SUCCESS in 34m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b38af0ef427344ebba89c7b3b5a111eb : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4e49317c58347c091dd957277d508a1 : SUCCESS in 20m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/12f2f85c87f74dc599ac0c99d1972f6b : SUCCESS in 19m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/97261091dc024d5baed2d1b83b65d591 : SUCCESS in 20m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/445fd112b0a648259b25859c48e0f3ca : SUCCESS in 18m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8093db7a8355488994b400b3658352e4 : SUCCESS in 18m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57b2317b27a3412fb3f14175f02922bf : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7fad3d2341e04d68bb90f48870229b3a : SUCCESS in 16m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/534da6448f164383b4fa04ac4438d198 : SUCCESS in 8m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2092ddb4ebf64300b4aab555fc55c7f9 : SUCCESS in 51m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/63d9baaeebf14a81a3169dd9b3ab6b14 : SUCCESS in 49m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f7d709ccd6224cc88f93324f6567ae80 : SUCCESS in 1h 00m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/430d44e1ee644e3782ae554a87071f32 : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f00551b455b4fe3be0c78b8f8696774 : FAILURE in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39dfad16f77c4376bdf78981dd281275 : SUCCESS in 25m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae15b48d2c664c5989ec1588c53eab14 : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":4},{"id":"3b1e730e7e500a226e0c90e05b13c5b3939f3ed6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 06:48:25.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/333107e0a9084a47b8181fe65478c0ea\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38837dc7d49241398d3a4e8f2122806e : SUCCESS in 52m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/397942c320574318a1e0d3635f15f4ec : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b04a0ac2a3e40d8a87f94f8edba796b : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a790f54fa84421dbdb7ce00e5ecd5f2 : TIMED_OUT in 1h 31m 12s","accounts_in_message":[],"_revision_number":4},{"id":"3a17a67ed4eef92b118a076b68a78a53f355559d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-23 15:15:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1e50b7a7f2acda2352be046920c7e6bdd697918c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 16:30:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4922fff0c572408696405cdd13107d4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13f8507c701c47f88865ebf0037fbc84 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d3b298c39e5f4a1b80f643f426de9be6 : SUCCESS in 59m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5c2a42ce5c4435f94fb3cb03b248273 : SUCCESS in 59m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7423e3ce8be14c4180a13e0fa5bb1446 : SUCCESS in 59m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0e2f70e28314a56ad0797b46d8a12fc : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ef40a1024c74505a689746473bf740e : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac42deb3b58143a08193c524240ab9c9 : SUCCESS in 23m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce5228902b484c3b9ad537b187572481 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27560fd59b6f4a64bc4de39f18b55027 : SUCCESS in 18m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07ca9ba713e54e7982a30a5aab89e387 : SUCCESS in 18m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9febd66e7b104cbeaa0eba2f8637cd49 : SUCCESS in 20m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1709bd7c3b374b91938c9fa51a42a1dc : SUCCESS in 28m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/83df798a8cde48a38fb018db183ce53d : SUCCESS in 24m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a1ad46a48c5f4f978d211cc51379b7eb : SUCCESS in 20m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1620979a795f485ba84df2f9960efd7b : SUCCESS in 21m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7d53d77f20d34022a6288beaa292149f : SUCCESS in 27m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ddb2f3719d7a411a989ea3d5c442a77b : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d0fce17de4c247d4b7dc4034ec7b0d53 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d8edc3dede5b4af49f30028557e451e8 : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6e58e2cefdb14277ad626b34df269ad3 : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9fd59588907e43e2a6e0750c61dac1f5 : SUCCESS in 13m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a041622e698243c0b173e012cc8c48fa : SUCCESS in 6m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4d4b0abc36e426fa1e9d7d363d9b132 : SUCCESS in 57m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5fcdb0e2bf5a42708d371f0269363656 : SUCCESS in 42m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7dd34b7b6de842baa0f3997a0f9834bc : SUCCESS in 57m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c08967449098469e8793004803c57075 : SUCCESS in 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93afda0650649c4a863ee5d48e29dff : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4c4498f64f054171b88323e6be4d9333 : SUCCESS in 17m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/026e1ebf384a452fb6fb91c9e61d36d3 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":5},{"id":"cc2311aa89eaa64263bf858bc98ee0d466cd83b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 03:34:02.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea5b98f9b9ab4d38b1a7838a6a931703\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/908143d6cabd44a395e681bdb3a4784d : SUCCESS in 1h 14m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d286f0fd8e964178a858dac68fd746bc : SUCCESS in 56m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/896c152bdcfc44f298b9a5e541f3f21c : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fd6dc56ff2243ef88c5921c9115b4ca : TIMED_OUT in 1h 30m 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c40934e20cc7253e698cad8e8b415896d2712d30","revisions":{"e035b2b413c4f8dc2fcafb374e59d2e040bac997":{"kind":"REWORK","_number":1,"created":"2024-08-13 16:00:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/926220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/926220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/926220/1"}}},"commit":{"parents":[{"commit":"9e9585e714f5642ad680eff0215656a092823bf5","subject":"Ignore 404s from handoffs when choosing response code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e9585e714f5642ad680eff0215656a092823bf5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-13 15:18:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-13 16:00:08.000000000","tz":60},"subject":"object-server POST: return x-backend-timestamp in 404","message":"object-server POST: return x-backend-timestamp in 404\n\nRelated-Change: I96f28ab0b2b5f9374c399e8905ee240e7b093f8b\nChange-Id: I6dda5f2e44a6fca8a0705d682a3323ca39e75268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e035b2b413c4f8dc2fcafb374e59d2e040bac997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e035b2b413c4f8dc2fcafb374e59d2e040bac997"}]},"branch":"refs/heads/master"},"99ac306db7f049dfef5cd2a9550ddad4a9ad3233":{"kind":"REWORK","_number":2,"created":"2024-08-14 14:11:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/926220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/926220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/926220/2"}}},"commit":{"parents":[{"commit":"a1735cbb8d01f4338021b2c4933a75a76cc07bac","subject":"Ignore 404s from handoffs when choosing response code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1735cbb8d01f4338021b2c4933a75a76cc07bac"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-13 15:18:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-14 12:57:05.000000000","tz":60},"subject":"object-server POST: return x-backend-timestamp in 404","message":"object-server POST: return x-backend-timestamp in 404\n\nRelated-Change: I96f28ab0b2b5f9374c399e8905ee240e7b093f8b\nChange-Id: I6dda5f2e44a6fca8a0705d682a3323ca39e75268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99ac306db7f049dfef5cd2a9550ddad4a9ad3233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99ac306db7f049dfef5cd2a9550ddad4a9ad3233"}]},"branch":"refs/heads/master"},"362b930e4a56207f8da860755d25fac20f3b7886":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-14 16:42:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/926220/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/926220/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/926220/3"}}},"commit":{"parents":[{"commit":"ff76290e67bb2c38d79a9a9f014769e08fb09972","subject":"Ignore 404s from handoffs when choosing response code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff76290e67bb2c38d79a9a9f014769e08fb09972"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-13 15:18:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-14 16:42:09.000000000","tz":60},"subject":"object-server POST: return x-backend-timestamp in 404","message":"object-server POST: return x-backend-timestamp in 404\n\nRelated-Change: I96f28ab0b2b5f9374c399e8905ee240e7b093f8b\nChange-Id: I6dda5f2e44a6fca8a0705d682a3323ca39e75268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/362b930e4a56207f8da860755d25fac20f3b7886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/362b930e4a56207f8da860755d25fac20f3b7886"}]},"branch":"refs/heads/master"},"c884533b5900291d4be8ffe063dbb5aff64408dc":{"kind":"REWORK","_number":4,"created":"2024-08-21 15:28:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/926220/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/926220/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/926220/4"}}},"commit":{"parents":[{"commit":"cfd6a841ce4b82146feb0d37291c90f6b52e3d50","subject":"Ignore 404s from handoffs when choosing response code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfd6a841ce4b82146feb0d37291c90f6b52e3d50"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-13 15:18:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-21 15:10:52.000000000","tz":60},"subject":"WIP object-server POST: return x-backend-timestamp in 404","message":"WIP object-server POST: return x-backend-timestamp in 404\n\nTODO: update/add test_obj POST tests with backend header\n\nRelated-Change: I96f28ab0b2b5f9374c399e8905ee240e7b093f8b\nChange-Id: I6dda5f2e44a6fca8a0705d682a3323ca39e75268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c884533b5900291d4be8ffe063dbb5aff64408dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c884533b5900291d4be8ffe063dbb5aff64408dc"}]},"branch":"refs/heads/master"},"c40934e20cc7253e698cad8e8b415896d2712d30":{"kind":"REWORK","_number":5,"created":"2024-08-23 15:15:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/926220/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/926220/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/926220/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/926220/5"}}},"commit":{"parents":[{"commit":"d0f6a0ce5624e271bfc3e2cca7d328a18df546f4","subject":"Ignore 404s from handoffs when choosing response code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0f6a0ce5624e271bfc3e2cca7d328a18df546f4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-13 15:18:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-23 15:14:34.000000000","tz":60},"subject":"WIP object-server POST: return x-backend-timestamp in 404","message":"WIP object-server POST: return x-backend-timestamp in 404\n\nTODO: update/add test_obj POST tests with backend header\n\nRelated-Change: I96f28ab0b2b5f9374c399e8905ee240e7b093f8b\nChange-Id: I6dda5f2e44a6fca8a0705d682a3323ca39e75268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c40934e20cc7253e698cad8e8b415896d2712d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c40934e20cc7253e698cad8e8b415896d2712d30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
