)]}'
{"id":"openstack%2Fswift~926453","triplet_id":"openstack%2Fswift~master~I71438a2f67b26d41392a131ed73003f7fc6b2d73","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-04 20:13:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I71438a2f67b26d41392a131ed73003f7fc6b2d73","subject":"sq: add some tests","status":"ABANDONED","created":"2024-08-16 00:11:36.000000000","updated":"2024-10-04 20:13:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b97c2677861fe51660007bcc381ecf81b86ff03e","_number":926453,"virtual_id_number":926453,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-28 00:47:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-16 01:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d0eaa13f7183b3c5e32dc1ce61919f3592dc08b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-16 00:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34685ae04789f03ccdc043a592e2cd33901a2066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 01:40:05.000000000","message":"Patch Set 1: Verified-1\n\n(25 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9d5a9961ee0473eb4c2e6b7d07bd3f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae2ed7b3e7b475b8ade6f8d3a38158e : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/5e1d486e7e7147af9999012ace511899 : SUCCESS in 1h 00m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a63481d8c1e48b687e1c74b33d5e346 : SUCCESS in 47m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebe3cd0776d74ce88e07e70ece089470 : SUCCESS in 42m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18d5cccd0fcb456699ab5f76cb075f49 : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/301b6cafbaee47c7ab779fa1a838e536 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/caa71f8a39224acb8eee4eb9d385ba7b : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a9af8fea106f4de4a00a6d97e5e2ba5a : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/942b81b679bc4f6fb485a2861202c683 : FAILURE in 18m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/afc120c66de442389110202705ca3352 : FAILURE in 18m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a3583c253c2747b09fc0681b195f1bd1 : FAILURE in 19m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/278485a7d8084d16a95b438c9af38a50 : FAILURE in 16m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/15d1886e61fa481a9a9731d08a8589e7 : FAILURE in 29m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f620778a03cf47c8b9dc233dfd36857e : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/86ad3d6f3b904848821660f89777cb80 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97a313372c6e450e84c90871f800807e : SUCCESS in 19m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3c48543af83c430983f4a269894eb199 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5acf833ced6f46c7ba037b7ca4700db4 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b12af52fad7a4074832957d44469db36 : SUCCESS in 19m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/931d25261476418b9090d67356ee2dd0 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8c1ec7c4d5946f69dbc4e37c0261671 : SUCCESS in 12m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d4421ddb29fc4e098c963b84c312d384 : FAILURE in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/38b6790f1ef24f82abae7e33e2274c5d : FAILURE in 47m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e1bd3ad96554b3cb709f097624f00c3 : SUCCESS in 57m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee60544f77814a318aba2079644e5438 : SUCCESS in 39m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/749795b726bb4507980d6bfeccf80eb9 : FAILURE in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5666f374a0b4d68a748a7744173fba7 : FAILURE in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50dff728d70b4393a2504dd8cad2f610 : SUCCESS in 21m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5648eb3500d747bfb15644929a09441a : SUCCESS in 58m 32s\n\nWarning:\n  Comments left for invalid file test/unit/common/middleware/s3api/test_s3request.py","accounts_in_message":[],"_revision_number":1},{"id":"7d57d4838708a6053d6411ded9b1cdfdc3b0097d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 02:57:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ffcf428829e4ec5b5c8bbfe147c39b5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/762219b36b994d528a18b3991267319f : FAILURE in 1h 11m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77f581361bd944aeb4979b1a7c66d4db : FAILURE in 1h 02m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6e5a3b0f0e14f10a828777be19a8035 : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93b4ee6cf58d4e638f5ad37bdcc9c767 : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e319a6a14486d3db59b68631f7c181dde8c803dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 00:06:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1a4cceaaffb294b3f314d6559b547c605c958909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 01:22:16.000000000","message":"Patch Set 2: Verified-1\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b410ada8c86241f889c7a2a6ce8335fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98e557ed5de454fa2b82641973b3da7 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/443cc0de080f493aa3998139708d06e9 : SUCCESS in 59m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/41f50172087b4b63bf707366bb23d855 : SUCCESS in 42m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/325d8f34ed2c4618aebf6f29ab2c4ca1 : SUCCESS in 1h 02m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/105f30b89f24456f886e46088691f625 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fd2dd59faac40c78cb352580f3130f7 : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04dcd0f22498434a93b94d0f813e167e : SUCCESS in 23m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18d71a0828654af59deb72f27fdc34fa : RETRY_LIMIT in 2m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/99bcdbd5cc62470c92dcd1823810c5cd : SUCCESS in 20m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1119df3624c440d0a31ebab8ae138a9b : SUCCESS in 17m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ae1bf8b0d5af46cbabade0171683b8c8 : SUCCESS in 18m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ddb11e9e74284fffab209b29da1897d9 : SUCCESS in 16m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/86d7cf7c9c89440c8922273e55454733 : SUCCESS in 24m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7797d56bf3d942febc401bce64544af8 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bad92abc0a4d42a4b9a445f36a0fc136 : SUCCESS in 19m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b81e9312273643318f019824c33eca18 : SUCCESS in 18m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/84d618385e3545ec8f49580a9a811cbb : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c6e74c8c1764edda53a2461252cd38b : SUCCESS in 15m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8152960c94754c8f9f33a9e5866ed1f7 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83e53adb362d44b0872d5e7c95ecb765 : SUCCESS in 7m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b052e14852374534a86a835922fa2701 : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3d28f2cade134bf0b8c84ea2b575c266 : FAILURE in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8df3ecbb0ee543daa0e6edbfb145aedd : SUCCESS in 52m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/054f85126afb431cae52c26cf2c9fe9f : SUCCESS in 44m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36fab9163bef4a54bf072d6b62db999a : SUCCESS in 47m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a201bdd42f4144a68854a84aa117fd8e : SUCCESS in 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53fc34d5b7404b1585e45ff48926c869 : FAILURE in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6aae6a9802d41bc8fb762f91fbd9e86 : SUCCESS in 22m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f050f11f96734b88bc66d9cec858960e : SUCCESS in 42m 08s","accounts_in_message":[],"_revision_number":2},{"id":"7397bc95c0c03112f90c0535d406a07d691d438c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 03:34:37.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5edb951ee95b4342a593281a4040ee82\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fa3dbcf20104e4b934cf577ef894219 : SUCCESS in 1h 27m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b614d6af695f4f35818fd8f08a033b0b : SUCCESS in 1h 07m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5ee176b43b54fbd89ae7758c8f847ce : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1aec16910b1748af822849971956a9fd : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":2},{"id":"780410dbf385622e6a63d001b24e96bff2c0f134","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 22:20:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6452fa77387aedeed0367578603d96761985cb2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 22:34:45.000000000","message":"Patch Set 3: Verified-1\n\n(18 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c61aa5ba98a45f7ae265fdcffa00559\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7807538a4425471fa1cc8b044c21e696 : SUCCESS in 8m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7575f7d2bc1c49419c19056b938485c8 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b95fbaf3c13a4d028f5f65896fbf670f : FAILURE in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0191d5a130df4c669c54674dcfd18260 : FAILURE in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"7104c8cf8114d5208d3a49291f10572cced0e4a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 04:41:20.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c89741cc2b8481198e16a9897d76d58\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4070075d24724240bfabdce75ba0ef2f : SUCCESS in 1h 18m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1e818a970484f12aede7a37b6e79262 : SUCCESS in 1h 18m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/79ff9d7c8061407a9db3ce3f8cf16dcb : TIMED_OUT in 1h 30m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ddfdc616aa2042e1af5f58627e185714 : TIMED_OUT in 1h 30m 55s","accounts_in_message":[],"_revision_number":3},{"id":"a659362155517ac891ae4752848f291473d67796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 19:22:17.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5580ab1e18d787c98f71af0a9024c02d1af0b031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 19:37:28.000000000","message":"Patch Set 4: Verified-1\n\n(17 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dae53625052a40ac9c072ad5638f4219\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0695a3479d2d477087fb77a52bee49d0 : SUCCESS in 7m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a50613f26b4449db9e98e09776e654be : SUCCESS in 5m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/31ab9331c06942028e58a761de03f797 : FAILURE in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b718798dc734b1c98eab6f8220ad9eb : FAILURE in 4m 02s","accounts_in_message":[],"_revision_number":4},{"id":"ab8c723074a7bd290c625383f84f112798cdb3f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 23:42:54.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"bdde0db5f87af15ff9e6ce3572bc73b55676536f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 23:57:57.000000000","message":"Patch Set 5: Verified-1\n\n(17 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/baa24b0bd3f441dab8faee61c83b6bba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15299996b48d42ef9e1f9d655e2bd8cb : SUCCESS in 8m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fcc5423719a84d75980d57b45843eaa3 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f4ce207a2fce473a99e9e89242c411d4 : FAILURE in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f5019df3cc4b05994107adc3801714 : FAILURE in 4m 04s","accounts_in_message":[],"_revision_number":5},{"id":"0e00ae23d3555ceffc978c8d1dd364c470cd52eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 14:16:47.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18bb322b110b4607be6bb70a44ae4147\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9244e6ec62ce487187aba2bf193646d0 : TIMED_OUT in 1h 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d975fb3ab3af4371a5b8380dd46e45ba : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d17d71df06e94f139a4354a0d6fce90f : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef831758b13a4373a6ef6f4f9f643503 : TIMED_OUT in 1h 30m 59s","accounts_in_message":[],"_revision_number":5},{"id":"bf07ad597f1be1f8f1d5cbcb7b0dda8b6db6421e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-27 23:33:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6dfeb6982f82fb173feec830052259bcb52f4d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 00:47:57.000000000","message":"Patch Set 6: Verified-1\n\n(17 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6aeb7c8d4a848808083c9ded503fefb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4cab4fcb7d41e3a57a975e3352a7fa : SUCCESS in 9m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6db0809ee9644ba1b3c2f53ebd539f99 : SUCCESS in 1h 04m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/377e136815684c7885fc1c0a09415749 : SUCCESS in 7m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f483c166b9344d55aadf0224062572b5 : FAILURE in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/585a3bc19e644cb18db1503eb7e26452 : FAILURE in 4m 02s","accounts_in_message":[],"_revision_number":6},{"id":"2c9f671cebd6dde294088cb8385c979bfdbda1a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 02:47:48.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/518bcfffafb74b6fa6ec01dc334e2511\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c377058911b4411a9b00f09895b2746d : FAILURE in 1h 44m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bb54067c9ff492c861dfdfa2e2794ef : FAILURE in 1h 13m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e44a43c3c814446a2edc215161d7286 : SUCCESS in 1h 12m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5270cb4366f4ca48e8d01d70aec7844 : SUCCESS in 1h 32m 09s","accounts_in_message":[],"_revision_number":6},{"id":"b97c2677861fe51660007bcc381ecf81b86ff03e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-04 20:13:36.000000000","message":"Abandoned\n\nShould all be in the parent now.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"07cf988a91927ebff11467c7233f63ad8b51b346","revisions":{"43957af4ec64e70132661e45591af8ac93a0bda5":{"kind":"REWORK","_number":1,"created":"2024-08-16 00:11:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/926453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/926453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/926453/1"}}},"commit":{"parents":[{"commit":"d2226fb61eb188bf2e2d7d49554619a587510df9","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2226fb61eb188bf2e2d7d49554619a587510df9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-17 18:31:55.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-16 00:09:45.000000000","tz":-420},"subject":"sq: fix a bunch of stuff","message":"sq: fix a bunch of stuff\n\nI still need to properly attribute all this to the various parts of the\nchain, but if nothing else I should get it off my computer.\n\nChange-Id: I71438a2f67b26d41392a131ed73003f7fc6b2d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43957af4ec64e70132661e45591af8ac93a0bda5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43957af4ec64e70132661e45591af8ac93a0bda5"}]},"branch":"refs/heads/master"},"0e31b12997f25388417a6b0ee82622ea1d19686e":{"kind":"REWORK","_number":2,"created":"2024-08-22 00:06:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/926453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/926453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/926453/2"}}},"commit":{"parents":[{"commit":"20a53d89c44504faa30a80b90fa29ed8d23b6ce5","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20a53d89c44504faa30a80b90fa29ed8d23b6ce5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-17 18:31:55.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-22 00:05:44.000000000","tz":-420},"subject":"sq: fix some stuff? maybe?","message":"sq: fix some stuff? maybe?\n\nThis is going to take a bit to properly unwind...\n\nChange-Id: I71438a2f67b26d41392a131ed73003f7fc6b2d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e31b12997f25388417a6b0ee82622ea1d19686e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e31b12997f25388417a6b0ee82622ea1d19686e"}]},"branch":"refs/heads/master"},"835301c76eda59ba4bcebd9e2470844fab7a3cac":{"kind":"REWORK","_number":3,"created":"2024-08-22 22:20:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/926453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/926453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/926453/3"}}},"commit":{"parents":[{"commit":"58a3e83a3bb6a03ad2b035f551a53fb274bf7e2e","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58a3e83a3bb6a03ad2b035f551a53fb274bf7e2e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-17 18:31:55.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-22 22:20:35.000000000","tz":-420},"subject":"sq: add some tests","message":"sq: add some tests\n\nThis is going to take a bit to properly unwind...\n\nChange-Id: I71438a2f67b26d41392a131ed73003f7fc6b2d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/835301c76eda59ba4bcebd9e2470844fab7a3cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/835301c76eda59ba4bcebd9e2470844fab7a3cac"}]},"branch":"refs/heads/master"},"fb97afca5a0db079e2bae182197caaf5d977c362":{"kind":"REWORK","_number":4,"created":"2024-08-23 19:22:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/926453/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/926453/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/926453/4"}}},"commit":{"parents":[{"commit":"e786ea229d537d61425c7bf170c7203fb697abbe","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e786ea229d537d61425c7bf170c7203fb697abbe"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-17 18:31:55.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:21:54.000000000","tz":-420},"subject":"sq: add some tests","message":"sq: add some tests\n\nChange-Id: I71438a2f67b26d41392a131ed73003f7fc6b2d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb97afca5a0db079e2bae182197caaf5d977c362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb97afca5a0db079e2bae182197caaf5d977c362"}]},"branch":"refs/heads/master"},"49dc073f4381f2a01490c11ba3dcdbccbd60b57c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-08-23 23:42:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/926453/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/926453/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/926453/5"}}},"commit":{"parents":[{"commit":"a683db748e917a631d1f0d4d7273a70341c572b2","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a683db748e917a631d1f0d4d7273a70341c572b2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-17 18:31:55.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 23:41:18.000000000","tz":-420},"subject":"sq: add some tests","message":"sq: add some tests\n\nChange-Id: I71438a2f67b26d41392a131ed73003f7fc6b2d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49dc073f4381f2a01490c11ba3dcdbccbd60b57c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49dc073f4381f2a01490c11ba3dcdbccbd60b57c"}]},"branch":"refs/heads/master"},"07cf988a91927ebff11467c7233f63ad8b51b346":{"kind":"REWORK","_number":6,"created":"2024-09-27 23:33:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/926453/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/926453/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/926453/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/926453/6"}}},"commit":{"parents":[{"commit":"ffe6c26d67868ab64986e93c38a1aba0d4ee5117","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffe6c26d67868ab64986e93c38a1aba0d4ee5117"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-27 23:32:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-27 23:32:48.000000000","tz":-420},"subject":"sq: add some tests","message":"sq: add some tests\n\nChange-Id: I71438a2f67b26d41392a131ed73003f7fc6b2d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07cf988a91927ebff11467c7233f63ad8b51b346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07cf988a91927ebff11467c7233f63ad8b51b346"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
