)]}'
{"id":"openstack%2Fswift~926538","triplet_id":"openstack%2Fswift~master~Ia95d335842608a72e81df9c10d5a018a1c83b98b","project":"openstack/swift","branch":"master","topic":"p-shard-repair","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-08-20 10:32:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia95d335842608a72e81df9c10d5a018a1c83b98b","subject":"Added a swift-graph-shardranges cli script / tool","status":"ABANDONED","created":"2024-08-19 14:36:07.000000000","updated":"2024-08-20 10:32:43.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9c5422efb440d920d89634f65872a6cd58495599","_number":926538,"virtual_id_number":926538,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 15:42:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-20 10:27:35.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"d10067322f854793d0e0897ea951b9fc30ab7572","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-19 14:36:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24e5a7ee2bef850a85b42b5c3b0165416722ad4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 15:42:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83515e66ab8e4b4985e02a34fc799729\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b6cb2bf60194b339facbf6db9449b0d : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/6bc2eb01fc844999b202085d1a732a6a : SUCCESS in 46m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb999d83c4df484e987680703f96f1a5 : SUCCESS in 41m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13ffc93dc9684b5fa7d8d94e0a564329 : SUCCESS in 49m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68f9993ebb754a6eae893beb9392933d : SUCCESS in 22m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/159d00d6d4094298955afddf2ac1c3d6 : SUCCESS in 22m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13d71f09c5b0416db289c9e384fc3d8b : SUCCESS in 23m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c902fcb13f24e3c930d585d8f3e1ef6 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3cf8f9f2d56e4e77a968c46001752461 : FAILURE in 20m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b0ad7d1816d4e39a0959fa196c115c8 : FAILURE in 18m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/36e23cb620324033bc7ef18c9efac0de : FAILURE in 18m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f73518c4110d4ffcbdc7b064fa4e4586 : FAILURE in 13m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7d8b30d01f684496836b57b2bb8a8ab5 : FAILURE in 22m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a4e0960145e4777b19618d99dee00ef : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b71a97d092941c791abd59080d43aa3 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6b838f0a2ca84a56985f0e438d8708e5 : SUCCESS in 20m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f655a974fbdd471682945e47a1eeeb35 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/80c3532fa45c4ff9ab2312f6f953061b : SUCCESS in 17m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a087136598344288aa563f8af5a118e8 : SUCCESS in 20m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/808f322b5d9d4b80856c3b3fa90d48a3 : SUCCESS in 5m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c5b62fd4d064fccb17d95222a8f9ea2 : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/73687da4b1e9410d838623d90ec40545 : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab15de9eec2d4ac683480fbdaca6da0d : SUCCESS in 49m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b941fd3bbbc4ae0a59dffb7c44b09a0 : SUCCESS in 52m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e32a2f372aaf4634ac023f3bcaccc851 : SUCCESS in 49m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/011dd900c897423591766bf8a2915c9f : FAILURE in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87e3ce31ab249bbb0483d32f51e55bb : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ced9db2317e54451842965fbd530e82d : SUCCESS in 19m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6217b10f64bc4f47b0602ed90cb18a00 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8ef3ddcabf1524b0affd19a958010f269e77cfde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 16:26:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27f7e5221e4d485ca5935eb1918d99c6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d5e18b1b3004e90b2ecd583f4bd204a : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f9313cef33742f1a09618951062f86a : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac1bd7caf9f745bb8c5926d51b50afe1 : SUCCESS in 1h 17m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/487e107d7a0c46d780762a1dd943d73f : TIMED_OUT in 1h 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ee0d9372ad27491f1a81c350d9ba71d78d050a43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-20 10:27:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e44ceec16fa3f034840d8d0c2cf2ffaac665787b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-20 10:27:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ee532a14687bf3caa279137b439410de7e6bb3f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-20 10:32:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c5422efb440d920d89634f65872a6cd58495599","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-20 10:32:43.000000000","message":"Abandoned\n\nsqaushed into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"84620916a42182c1bc50ef7d801f6007ca55fa9e","revisions":{"4dd0ad3cf5c2f9efef82628bbc0e68b92075d00d":{"kind":"REWORK","_number":1,"created":"2024-08-19 14:36:07.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/926538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/926538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/926538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/926538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/926538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/926538/1"}}},"commit":{"parents":[{"commit":"f3256c8fc0c52306022d48a54acfac9046d3ed1e","subject":"Added a swift-graph-shardranges cli script / tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3256c8fc0c52306022d48a54acfac9046d3ed1e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-19 14:35:56.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-19 14:35:56.000000000","tz":-300},"subject":"sq? pdb is no fun when stdin is closed","message":"sq? pdb is no fun when stdin is closed\n\nChange-Id: Ia95d335842608a72e81df9c10d5a018a1c83b98b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dd0ad3cf5c2f9efef82628bbc0e68b92075d00d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dd0ad3cf5c2f9efef82628bbc0e68b92075d00d"}]},"branch":"refs/heads/master"},"84620916a42182c1bc50ef7d801f6007ca55fa9e":{"kind":"REWORK","_number":2,"created":"2024-08-20 10:27:03.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/38/926538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/926538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/926538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/926538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/926538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/926538/2"}}},"commit":{"parents":[{"commit":"e6b73612d1dcd0bace2361338e6ae82eb591282b","subject":"FakeStatsdClient: Stop issuing DNS calls for host.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b73612d1dcd0bace2361338e6ae82eb591282b"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-02-11 04:46:32.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-08-20 10:26:48.000000000","tz":600},"subject":"Added a swift-graph-shardranges cli script / tool","message":"Added a swift-graph-shardranges cli script / tool\n\nTo correctly use this script you need to have python-graphviz and\ngraphvix installed. If you do you can use it to generate a shard range\ngraph.\n\nAs it currently stands it only supports getting information from STDIN.\nThis was a choice because we\u0027re thinking of wrapping it in a docker\ncontainer to include all dependencies, so would make it easier to use.\n\nCurrent usage is:\n\n  $ swift-graph-shardranges -h\n  usage: swift-graph-shardranges [-h] [-i] [--json-input JSON_INPUT] [-f {svg,png,pdf}] [-o OUTPUT] [-a] [-m MIN_SHARD_AGE] [graph_name]\n\n  Display ShardRange graphs\n\n  positional arguments:\n    graph_name            Name of the graph\n\n  optional arguments:\n    -h, --help            show this help message and exit\n    -i, --info            Data is from container-info\n    --json-input JSON_INPUT\n                          take input from file instead of stdin\n    -f {svg,png,pdf}, --format {svg,png,pdf}\n                          Format of the graph to produce\n    -o OUTPUT, --output OUTPUT\n                          Provide an output directory. Defaults to local\n                          directory (.)\n    -a, --analyze         Run ShardRanges through the analyzer and highlight the best path\n    -m MIN_SHARD_AGE, --min-shard-age MIN_SHARD_AGE\n                          Minimum age of a shard for it to be considered as an overlap that\n                          is due for repair. Overlapping shards younger than this age will\n                          be ignored. Value of 0 means no recent shards will be ignored.\n                          Defaults to 14400.\n\nIf you pass the \u0027-i\u0027 it\u0027ll assume in the input is the output of\ncontainer-info:\n\n  $ swift-container-info ... | swift-graph-shardranges -i\n\nIf you don\u0027t it\u0027ll assume json:\n\n  $ swift-manage-shard-ranges .. show | swift-graph-shardranges\n\nOf course you can also load in some json from disk:\n  $ swift-graph-shardranges \u003c \u003cpath_to_json\u003e\n\nIf you prefer not to use stdin then you can use `--json-input` to\nspecify the shards, but this doesn\u0027t work for container-info format.\n\nThis will attempt to open the file if you can, but in anycase it\u0027ll drop\nthe graph file in whatever format you specify in the output directory.\n\nIf you use the \u0027--analyze\u0027 option then it\u0027ll run the ShardRanges through\nthe s-m-s-r analyze function and then mark the good path in green.\n\nChange-Id: I5b02785631887f1a41430013ca39d35f5523baf3\n\nsq? pdb is no fun when stdin is closed\n\nChange-Id: Ia95d335842608a72e81df9c10d5a018a1c83b98b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84620916a42182c1bc50ef7d801f6007ca55fa9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84620916a42182c1bc50ef7d801f6007ca55fa9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
