)]}'
{"id":"openstack%2Fswift~927003","triplet_id":"openstack%2Fswift~master~Iaaf494fd4e02017005cb3811b673f967bd6b5e1d","project":"openstack/swift","branch":"master","topic":"p-db-state-var","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-08-26 21:30:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaaf494fd4e02017005cb3811b673f967bd6b5e1d","subject":"sharder: rename state to db_state when it\u0027s the DB state","status":"MERGED","created":"2024-08-23 09:56:21.000000000","updated":"2024-08-26 21:33:45.000000000","submitted":"2024-08-26 21:30:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927003-p-db-state-var","meta_rev_id":"574a540703a5e6ebee494fc7e0b3ec7e5d486e90","_number":927003,"virtual_id_number":927003,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-26 21:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-26 17:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-26 17:45:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-23 11:07:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 17:45:16.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"5034421d6db1602459e1c0d5750ecad852c3f277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-23 09:56:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"589b4de06e5c0438b6599a16a5da6f642cb8a430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 11:07:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fed572f1ee664ff586e46f896330fdbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d3a89dc38a45f6b85d0ebb39d7cd45 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d7c162a3065f4111adf5b8e46834e593 : SUCCESS in 58m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/077416943d5c40dab1973b751a97947c : SUCCESS in 42m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0655597f6fc746fab292c8a5539c60bb : SUCCESS in 1h 04m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89d7a248602d4816b5e77785dce8eb99 : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c14264fdaaf646a2b9f9a16fd2abc79e : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44c000d8bc2146b2b1166ad0852b95e8 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e9216136fbe47b5a3495c173bc1a33e : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e82a767b9c54c578a5842cdf48afd29 : SUCCESS in 21m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/88749e1f012d4d9e9ac0eed79ae6a211 : SUCCESS in 21m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1a40c80964e648d7a680b078b574c538 : SUCCESS in 19m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cd7d6b07851341148f404fd202412f93 : SUCCESS in 20m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/488f47385dec450f85485fd08e6c0891 : SUCCESS in 26m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/658d8c611d90429c96b136593b577b90 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a57356e0d2fa42e98fd21bed0019e819 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cdf5cdf1e5794906a60c65e2b3596b75 : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/db111192297f46368c1a1e38ae8e82a4 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a4b732507b41452cb119bb7d24047d53 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37852d1b0b214164901ef577aca87505 : SUCCESS in 16m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45384dede79a4667bef92575f8adb7ac : SUCCESS in 5m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/71312246e4c047919758ac2b1ff7acb6 : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a22177d7c4d5481fbf610729688cbcd0 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd747f9f2b924110b6363aa36e6d508f : SUCCESS in 46m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b214cca7fbc2474fb84ff0a4b149d219 : SUCCESS in 49m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8949be09b6524a67a3f991d0cb267da0 : SUCCESS in 37m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f8c659684cd4609b643108cc87819c4 : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a3a63f56c794317a06f8b52289239b2 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c65cecdb41984725befc61f3dcff2d3c : SUCCESS in 16m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69f55719cc3c4e109ac33be77142fccd : SUCCESS in 59m 38s","accounts_in_message":[],"_revision_number":1},{"id":"939528d24a15791ae22c3dfa0a001586a83d25ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 07:59:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e257244aed4481c8dd3601a336c6254\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d13202d0927465c89027ae89a26d1d9 : SUCCESS in 44m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb3f5c92185644c38e7c41b8beed59f8 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3189d6cffe854143850104cd4422eb36 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/61e45f549dd0411d9f6ce17b625e84ae : TIMED_OUT in 1h 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8f6ff261c5a41e6ae310eb7182d5c324c68c0d96","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-26 17:45:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1bbe96a23bdbde32abf8931902a1d39a5cf998a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 17:46:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"161ff0bd14f7bb7e43c7af1bd48832d1560996d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 21:30:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb7ad347758f4d059cbe27dd33826275\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4d08f458bf45fcaedf8f6127676525 : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/e2ecfeb1ea5d4f4a9ef1f4d1badc9daa : SUCCESS in 57m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d4cc6950b85425488ae0019ebd60ed1 : SUCCESS in 49m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/424afb675e344261837e2c00d42323ad : SUCCESS in 55m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/25a203ad3b4b4c9e8d2a5e677b2fd11c : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/92cef4e9bcf44536a3d3f4440ba99046 : SUCCESS in 19m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d1c22f99a26b43ee8bb984d738a344e0 : SUCCESS in 22m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8a01a1de2e44abba0624d1e0a298444 : SUCCESS in 22m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/87e00fc3f7ed4c1a94f4afb29864ba3f : SUCCESS in 19m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/67f0f93e26b34a33b2d2e7ea489645ac : SUCCESS in 32m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1433cca2e7014c4f8741957344a3fb6c : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a425cc49449c495e920f43b6c4501eee : SUCCESS in 16m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7432db1e21e14cd6b9431a0c44444c92 : SUCCESS in 18m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7d79a0af56534554aa959429c36e9bdb : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6afc4329c8ab4cbbb7495d5d2b924030 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0e5aab270a0044cf8385f226ea19333b : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78730a052d43416ab6f975737db0cf73 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e14dfc283324151945864e54c843aef : SUCCESS in 8m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ae8ac9f2ac44c8290e427681ac2b615 : SUCCESS in 52m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a67c7524810a483e9026cd24b3da7a49 : SUCCESS in 50m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/228736dee2f5427fa426adb5784bf24f : SUCCESS in 50m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/944438e6e04549c091b3fb70bddb7f1f : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378dd1553a0d4d1587ba94664d192a16 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f795a54dcb64e969aa32cdee1e6ae7c : SUCCESS in 23m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7d65eb12ea542aebf0e27a3ccd1bef5 : SUCCESS in 43m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0a86ab38dcedc0df265b73b7a335487dbfb0bad5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 21:30:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"574a540703a5e6ebee494fc7e0b3ec7e5d486e90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 21:33:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37d79d476558485f98c87d5b0a08307e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22b50c24d33b4d73bf00c50fade29071 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e843aff7e3aa408f88acd6167383cbfa : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87bf4bf5da2cb0c8f640d96cd4304bec422dbaab","revisions":{"87bf4bf5da2cb0c8f640d96cd4304bec422dbaab":{"kind":"REWORK","_number":1,"created":"2024-08-23 09:56:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/03/927003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/927003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/927003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/927003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/927003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/927003/1"}}},"commit":{"parents":[{"commit":"d830703a32d16d4618a30684da92b78bf48deff6","subject":"Setup proxy object GET and HEAD tests correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d830703a32d16d4618a30684da92b78bf48deff6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-23 09:53:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-23 09:53:07.000000000","tz":60},"subject":"sharder: rename state to db_state when it\u0027s the DB state","message":"sharder: rename state to db_state when it\u0027s the DB state\n\nSharding involves both container DB state transitions and ShardRange\nstate transitions. To avoid confusion, always use db_state as the var\nname when referring to the container DB state value.\n\nChange-Id: Iaaf494fd4e02017005cb3811b673f967bd6b5e1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87bf4bf5da2cb0c8f640d96cd4304bec422dbaab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87bf4bf5da2cb0c8f640d96cd4304bec422dbaab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
