)]}'
{"id":"openstack%2Fswift~927177","triplet_id":"openstack%2Fswift~master~I86b0285d478ca370c6df22a2cef5c8480b1c8674","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-09-03 18:02:30.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-08-21 17:40:35.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{"39146":{"account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"last_update":"2026-09-03 18:04:44.000000000","reason":"\u003cGERRIT_ACCOUNT_39146\u003e replied on the change","reason_account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}}},"hashtags":[],"change_id":"I86b0285d478ca370c6df22a2cef5c8480b1c8674","subject":"s3api: Reject unsupported Transfer-Encoding","status":"NEW","created":"2024-08-26 16:06:40.000000000","updated":"2026-09-03 19:10:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"ae04b3e9bf39cf44b1202ce668c1182fc2f3ac94","_number":927177,"virtual_id_number":927177,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 19:10:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 18:23:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 20:18:56.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2026-08-21 17:40:35.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-08-22 02:02:07.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-09-03 18:04:44.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"CC"}],"messages":[{"id":"45d0da1bbb3c68c0512181b44d81cfa2d169660a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-26 16:06:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"986f064df39cd2aa4f488afad3b2ce381a94f5c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 18:23:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e5f1138910848c5ab68fafb65549814\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3219df507705485d940dbe81f4bfacb6 : SUCCESS in 11m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/2ac8cd22217149f49e6cf442c0b73904 : SUCCESS in 59m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c16cdee540da406b934648b711ff71c0 : SUCCESS in 48m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c34855cf6a0546089618ab451154e2f6 : SUCCESS in 58m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/908d075d8bff4e568c1c436017bf83a4 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2cd47dc34ff48d591f84a6b61909a89 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/322633d9e7fc4388bc1a695ce39e1feb : SUCCESS in 26m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0583697785264361a4982b16a30f33c8 : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b1e55e71fed4dec8269296ccf22b912 : FAILURE in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2defa8d5bf3e40598322f94abec95eee : FAILURE in 19m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/df4280d44af34649a35497d01be1eba6 : FAILURE in 18m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c09a3282e5eb44108dafd7f340b281ec : FAILURE in 19m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0888782f42244c16b3d8bb89666fda73 : FAILURE in 26m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0746d5495eca40eabf834ccdba8d7a7a : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/00f2a90646554e6ab1c9910191329f09 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/862277f1686e46489f329a90860bba68 : SUCCESS in 18m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/34bdf63e1db54bc0bc42b0a424d29105 : SUCCESS in 21m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/02fa7e3161f8477cbbd976ecbb8454d1 : SUCCESS in 21m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b4c7f7120bcc490da206cc51c7143f63 : SUCCESS in 21m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fba53ce993354607b29763f57cdaf0c1 : SUCCESS in 8m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/297a12f840cb40b6bc7fce41666bc0ed : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2e66161d51784a678b43e84f225495b5 : SUCCESS in 9m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f52613e83ad64a3791780aa242b07fc0 : SUCCESS in 46m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d41b1f3fbc3e427fbdbd426189f95e25 : SUCCESS in 53m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/203845b650fd49fdb8a208fe9b33784c : SUCCESS in 1h 05m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cec1bb4e291f47e8bccec6d8dae08079 : FAILURE in 25m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0bb7e1823b748b6961b11523d0aeac0 : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/379c129eef1a44ad9195f4231a589709 : SUCCESS in 18m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a40c2ed89474602a09b3f93e3d4d0ba : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0a7744b9ce26729808b46d1fd2a72380522b0fce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 21:40:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7913919b046f435d8be345d29686c1c4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2bd7838f3f5c4e989a07b8a9995c1856 : FAILURE in 1h 27m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/062708b3f1b74d26aa803e99732265ff : FAILURE in 1h 24m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4659d9b3436947949c8d3cb7a84db3ae : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f6e0c830eee4b09969d3e05f6b94432 : TIMED_OUT in 1h 31m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f85e507c36f95ca2bc2534f4454095a85742fcb5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-06-19 21:09:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"bcdd5e87c754a38c8a0b6a695b476d302aba7b30","tag":"autogenerated:gerrit:restore","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-08-20 17:49:10.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a41a3aac21c30f123b20f55686049482de53ece7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 17:49:20.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 927177,1","accounts_in_message":[],"_revision_number":1},{"id":"6dc069ae3378e6b4cf157399e3b9bbd0939f59b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 17:49:23.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 927177,1","accounts_in_message":[],"_revision_number":1},{"id":"7b23b1a500631412a68f5fc9e3f9dd666ac2bea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-08-20 18:14:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f79f4dc41b14acab1f21b29de1e913564176932c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-08-20 18:15:50.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"36883a22700089944fe1ab7e83e95952285ff0a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 19:06:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/830408a5ccc643a0a1544ceeeb6f4d43\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5616e003e64244b397ad286b1a507cd6 : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":2},{"id":"392d3d004f95e74e8840ada30e15bfc2928ec90d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 19:35:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90c0138c95474dfb81763edaff116327\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ef1ec9b0eb44c8ac26a07e0a5859b9 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/635468fe84d942349dc708c344a9d951 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/c8a43f6e54c74888b157ea5b5cfd79b2 : SUCCESS in 1h 15m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6046574fa7e64a96b6e1a6d1ae225d4f : SUCCESS in 32m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdb1165a50bf48dd9d6f1b02bcca76aa : SUCCESS in 1h 01m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b831044144f4491e87c682c2e621ff40 : SUCCESS in 30m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e37666920a94f20b086312b80aa164d : SUCCESS in 10m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d86b40c8c0c49c69d49ee2394e56b75 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5471e0ae8f3f482ba595858b939da1cf : SUCCESS in 11m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bee594e55d594ac09344b71c5f19a5c5 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1a4a3a183e3244c0803e7d5c442f6aa5 : SUCCESS in 27m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/895baefdad674521aea9670fc8aad79e : SUCCESS in 9m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d6ea403706284135a14443495475aa69 : SUCCESS in 12m 58s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3c62f005045846a1b14b6bce83acf2b1 : SUCCESS in 20m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/941e52af6d3f46a1a4b11addec78a9c0 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2b7481b77c67421b9fb58910e56d5dd8 : SUCCESS in 8m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1c72153c9a7744c9acc0ad0ead94f8c7 : SUCCESS in 9m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b32202707a4f479f86f16f6c96ecf165 : SUCCESS in 10m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eec8be01676941c8bb5c070576425914 : SUCCESS in 5m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5bd92091c3644186912257e8fbd55a7c : FAILURE in 4m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/80eee51d0b2e4ec6b7624bfd18ebbe4b : FAILURE in 9m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a08d5d61c094436e9e338055067c2a05 : SUCCESS in 59m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0c984a6eae14c688a1684e5de57996e : FAILURE in 15m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac8abe0e1e25474e9261d5a8605f7a8c : FAILURE in 34m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/961e115e26aa4489ba9a4d7b3fcac6b5 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d8d6297510441a9a938a5e4f5e35de : SUCCESS in 2m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/70546f7343f84a59b4646b5328cddcdd : SUCCESS in 19m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adbea7a0134347a39fd1852144c76eb0 : SUCCESS in 27m 14s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"153ace95250b3a60587c838fc27b7852171afbcf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-20 23:38:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"996a901649bcf5de0af16916dc879c0833d05e6f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-20 23:40:05.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"8b1d89f63e960e3a1ec14c3a4be81ee754bdff93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 00:41:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d796035e89874e94a39b26c8bbcf8684\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7d52282898419a8c1a5ccf85fd35c8 : POST_FAILURE in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae4556b08412494b86b55cc87c7f049a : SUCCESS in 6m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/98ff13a593b94ec08cb88da8ab2ecf4b : SUCCESS in 29m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ebc0aa29da564b50bd623ff559fea985 : SUCCESS in 1h 00m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b0f9ae75572483c8fcb9f5c37cfbe01 : SUCCESS in 48m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/586ef5d031e743168f14cd8a9050cff5 : SUCCESS in 43m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/faded2b5f24d4649a0df9002e4441631 : SUCCESS in 14m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71572ad6dc2142989872a72796e3b5b8 : POST_FAILURE in 15m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82f5699bb173469ebaf6428319e3616a : POST_FAILURE in 15m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d7c9bedf82b4ee0bda01a26c029f94c : POST_FAILURE in 9m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04635903511c4b0ba5603508197f2b66 : POST_FAILURE in 14m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0ae03d4b7ea3403e960c61bea5538073 : POST_FAILURE in 21m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/910f245a2ec64641bbcab7ba20fe1329 : SUCCESS in 11m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/44f8fa13387e43149baeff785228ac72 : POST_FAILURE in 14m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f463c9cd7c194b42b8424c60bec1f9d2 : SUCCESS in 8m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/60c2ec9bb59349d6bc3c824f22e16943 : POST_FAILURE in 16m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a5b94999b7d241869688a3319dd56997 : POST_FAILURE in 18m 28s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b92386cf99a94664998b653ff3bc7c28 : SUCCESS in 10m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/793641f1bed444a89d4880ada87d6696 : POST_FAILURE in 8m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f9f9c0646cb451cb83c3748fecf0943 : FAILURE in 6m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ea098c6efa224bd8bc20de80bd3b8672 : SUCCESS in 5m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cc5ddd92fa544b3832c5a91137221e0 : SUCCESS in 36m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97c27ab4ef8a4ab9a65ce4dce1e4d052 : SUCCESS in 31m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5df397bebfe544d48804794037dd25c6 : POST_FAILURE in 26m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88f4e68a63b743a8b4d1a965d1a3ebdb : SUCCESS in 23m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f55c70f8478343699d515751d9ae711c : POST_FAILURE in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0db7387d2912422eb45b25bab1e2e4d2 : SUCCESS in 14m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e553f17d934f43bea7da36ed6e48d644 : SUCCESS in 50m 08s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"1d73655217daf51018eddad8036248f3be70dbdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 00:57:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4b58e7980b94540b5dcc2f0dfb62e9d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab173acc5b494d0bb4418fc2a8fb9db9 : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":4},{"id":"e216ea783e4fd050f3a1e9acc285a7daa0246a97","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-21 17:40:35.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"0131db05beb6e93be8df38043c3af1bb1ed1ad16","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-22 02:02:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d29bad471bbeb024e7524ad38251823f58f0cdbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-09-03 18:02:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9eeba870d5d49f600bcc2d07980f16493c646d05","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-09-03 18:04:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"79db17858db11e62f99cd153a9490cea20da8bd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 19:06:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fda507fe00cd4ace99d00e482e4b9239\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/55f954e7e53741af84126e26801efa1b : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":5},{"id":"ae04b3e9bf39cf44b1202ce668c1182fc2f3ac94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 19:10:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8438e1ffac24b5b9e3d2a40fa22f1ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11d63daf9a1347f69608f19fa14b7ec5 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c8bba7dcc4445ef890502d0ced7397d : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/a35fecd590cc4032abe79b9640f9304d : SUCCESS in 1h 03m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b273f412ca2849bdb68571b926b2026c : SUCCESS in 42m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a2c5f51c869c4d8ea2b4790bcf17f448 : SUCCESS in 36m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9ee46e80af644d78ac554a87240c8d6 : SUCCESS in 1h 02m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e19757a9b7614221ab57c1a38dc06dcc : SUCCESS in 10m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e5f5b52d1fa4c068f10a5a3d4ce947e : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf44ab92695f4c598371d81f7e507816 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e416909c72343d1bd29f0d06053a840 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1663c876270241b6b6e775d3abae07d7 : SUCCESS in 21m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1e120b4b8ad543b5b2e4938dd9d7b3e9 : SUCCESS in 22m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b70c9bd7115a41ae8811380d1a55ce32 : SUCCESS in 24m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2496c8c7856e4cc59c369ce0daa7ff4d : SUCCESS in 18m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/57904f18d20e45149dc8644b7070229d : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3d29236a50214562a8d35ffccbcba4bc : SUCCESS in 8m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/03866a55b90d44e792c077e1baa9fd71 : SUCCESS in 12m 35s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4a42ef0ad574366acf6dbf718f24876 : SUCCESS in 12m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a34a8d23d664d949905b1506fc30609 : SUCCESS in 6m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fb6b4339859f4f49aa2482de4de0dc9a : FAILURE in 31m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/968321a82cc648758e07f18663f760a7 : SUCCESS in 11m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cf8b63fda234cdea68f87eb87276932 : SUCCESS in 1h 00m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4d0bdf589144302aa0b259e251fa3e4 : SUCCESS in 34m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f85a55b18b146f9b43ac82c18ae4fb1 : SUCCESS in 31m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f1ffe5e871b4fee97ddb1142ce4d72b : SUCCESS in 27m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b8b2ed97654c188c830354dc40e9f7 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3099998446e4a61945fbbe746a7f946 : SUCCESS in 14m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c10c82e5b5c4fdcb23708ccded3e2be : SUCCESS in 56m 36s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d869af714c0b223ed6a66d9751150590a629a234","revisions":{"8dc89d23f8fba92983729a3f84840994c9fcce73":{"kind":"REWORK","_number":1,"created":"2024-08-26 16:06:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/927177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/927177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/927177/1"}}},"commit":{"parents":[{"commit":"70948481f92a26466c23fa2d2cc072b77c6e0e67","subject":"s3api: Clean up some a bunch of errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70948481f92a26466c23fa2d2cc072b77c6e0e67"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 22:16:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-26 16:06:24.000000000","tz":-420},"subject":"s3api: 501 \u0027Transfer-Encoding: chunked\u0027 requests","message":"s3api: 501 \u0027Transfer-Encoding: chunked\u0027 requests\n\nApparently that\u0027s what AWS does??\n\nChange-Id: I86b0285d478ca370c6df22a2cef5c8480b1c8674\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8dc89d23f8fba92983729a3f84840994c9fcce73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8dc89d23f8fba92983729a3f84840994c9fcce73"}]},"branch":"refs/heads/master"},"40d3bdae020242241594afd5e15645a622059f20":{"kind":"REWORK","_number":2,"created":"2026-08-20 18:14:37.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/77/927177/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/927177/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/927177/2"}}},"commit":{"parents":[{"commit":"e19c896da0eeaf09419726106c46acc3e8bdce39","subject":"Merge \"s3api: restrict headers sent with copy check HEAD request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e19c896da0eeaf09419726106c46acc3e8bdce39"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 22:16:19.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-08-20 18:14:19.000000000","tz":-420},"subject":"s3api: 501 \u0027Transfer-Encoding: chunked\u0027 requests","message":"s3api: 501 \u0027Transfer-Encoding: chunked\u0027 requests\n\nApparently that\u0027s what AWS does??\n\nChange-Id: I86b0285d478ca370c6df22a2cef5c8480b1c8674\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40d3bdae020242241594afd5e15645a622059f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40d3bdae020242241594afd5e15645a622059f20"}]},"branch":"refs/heads/master"},"5ce538640681449dc9f6d19b941420d33f378543":{"kind":"REWORK","_number":3,"created":"2026-08-20 23:38:07.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/77/927177/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/927177/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/927177/3"}}},"commit":{"parents":[{"commit":"e19c896da0eeaf09419726106c46acc3e8bdce39","subject":"Merge \"s3api: restrict headers sent with copy check HEAD request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e19c896da0eeaf09419726106c46acc3e8bdce39"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 22:16:19.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-20 23:35:36.000000000","tz":-420},"subject":"s3api: Reject unsupported Transfer-Encoding","message":"s3api: Reject unsupported Transfer-Encoding\n\nAmazon S3 rejects ordinary requests that specify Transfer-Encoding with\na 501 NotImplemented response. Swift currently allows unsupported\ntransfer codings to reach later request processing.\n\nValidate Transfer-Encoding while constructing S3 requests. Allow\nchunked or identity only for streaming requests that also use the\naws-chunked content encoding. Remove identity before constructing the\ndownstream Swift request, since swob otherwise rejects it while\ndetermining message length.\n\nLog rejected Transfer-Encoding values so operators can identify affected\nclients. Add unit and compatibility coverage for ordinary chunked and\nidentity rejection, supported signed and unsigned aws-chunked streams,\nand downstream message-length handling.\n\nAssisted-By: Codex:gpt-5.6\nChange-Id: I86b0285d478ca370c6df22a2cef5c8480b1c8674\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ce538640681449dc9f6d19b941420d33f378543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ce538640681449dc9f6d19b941420d33f378543"}]},"branch":"refs/heads/master"},"67a4f03e42a9d623208529f8cccac16ac275f0e0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-08-20 23:40:05.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/77/927177/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/927177/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/927177/4"}}},"commit":{"parents":[{"commit":"e19c896da0eeaf09419726106c46acc3e8bdce39","subject":"Merge \"s3api: restrict headers sent with copy check HEAD request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e19c896da0eeaf09419726106c46acc3e8bdce39"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 22:16:19.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-20 23:39:08.000000000","tz":-420},"subject":"s3api: Reject unsupported Transfer-Encoding","message":"s3api: Reject unsupported Transfer-Encoding\n\nAmazon S3 rejects ordinary requests that specify Transfer-Encoding with\na 501 NotImplemented response. Swift currently allows unsupported\ntransfer codings to reach later request processing.\n\nValidate Transfer-Encoding while constructing S3 requests. Allow\nchunked or identity only for streaming requests that also use the\naws-chunked content encoding. Remove identity before constructing the\ndownstream Swift request, since swob otherwise rejects it while\ndetermining message length.\n\nLog rejected Transfer-Encoding values so operators can identify affected\nclients. Add unit and compatibility coverage for ordinary chunked and\nidentity rejection, supported signed and unsigned aws-chunked streams,\nand downstream message-length handling.\n\nCloses-Bug: #2164595\nAssisted-By: Codex:gpt-5.6\nChange-Id: I86b0285d478ca370c6df22a2cef5c8480b1c8674\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67a4f03e42a9d623208529f8cccac16ac275f0e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67a4f03e42a9d623208529f8cccac16ac275f0e0"}]},"branch":"refs/heads/master"},"d869af714c0b223ed6a66d9751150590a629a234":{"kind":"REWORK","_number":5,"created":"2026-09-03 18:02:30.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/77/927177/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/927177/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/927177/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/927177/5"}}},"commit":{"parents":[{"commit":"e19c896da0eeaf09419726106c46acc3e8bdce39","subject":"Merge \"s3api: restrict headers sent with copy check HEAD request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e19c896da0eeaf09419726106c46acc3e8bdce39"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 22:16:19.000000000","tz":-420},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-09-03 17:47:03.000000000","tz":-240},"subject":"s3api: Reject unsupported Transfer-Encoding","message":"s3api: Reject unsupported Transfer-Encoding\n\nAmazon S3 handles Transfer-Encoding according to the operation and\npayload framing. It rejects ordinary PutObject requests using HTTP\nchunked transfer with 501 NotImplemented, while allowing the header for\nsome XML operations and SigV4 streaming uploads.\n\nValidate the Transfer-Encoding value while constructing an S3\nrequest. Allow \u0027chunked\u0027 or \u0027identity\u0027 for AWS-chunked PutObject and\nnon-copy UploadPart requests, and reject unsupported operations before\nforwarding the client body to Swift.\n\nKeep the header intact until SigV4 captures the canonical request.\nRemove identity afterward because WebOb rejects it when determining the\ndownstream request length; chunked requires no rewriting.\n\nDefer validation for bounded XML operations to S3Request.xml(). This\nallows a single chunked coding while retaining body limits and\noperation-specific error precedence.\n\nAdd unit and cross-compatibility coverage for supported streaming\nuploads, bounded XML operations, invalid coding lists, error\nprecedence, and downstream body handling.\n\nCloses-Bug: #2164595\nAssisted-By: Codex:gpt-5.6\nChange-Id: I86b0285d478ca370c6df22a2cef5c8480b1c8674\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d869af714c0b223ed6a66d9751150590a629a234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d869af714c0b223ed6a66d9751150590a629a234"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
