)]}'
{"id":"openstack%2Fswift~927425","triplet_id":"openstack%2Fswift~master~If6731a6b6b16fbc4eebb61254ed10b53e1767a0f","project":"openstack/swift","branch":"master","topic":"euclean","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-08-29 00:21:31.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-10 10:23:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If6731a6b6b16fbc4eebb61254ed10b53e1767a0f","subject":"diskfile: Treat EUCLEAN like ENODATA","status":"MERGED","created":"2024-08-28 18:48:58.000000000","updated":"2024-09-10 10:24:23.000000000","submitted":"2024-09-10 10:23:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"927425-euclean","meta_rev_id":"11ae3444cf7e84b422c7fd65652ac3cada6f3f95","_number":927425,"virtual_id_number":927425,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-10 10:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-09-10 09:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2024-08-29 00:21:31.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-09-10 09:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-28 20:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-28 20:28:53.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-08-28 20:33:46.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-08-28 20:59:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-10 09:15:08.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"6ff077ef2c07257fa19426d6ee3c310348c13911","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-28 18:48:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fee7aaafc0b08910181835476c1106c50dab988","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-28 18:53:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c48e047ca20a3f156f88118c8f54b886100e114c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 20:27:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67bcaf2d6af42db8498a0d16ef2b945\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eeecb6637dbe4e339a0bc69ac2fbbc4f : FAILURE in 54m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c47a38cb20b84cfa903be85c7ef9f9cd : FAILURE in 58m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f39ebb8b7c214d219b47afac20132b77 : SUCCESS in 59m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e43762e83b924984af4c845a267e9ff1 : TIMED_OUT in 1h 30m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5a6c8ff3060a0b3fd9f38a4ffa32840cc211c510","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-28 20:28:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1292e51a7e2ff9b1edfd8a011894aea1d9b7da20","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-28 20:33:20.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8ca918a185a1e3990f745032c6802fd4dba067e4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-08-28 20:33:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0025a7d69fc3f07322eee2f3105bcd1206cc695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 20:59:13.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59deceb391664025a5027d55e9c43248\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c5d560940a432592af1ae17f12c60e : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/58dc98825af343049f64519c3fca423c : SUCCESS in 59m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f9bc2ade6f04d589997dac2c4663755 : SUCCESS in 49m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac84ca75f36d49549e0cb0b855956d01 : SUCCESS in 53m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63cf9eb90ea74cfa9a5623c7737dc867 : SUCCESS in 28m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/daa66ac2bb084ce0a9187f0b21fbdcfd : SUCCESS in 23m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1bdcefd076a498eb4d6510833ec8c1b : SUCCESS in 23m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9672e10e7baa4a27bbb3d8c25a8d4edf : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f706d6888c2c443294109344fd3f90f5 : FAILURE in 24m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2493ce6d74264dfebd96c740e8b34c94 : FAILURE in 16m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/462f93455be440eb93add5ba11c4e793 : FAILURE in 21m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6e971e52f97f4cfbafd51efc210d734e : FAILURE in 19m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b48d67c8aeca432f97bc39c027cb7adf : FAILURE in 34m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e95acc7acfb34a928c25eb9334d00860 : SUCCESS in 24m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e16c0a3a80ac4d07852c9207f5bf0713 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e67f8c3ea4e24741a4246d9e8b7320c2 : SUCCESS in 22m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e7e89cb50ea2448da4e3dd24f72a7781 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e8b9f51f94784f81a3c049168ca6d2f1 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dcbb822a74ee4ef6acb210c2d92d005d : SUCCESS in 28m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/68157135ee924a1f9b5d1d0987483932 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9cef26bcbbd546b7b6608502e1fb6efb : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cbe3b9052dc84c3e851785b0b2734f51 : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/93592c4d73964feb8a32c726c42b4416 : TIMED_OUT in 2h 01m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07aaf650c7e9495ab399229d91c8821f : SUCCESS in 44m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/050d44aab13846548635f805a781942b : SUCCESS in 56m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18738b35192140db9c7c243b4be4dffe : FAILURE in 22m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48509839c544921b7239ade9a91821e : FAILURE in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0bdd49e4d1204a90837515b02332690f : SUCCESS in 18m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c05021b5bc604c76a6f26517fbd7c820 : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2dceaf0339d3c2c8f507ccc89514b3e65170a431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-28 22:38:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"34b139afa1bc0dd15eb2dea17331cb3d281b7f38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-28 22:58:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b6d458c1bd22423669323ae42a1039dd8cd107b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 00:09:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a190a41fce05474f92696ed46ff3261d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25082dbad8694857bdd7e227148af675 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/4f05087627e14c81ae721b7a5bd699cb : SUCCESS in 1h 00m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/850f5f8873704de684d58d3c6f5157cc : SUCCESS in 36m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f70ea7a10994c958a5f44f253024647 : SUCCESS in 59m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87290dfc700848cb9ee142a0c21cf8f0 : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d20d368335754ddb9560baad734f592b : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e8211ec169a49b5a79f1eb5bcef762e : SUCCESS in 18m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a568c59aa42d4bcbb70ac2f7ed997932 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/308eb05b9f2c4a3abe51b583fde5d1e1 : SUCCESS in 16m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ebef6fcebace4d199f04bc553f168fb5 : SUCCESS in 15m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03fe8c0bdff64a9294dc80057590ab8b : SUCCESS in 15m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ea6b8f04fdd140419a386b90531e1ee6 : SUCCESS in 16m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/053b638f947a49318cde8fa279c7824e : SUCCESS in 27m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06133e709b1646fda5a2e3f62a1a5039 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/948b9bf2cc7349f2b1aa67398db0cc86 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/979a3091edcc42019895fd03f0418aeb : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e4851e9074044c279220d41d3d8ec4b1 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4743b1ea95b146cb9d56bd2534869086 : SUCCESS in 14m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fd2068691d544584bf1b0736134b1514 : SUCCESS in 17m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d54d839f4d94f73ae74033ff2818901 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e4d6c681de654c0892964453cdae223c : SUCCESS in 12m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dc06e2aa3833499e89802078ad395ea7 : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3c9d28db8d8464d9f108907cbce65a2 : SUCCESS in 43m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08e4a0755675465485115a1fa487c9d8 : SUCCESS in 47m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b35bab1d6674081a71984d51fc9f561 : SUCCESS in 34m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53eb882a4e9642ddb8bbc68b3911f27c : SUCCESS in 20m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f5ba85ecc24474988e621fdac426d8d : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a45728ea48440c8a9e73097ead67566 : SUCCESS in 17m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/136e933c64e54759823f3076a1d705ce : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":3},{"id":"78c7ce4fdb805eea0617c7f71aa097bd2f3875b9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-08-29 00:21:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"676e004278300fdecac3263dafd65f76e2558a01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 00:33:07.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/644f042be0e2434eabe84db94748e0e9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0c6534c19dd4158b825b5f3c72d28f1 : TIMED_OUT in 1h 30m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2de5cd89966e48338cb3631e0c16ed7c : SUCCESS in 1h 13m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/41ce754d190b48668125295a87ca86df : TIMED_OUT in 1h 30m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/346d3ab3663a4ecfb3c0e018680ac493 : TIMED_OUT in 1h 30m 54s","accounts_in_message":[],"_revision_number":3},{"id":"5d1361ca5477a0bb83257f0ded8ad3d4fe9e9845","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-10 09:15:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4af63882888f62cc27250a29d5c932e3787bf045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 09:15:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"77c45d2ddc43bbff95cc3eca0cc26bfbe9727987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 10:23:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e3ac4e75174f6581670967121ecca5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f3d324cd7c4ae897580b86b3483237 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/b4174839545744449fb3811bfd08e2a5 : SUCCESS in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5295ea5a902f40eba0962c1704a6bf1e : SUCCESS in 51m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e5e9b78cbe64802bf284b30465ff15c : SUCCESS in 1h 02m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9a196defe6cf470887e14bf691844e18 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/587a9ff1df12402aa79f4870b9f60c75 : SUCCESS in 20m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03dfe218d855458fb0fac9c1d0d5c5da : SUCCESS in 17m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d38530a346bb4ee3bbe6d769cb9d37d7 : SUCCESS in 22m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6a6b328d121541a295aa216d55534353 : SUCCESS in 19m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b75467cc9505445bb73d94d0465e7d9b : SUCCESS in 22m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6221f3866ad249f683e049aeecc37ad5 : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f928f51a4a5e4e35b1345cc52d82af14 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/80ca037d87044496a93fd3a5dbfbf8d1 : SUCCESS in 23m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f1c8e20e7fd644b3a05237680c2285d6 : SUCCESS in 12m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/de3143d1e7d442968dfc8429875c5ed3 : SUCCESS in 21m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc61b99a1c1d4c14bfd2b06c6c5e4505 : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/90cb739c8edc4075908344e38a0c7c45 : SUCCESS in 4m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/58df0bdb1e1e495a8f020664039c6b2a : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/398e99f719e346c3b9d044aeabaec1a6 : SUCCESS in 52m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51aa7a4c8c364b86b6c7e4e1b3b46bba : SUCCESS in 53m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5efc6c75fcc463b8c209ddc2d48cd15 : SUCCESS in 54m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6762c7a9186344c19271a6fa975b9eda : SUCCESS in 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92faa34bfcee4d0f8b1be42eefc40333 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/428db3c7adb14d08a57bc9d48b1bde4a : SUCCESS in 19m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a954bd9a3df480cacaaa87ee0aa68df : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":3},{"id":"5257b010f2d064388e8a14f7e37e186552118466","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 10:23:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"11ae3444cf7e84b422c7fd65652ac3cada6f3f95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 10:24:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7db8df85b3ae4e4194b2537fda9ef506\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ab4c77b7414000b5db78ba61318f2a : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e5d273396d804d34bd4742168ef7f057 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5e079635483baa34d781bfcf78129a7057e627e4","revisions":{"9209ba79025ca46e6eeb709aa31386f199020684":{"kind":"REWORK","_number":1,"created":"2024-08-28 18:48:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/927425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/927425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/927425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/927425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/927425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/927425/1"}}},"commit":{"parents":[{"commit":"6a0153f54542da474955802ac01cf174c7ce8573","subject":"Merge \"docs: Simplify reload-process-tree SVGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a0153f54542da474955802ac01cf174c7ce8573"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-28 18:46:28.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-28 18:48:15.000000000","tz":-420},"subject":"diskfile: Treat EUCLEAN like ENODATA","message":"diskfile: Treat EUCLEAN like ENODATA\n\nFound a new way filesystems can break in prod:\n\n   object-auditor: ERROR Trying to audit .../7a7c4af06d2616f23eda274e2ad9c948:\n   Traceback (most recent call last):\n     File \".../swift/obj/diskfile.py\", line 2630, in open\n       files \u003d os.listdir(self._datadir)\n   OSError: [Errno 117] Structure needs cleaning: .../7a7c4af06d2616f23eda274e2ad9c948\n\n   During handling of the above exception, another exception occurred:\n\n   Traceback (most recent call last):\n     File \".../swift/obj/auditor.py\", line 238, in failsafe_object_audit\n       self.object_audit(location)\n     File \".../swift/obj/auditor.py\", line 261, in object_audit\n       with df.open(modernize\u003dTrue):\n     File \".../swift/obj/diskfile.py\", line 2657, in open\n       \"Error listing directory %s: %s\" % (self._datadir, err))\n   swift.common.exceptions.DiskFileError:\n   Error listing directory .../7a7c4af06d2616f23eda274e2ad9c948:\n   [Errno 117] Structure needs cleaning: \u0027.../7a7c4af06d2616f23eda274e2ad9c948\u0027\n\nChange-Id: If6731a6b6b16fbc4eebb61254ed10b53e1767a0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9209ba79025ca46e6eeb709aa31386f199020684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9209ba79025ca46e6eeb709aa31386f199020684"}]},"branch":"refs/heads/master"},"0bde02013c0141171562d8b7f13f1b2a058d4166":{"kind":"REWORK","_number":2,"created":"2024-08-28 22:38:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/927425/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/927425/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/927425/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/927425/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/927425/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/927425/2"}}},"commit":{"parents":[{"commit":"6a0153f54542da474955802ac01cf174c7ce8573","subject":"Merge \"docs: Simplify reload-process-tree SVGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a0153f54542da474955802ac01cf174c7ce8573"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-28 18:46:28.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-28 22:38:28.000000000","tz":-420},"subject":"diskfile: Treat EUCLEAN like ENODATA","message":"diskfile: Treat EUCLEAN like ENODATA\n\nFound a new way filesystems can break in prod:\n\n   object-auditor: ERROR Trying to audit .../7a7c4af06d2616f23eda274e2ad9c948:\n   Traceback (most recent call last):\n     File \".../swift/obj/diskfile.py\", line 2630, in open\n       files \u003d os.listdir(self._datadir)\n   OSError: [Errno 117] Structure needs cleaning: .../7a7c4af06d2616f23eda274e2ad9c948\n\n   During handling of the above exception, another exception occurred:\n\n   Traceback (most recent call last):\n     File \".../swift/obj/auditor.py\", line 238, in failsafe_object_audit\n       self.object_audit(location)\n     File \".../swift/obj/auditor.py\", line 261, in object_audit\n       with df.open(modernize\u003dTrue):\n     File \".../swift/obj/diskfile.py\", line 2657, in open\n       \"Error listing directory %s: %s\" % (self._datadir, err))\n   swift.common.exceptions.DiskFileError:\n   Error listing directory .../7a7c4af06d2616f23eda274e2ad9c948:\n   [Errno 117] Structure needs cleaning: \u0027.../7a7c4af06d2616f23eda274e2ad9c948\u0027\n\nChange-Id: If6731a6b6b16fbc4eebb61254ed10b53e1767a0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bde02013c0141171562d8b7f13f1b2a058d4166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bde02013c0141171562d8b7f13f1b2a058d4166"}]},"branch":"refs/heads/master"},"5e079635483baa34d781bfcf78129a7057e627e4":{"kind":"REWORK","_number":3,"created":"2024-08-28 22:58:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/927425/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/927425/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/927425/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/927425/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/927425/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/927425/3"}}},"commit":{"parents":[{"commit":"6a0153f54542da474955802ac01cf174c7ce8573","subject":"Merge \"docs: Simplify reload-process-tree SVGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a0153f54542da474955802ac01cf174c7ce8573"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-28 18:46:28.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-28 22:58:03.000000000","tz":-420},"subject":"diskfile: Treat EUCLEAN like ENODATA","message":"diskfile: Treat EUCLEAN like ENODATA\n\nFound a new way filesystems can break in prod:\n\n   object-auditor: ERROR Trying to audit .../7a7c4af06d2616f23eda274e2ad9c948:\n   Traceback (most recent call last):\n     File \".../swift/obj/diskfile.py\", line 2630, in open\n       files \u003d os.listdir(self._datadir)\n   OSError: [Errno 117] Structure needs cleaning: .../7a7c4af06d2616f23eda274e2ad9c948\n\n   During handling of the above exception, another exception occurred:\n\n   Traceback (most recent call last):\n     File \".../swift/obj/auditor.py\", line 238, in failsafe_object_audit\n       self.object_audit(location)\n     File \".../swift/obj/auditor.py\", line 261, in object_audit\n       with df.open(modernize\u003dTrue):\n     File \".../swift/obj/diskfile.py\", line 2657, in open\n       \"Error listing directory %s: %s\" % (self._datadir, err))\n   swift.common.exceptions.DiskFileError:\n   Error listing directory .../7a7c4af06d2616f23eda274e2ad9c948:\n   [Errno 117] Structure needs cleaning: \u0027.../7a7c4af06d2616f23eda274e2ad9c948\u0027\n\nChange-Id: If6731a6b6b16fbc4eebb61254ed10b53e1767a0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e079635483baa34d781bfcf78129a7057e627e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e079635483baa34d781bfcf78129a7057e627e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
