)]}'
{"id":"openstack%2Fswift~927630","triplet_id":"openstack%2Fswift~master~I45113d38c9f3d523fb6f3a02adc0f3856da6aaaf","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-08-30 20:23:57.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"37271":{"account":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"last_update":"2024-10-03 14:12:46.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-03 14:12:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I45113d38c9f3d523fb6f3a02adc0f3856da6aaaf","subject":"Add Chinemerem to AUTHORS","status":"MERGED","created":"2024-08-30 17:39:20.000000000","updated":"2024-10-03 14:13:57.000000000","submitted":"2024-10-03 14:12:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927630","meta_rev_id":"9d2235064dcc74b024a967ab8d4ce3903b3015ce","_number":927630,"virtual_id_number":927630,"owner":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 14:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"date":"2024-08-30 20:23:57.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2024-10-03 13:05:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2024-10-03 13:05:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-30 17:43:37.000000000","updated_by":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-08-30 18:07:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-30 18:18:32.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-10-03 13:05:48.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"4bebcd7bf3bcecd4b5b860da77aa528bb020981f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-08-30 17:39:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d68f62c8d636b41453a44b84ee16641464a0d394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:07:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc91c794929498da55fd0dbc0b739e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/834b38a1b79d427cb205c6352555d82f : SUCCESS in 8m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef172620c3c1405bb5b4242f55bcf211 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d7dd3b8f8177411a90efa97e81c1c4ed : SUCCESS in 19m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fde772fb063344419e0c6dd5727b0c32 : SUCCESS in 19m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e34bc3f35f5b43408dfcc70087315328 : SUCCESS in 16m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7f44dab5df1549c8a48282049a13ad29 : SUCCESS in 21m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0e056f17e0e94f49ba71a9c2ce6cc450 : SUCCESS in 26m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13e00c498518498fa40d63f011a3fe3e : SUCCESS in 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81f3d17a51e84e0e8edb99ec07b051ca : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"69e0faae565b1a4dac886ff6baffe675fd445da6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-08-30 18:18:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ecf9e5e2c55dc4e7f12f9ece1415fca6376017a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-30 19:20:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36606a5db35596461d81fef1831f80ffcb9b3c32","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-08-30 20:23:57.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fca1d0b1c1ac092146387896b87a69c58e183cdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 20:43:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccbee6d40fa44a03a882f267e0e0dca0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6289c4fafb15471a962bd51092a70261 : SUCCESS in 50m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/781bb3f97a6c4b78931e3e63b1bc1a99 : SUCCESS in 1h 10m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/131743ae46bd48bc98e69ec65ec9e2ba : TIMED_OUT in 1h 30m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f65493b2728148778671d9965e45e61a : TIMED_OUT in 1h 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"39470a2eecf2d285d3f2debcbf41e5080a4a7010","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-03 13:05:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf704555ffd7f5ff31c07bb539706df1a0439063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:06:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7f792de40980163d7a463674a2dfe3be4788ed8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:12:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf4a33bde1e54559a139acebb03a6fb8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be818d2f8ad14c5f9b12bbb5a1cd86b6 : SUCCESS in 10m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15e8a45c61634ccf8cce18e14035d7cc : SUCCESS in 1h 06m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fdc77623534f45a38480c329ed4e1d44 : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a3d3ef8dc6844d3aa711d27fa99d459e : SUCCESS in 18m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/deea59ff473a44628a0fcc52e07bc9e1 : SUCCESS in 10m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e70d9b8924bd412cacc34fdef7af5ae7 : SUCCESS in 21m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1570dacc9daa4e4d901b44157d2046bd : SUCCESS in 10m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/50041f1c40ab45878af1e153258238a1 : SUCCESS in 29m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ead7708732b4a71b3c9b693396d5b5e : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e79b5121cd674b509049574a172ca23e : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd1bd7cc718b4fea8d25d80181d5473b : SUCCESS in 18m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a211a86426e846ef99bd64d6a5adcc0e : SUCCESS in 8m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1eda9f9190f445f4a8878e4b80393230 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e823ed6f72c64cbdbb007d934c978d15 : SUCCESS in 9m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd83d8ae5bb340149cbe3f1f03e5cbed : SUCCESS in 4m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1130b4e85dc74a29aa591bf9d02a7e46 : SUCCESS in 6m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1760bf9738a24cd3ae2fd3eb855fb8c3 : SUCCESS in 22m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d869a8fd7d5a4dfbb6bacd0dd8f0c4ac : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e0eaf4ee57a047e5043eae78b84e54f77b35f806","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:12:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9d2235064dcc74b024a967ab8d4ce3903b3015ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:13:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32930f32a50847fbbdaa795bbd7dc18c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d960cc68c9ac4fd5b87c715d21ec9950 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/28fed989a30a460984b4e067b6d28a9e : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f277b2e9b1c7e03e0e0129363e08c2b28cf8a6d5","revisions":{"f277b2e9b1c7e03e0e0129363e08c2b28cf8a6d5":{"kind":"REWORK","_number":1,"created":"2024-08-30 17:39:20.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/30/927630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/927630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/927630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/927630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/927630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/927630/1"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Chinemerem","email":"chinemeremchigbo@Outlook.com","date":"2024-08-30 17:33:26.000000000","tz":-420},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-08-30 17:38:33.000000000","tz":-420},"subject":"Add Chinemerem to AUTHORS","message":"Add Chinemerem to AUTHORS\n\nChange-Id: I45113d38c9f3d523fb6f3a02adc0f3856da6aaaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f277b2e9b1c7e03e0e0129363e08c2b28cf8a6d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f277b2e9b1c7e03e0e0129363e08c2b28cf8a6d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
