)]}'
{"id":"openstack%2Fswift~927822","triplet_id":"openstack%2Fswift~master~Id1a4c0b759cd6097c5c35683aa77f6477726786a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-03 19:57:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id1a4c0b759cd6097c5c35683aa77f6477726786a","subject":"sq? refactor and more tests for s3 v4 sha256 checks","status":"ABANDONED","created":"2024-09-03 13:47:13.000000000","updated":"2024-09-03 19:57:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"85930fde1f4a509aa43ad285b53d33659431a1cd","_number":927822,"virtual_id_number":927822,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-03 15:29:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 15:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-03 15:29:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"603282fd127d2a4b1fe2ad1cfc0c0c8e45ebe1ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-03 13:47:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0ec687178a4ce4ec5784d62ca9b21e2a8497fb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:12:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f86c62e9ea4b29a98c60d795aa17b2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fb3a8792aac4d3fb558e5ee16e4fbe9 : SUCCESS in 1h 02m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3af7f9344434a22ba062f129da6feca : SUCCESS in 39m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/91362d63598c48f18046d24f541a71dd : SUCCESS in 57m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d99c9c87846f456c9445ea66a9834f17 : FAILURE in 1h 22m 56s","accounts_in_message":[],"_revision_number":1},{"id":"85741dde971bd15734cd556cabb40e8bb5713752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:29:28.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10686a7fdd6447be98a39d02ade0ee5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dfa992f9dfc499fa3cd24af20801e42 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/555d87ae59c14bf591cd68981e9fdcf3 : SUCCESS in 1h 10m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4dd9bd0c24ea4babaeba0fb9670ebece : SUCCESS in 50m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cadbd0d39d9343569c4404a3224d247c : SUCCESS in 53m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/864ad2e6ec1a442983a0529e2f5f727d : SUCCESS in 27m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ab816b8bca6422495c55d21c5dce902 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abce1045cad44512ab100f38e3306eca : SUCCESS in 18m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12980712080546cf9a9768020f947c4c : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/10e39563765e4a698c4091532b41cd4d : SUCCESS in 19m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8ce88d2c1824f0885ac6d6994ca099b : SUCCESS in 20m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cbb5d0b6cc6e4b64aed5ebf97dd1dad7 : SUCCESS in 27m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d24fb26c9d8d4a78a77f928c84d11878 : SUCCESS in 33m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b7f2dd24e981408d90d48f84684d7a20 : SUCCESS in 27m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5f6cfba110d14678b3566a54f8b9b27f : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e5019d64bb914d6aac7c6d9adc09cdcc : SUCCESS in 22m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6fec799fb1a244f7a61845014ffe46a8 : SUCCESS in 20m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/21439a6bf46e4b52a037c20fce5c111d : SUCCESS in 21m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5ba42d6635bb4168830faf3c75435a72 : SUCCESS in 19m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cff4a7b04cd74bb5bc13540fbeb279a2 : SUCCESS in 22m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/629c4de6de644ce385bcabe1ff564bd7 : SUCCESS in 8m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0160338df6314298b99dd5e3444ae18f : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c0129e307d7457e910b78214b98c3b2 : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9763466b4275418e8a61fda59e05e5a1 : SUCCESS in 50m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/662521a3c18f4cf9b10783ab533a120c : SUCCESS in 47m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dbd34a989afa45f292ffc818b37ae57f : SUCCESS in 1h 00m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8032db70659417c9b98821f3b7a65d8 : SUCCESS in 23m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e34828b345af46c7b6b17f894eb25c00 : FAILURE in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b40128721ca46dfbc26939161144894 : SUCCESS in 18m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35eabd2283c64a4fad09095a261c32f1 : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":1},{"id":"85930fde1f4a509aa43ad285b53d33659431a1cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-03 19:57:27.000000000","message":"Abandoned\n\nSquashed into parent. Thanks again!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65dbc80775ab94aacbc134232520166ca85dccc3","revisions":{"65dbc80775ab94aacbc134232520166ca85dccc3":{"kind":"REWORK","_number":1,"created":"2024-09-03 13:47:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/927822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/927822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/927822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/927822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/927822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/927822/1"}}},"commit":{"parents":[{"commit":"70948481f92a26466c23fa2d2cc072b77c6e0e67","subject":"s3api: Clean up some a bunch of errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70948481f92a26466c23fa2d2cc072b77c6e0e67"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-03 13:46:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-03 13:46:28.000000000","tz":60},"subject":"sq? refactor and more tests for s3 v4 sha256 checks","message":"sq? refactor and more tests for s3 v4 sha256 checks\n\nChange-Id: Id1a4c0b759cd6097c5c35683aa77f6477726786a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65dbc80775ab94aacbc134232520166ca85dccc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65dbc80775ab94aacbc134232520166ca85dccc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
