)]}'
{"id":"openstack%2Fswift~927857","triplet_id":"openstack%2Fswift~master~Iae48e8db642f6772114c0ae7c6bdd9c653cd035b","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-01-08 19:35:54.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2025-01-08 20:59:29.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-01-08 20:59:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae48e8db642f6772114c0ae7c6bdd9c653cd035b","subject":"obj: Add option to tune down etag validation in object-server","status":"MERGED","created":"2024-09-03 17:50:30.000000000","updated":"2025-01-08 21:00:37.000000000","submitted":"2025-01-08 20:59:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927857","meta_rev_id":"4c99a9e56c4b90c6d7a9dbe15f793fd45dcca58e","_number":927857,"virtual_id_number":927857,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-08 20:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":2,"date":"2025-01-08 19:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-01-08 19:35:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"CC":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 18:50:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-04 17:07:23.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-09-04 17:10:58.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2024-10-22 23:05:23.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"}],"messages":[{"id":"eb2929b27e389528c9389058abe54388ed88f95b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-03 17:50:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"060cef197e12339bd032c5db509ccfece4db28b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 18:50:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a90edb4f027b4ddea529793b30a93f5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8a6d3b5d9144f0a961390b3e7b3670 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/e28e156ec77149f2a138d22823eadc66 : SUCCESS in 53m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/447b2d846994471782d3a42feca77a4c : SUCCESS in 38m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15ce90ec67e54bcca81e8e15f830fdd4 : SUCCESS in 43m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e93d17adffce4c34a1c9b1f854198729 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b245fc97904d4366a853345202824e0a : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ceb1f5c2e3904c8db1b379059faa374a : SUCCESS in 18m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5013c88478e84799900489b274b792c7 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80a8e410995249c0b1ab5152835778cf : FAILURE in 14m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a212bd5a8c2f492c860ad238d20faab4 : FAILURE in 14m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5d8820f3b52048628e29362b496df93d : FAILURE in 14m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ba0b40f6bf51466c8567d6f37902db60 : FAILURE in 17m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cea1d6d268a346a9a80fd9118b1cca69 : FAILURE in 17m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/508ca2cd6b614b2e8744a0fb5a8241ae : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1aa85cb2cca04361bdd161e29d3c7656 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a11e877651964bc584d18c51a32aa855 : SUCCESS in 17m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb3bbe8b7291424dabeebfdf2926e655 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/00bbc18809364c3687c7491a23ad9d73 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7d4c5f3f008f4f2cbfd171b62f0198db : SUCCESS in 20m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c47eb201d3a48fe990ac6fc8f2a7822 : SUCCESS in 5m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/898b9fe73cae4e4aa059f246edddf7c8 : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/05128bf6896349e8a2becd6a315b21ff : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b494af7d7da34bc087fba9be726b5521 : SUCCESS in 50m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/87877d44354942c6895da603a4c45810 : SUCCESS in 49m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f05b651e0ac4ae6989f51942dc786eb : SUCCESS in 48m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4a7cad1f32d4dd4ae6d28c25eaad6fd : FAILURE in 18m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89986a04d68144b59110aee3ec948211 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b8d013fdf10d4f83885f94e78355f28b : SUCCESS in 17m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcbaea832922448b82ab76dc94b033c0 : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ad7a0872c898e7f1b03e7605c7f8b6fe2dfd64bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 19:24:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/336650f9180d457284a64054938a5ddd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8c35bc4e4534dde80e2cba710d165b6 : TIMED_OUT in 1h 31m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5dd75056c05349b7b54181b64d06516b : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9445770c665c493a9066e0c1bc7a3bf1 : TIMED_OUT in 1h 31m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/12600dd62acc4177b879174969fb8fba : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"db4b6632d3a68ecb691ac3a56df91de31654fea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-03 20:21:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"92a30c5a3b546715faaba8e338aa24dac0787c5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 21:30:44.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf7f6f5c3f094d318cb31039df6ac7ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/272d4ee446854b809924bd99c4b9fc6e : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/0fce4915893e47048bb1868d10bf7837 : SUCCESS in 57m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a3dfc658ee19443e97d4e69ea02afbc8 : SUCCESS in 47m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e8a2aa0f2b74c5c9e7d7e69424f57a3 : SUCCESS in 57m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6784f3b82d041e2869c68227763b87b : SUCCESS in 23m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/650f6c2efc4b4eefbc2ccc8917f6fe71 : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3edeb72145524ebbb79069e852a0bb8e : SUCCESS in 23m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/537da9f1bbc94198bc5357752f45e9ca : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d8fabeb691b24e11af1b5dd742ab349f : SUCCESS in 16m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8da3011717154576ae41036d99ee906a : SUCCESS in 20m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/674fc3923e7c4e518f96be1cdd4d8dd7 : SUCCESS in 21m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0de75fbebcff4d639f1ebc66f2cdaae1 : SUCCESS in 18m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4b27fcc53dfd4e6ab322807957168f8b : SUCCESS in 18m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32d3f09cb8de41a3b9b5dbf5c1224b26 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20a39bd94ab347e6a061e69cfebec84e : SUCCESS in 20m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6f7bd9ee60b540058c01058cf6311456 : SUCCESS in 17m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ae8d4a62cef8433997f1d66d0170ebe2 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f12159ddd851413e99a9faf1d4a2c6a2 : SUCCESS in 20m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b64416d436aa40a8809e22da36d42d20 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fa84c63a693347ea86b05c9eea95a190 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8de54dbf4ee641c6a69d825105c6739e : SUCCESS in 13m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9453d426c7af4bcf974f677f1b925b89 : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/55cc2c498e784dc194c197474263f842 : SUCCESS in 47m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b136205274124377998d329305a59896 : SUCCESS in 56m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82e263e1eaf040de9f4fbcb214c419c0 : SUCCESS in 46m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2f5630fa10b4e57b063294cac4b2162 : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9b81c2ca284cdebaae76d0fbadf683 : FAILURE in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0974229482af4825aa84859a59c5cb0a : SUCCESS in 17m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5499c16daa2844d7985389e2aab0e556 : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2fcd0e1740a2142108871c3bd6c24fed06cd72d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 22:31:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffd094b2810f42d79a6c3b3c5458ddf2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68ff6f05946d4c4a90f3f20c15209be9 : SUCCESS in 1h 25m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/540c837174774102b2326ed69e1f5e25 : SUCCESS in 1h 20m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/da2ae427f30e409e8040eef3d810db6b : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/827fe3ea78ac46299ae5a021e49f9697 : TIMED_OUT in 1h 31m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e4748d0931e0bf582f8c8b1b39f034455b6e3ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-04 00:29:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e692e3dee8d9e7fa0b205379eae7b353a576a2a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 01:37:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a352697240453fafbebb8e84520e74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e4cb6335c749b99a7e66457e697ca8 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/9a039c5ae9f34eeba9a6fcc6e5f7bef0 : SUCCESS in 57m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/37cb82222079451f826f16aacb67fbc5 : SUCCESS in 36m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9c8455c3021484eb8d6b5e8f8d5cd76 : SUCCESS in 59m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95551612f39e46b39bc2ec9f77a39b77 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8fe05dfe5a784eabb4b8ce413f56c65d : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/281ffc1c73b544f9a09ed96db9a5628c : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bad415d11f4848b6a46f801de44801c1 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5712b9a165614c58bb081f03dfaaf484 : SUCCESS in 18m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a31bcef735b4454ab9b520aac9ebd69e : SUCCESS in 18m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b79ecfe2fcf04db5850afd4c29718fbc : SUCCESS in 14m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7ef45b0793be486a802a85f761be5b9c : SUCCESS in 21m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b4f5257c8b454f8d9e8896b9611f55b9 : SUCCESS in 25m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3eec10b09fbc4bd9b21d2f6e3e984957 : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0681c15997254a1b82a9d7d7b199de49 : SUCCESS in 18m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2c0ed38566ca4087b98629a1936bf1ff : SUCCESS in 19m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1582e0739b0a460e9ddc953648b749b0 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/30321dde03b0456a8083c303647db490 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6cd611f3bee74080902c1a8786044028 : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4af2f8c32da34bd0a27600407204a138 : SUCCESS in 8m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/079ffb5a61e745888879efe3c7da6fc9 : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d2b9c28d1ec64e5da51d74eb8083c029 : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/240680813a5f4ed0969f8616c95650b7 : SUCCESS in 45m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bb5e88c460d34bed8f769c54fd8c0b17 : SUCCESS in 52m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8f690b53a3ba4b25aa8de0ecf2e830c6 : SUCCESS in 49m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6acb409004d4937b9b7a9f408b0d3be : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86dcf461f1824672971995ea96d19d02 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c01fd18ca1fb478d9167f843073ac982 : SUCCESS in 22m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/438beeed8cfb4183940eadba05733ced : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":3},{"id":"67ded7262a2f694420be4db5066de7f16c1a74d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 04:51:46.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e308788fd142efb3a9ef318fe70523\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e361c5cfc214c5e8fd7be05c41ab02b : SUCCESS in 1h 26m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/39845c18c03440a988420d7c8ee8ac83 : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1ff883ad66e461ebd60c36fc5a32519 : SUCCESS in 1h 06m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce5822a01f9d46e7811ed10c3cdcc511 : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":3},{"id":"cb52b92714fd5ae23772a5027d03af2ca951a571","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-09-19 22:49:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"85ba04cd95d3c3c376fbdccfa1de2de350e99036","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-07 23:22:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"85a5ad07b0928c6862f2281780b17edafc29f05a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-07 23:22:38.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dfb81ae0e388269a76111f431eb2c0058fa61cfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 00:21:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42bcc1b41eea4b22a7ecb30304b4a082\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be0d69b03864d6cb9bc41deef288a39 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/adc85e4c2b174812a0043acc491b3b28 : SUCCESS in 52m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c9e7e17199714a01ad5ab5591b944f26 : SUCCESS in 50m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/72b4105f10e549a2933f80a292383a05 : SUCCESS in 43m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14ed70c618964bc5979a490a2379182b : SUCCESS in 38m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c3ae7f2622b4910afacc28191ba320d : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43b4d7cc2ded4d3a91b4000feb1edeee : SUCCESS in 10m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9394c7342564df8bf9a092720c16971 : SUCCESS in 17m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2ae821a8b40e4b5ba09adf04081cb9fc : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/57e5867bbcff462898c4409ae57c4312 : SUCCESS in 10m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bfd1efb0e728473e9ee84bbd134784d0 : SUCCESS in 18m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a08d68dd2a074391ac9b9a6315e9847a : SUCCESS in 10m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a9314ac4e06741f6a3949054da757bfd : SUCCESS in 22m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8244c3f99cd748bb8843310c45b54e8c : SUCCESS in 29m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1367b747b2844aa7b9d2ed3886946b93 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bae97f615c984c84ba28ed6e6c444a95 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/daf4b04d9913466a87713503919e936a : SUCCESS in 19m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/608c988da56e4e90834cb019318087a6 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e40cb9f61bf490cb0b0004e3a3f2c81 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aa2a9bec7983439a82b8b0957834f43a : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bece03cd542a4b078d58a5d24efd194f : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aa9c5832245142a38bf9c2a343b6d4af : SUCCESS in 12m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/83bb97db35eb43d48912f4050b4111af : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/25c95b3b4b7e40978470f68e34470673 : SUCCESS in 41m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1af13abe3514df5bb53c48a3c925702 : SUCCESS in 52m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf00ee20d5ea41e38dcebd5a52803c98 : SUCCESS in 33m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c2120d339804c8f8612dff64ad27fed : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c28a457e08f4ff9b88e06d19b9b86ea : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a4bb8c6659347e2bba25b86493e6f36 : SUCCESS in 14m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a36fe3eb92cd45b2bde7cf3aa2f5bf79 : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":4},{"id":"f0b8828782d4959581da46a0f706c8dd7455ac7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 00:29:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/716cd517a18945698cec3bd546bf59c5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9845d09e64364f7c9d0b9b0fe5ee75ed : SUCCESS in 46m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c2600b13df742ecbe8b639e7b960ae2 : SUCCESS in 1h 05m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/705358b67abb4547a5efc1adbe4f3ec8 : SUCCESS in 40m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/afa972c49409409896c71f9d28c9c4c4 : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":4},{"id":"e5d6a03cff1119ddc2e9a4ec0805ae33da9e1e92","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-17 15:58:15.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f41290addeed8439c1fad85d51fa9bb94834aee3","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2024-10-22 23:05:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d96d49b5a4ef5d399c24e6f4dff387efcc18c73b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-12 17:41:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2974ad9f2bb05972b544f13e56482b64c40e6e2b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-12-13 00:08:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a7e7a1db7e1ad63a97ed3f91e5861f1cafa6fa6c","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-12-17 04:54:41.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"42ea92f040283a4ec3302a6f75913abfbe2b4239","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-08 16:59:35.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0c4299695a834a56f83cfd9eb83ab03f0a01c845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-08 18:21:31.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9ea493a38e89b4b77ebdca836ca17402d8d8b5d9","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-08 19:35:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c3db1f0ee81f8e143c1c0ab0766e9ba920b279da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 19:49:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0025b1bbb1b643d8b3d77eb09fc06a8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb8525cfbb14ad7b71dcd9a8aabff32 : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/dd85127939e64b598c326f71bb2d3e27 : SUCCESS in 59m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/26c95814a4dd471f9cbcc32b583b9c4c : SUCCESS in 59m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b3a65e7f0ab7480c93bcf4326cd92e7f : SUCCESS in 46m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1972cc0b3514a32934bb78afa47c387 : SUCCESS in 59m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/459d1073a65d4c258b0e8ddcef2958ce : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb31d8558e194a98a0186f2d20e5a389 : SUCCESS in 22m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42a5a1d8de5c40a292e136c5fcaba078 : SUCCESS in 17m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cbc9db09ac71490590739d0ed0ef67c1 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d5a991ad8acb45f0959ba29b8c2bf399 : SUCCESS in 17m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a90821a730c24723942761229e20a513 : SUCCESS in 24m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3934cc9c8223408ba49da78ffd34e010 : SUCCESS in 28m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d59707632b77433d8c5a64fcdf03766c : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b41608458e3d4224ba3fd537a44337c9 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7519a3569b0349c198bd14d30705fc86 : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4f79bcb1827f41698712aa753b297447 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/333f277850a24e51ba8f4b0e83245ce3 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eeb83cedb9f7406aaef216e8bc609a78 : SUCCESS in 7m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/30852585fc4949bc92fadcf03351b524 : SUCCESS in 55m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/916b5fe32f81497d8877f9b8faea2499 : SUCCESS in 47m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d81642bbcb16427c9b22524e7884dbc9 : SUCCESS in 1h 07m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb9c4c1b976f4963904fc82532a72ccf : SUCCESS in 23m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87f82df04c134ce5b34d3018e9082fb3 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa82629a02f44db4ad5da43342e31940 : SUCCESS in 20m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46d8172e27264498b30e7eb7c993357b : SUCCESS in 36m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e400098953044f83a2fef69d4a600921 : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":5},{"id":"f4cb6ebf667ad7e23b2d90e40e09d8868dcdfd8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 19:49:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cf7bf4d01bbed61612b166c1cfecc327d0bb2c62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 20:55:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e1c8c52bae34fc8836c7c3470ee8298\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99f3d104c3ec4f6583918124caa70e56 : SUCCESS in 2h 08m 10s","accounts_in_message":[],"_revision_number":5},{"id":"c9b6a20f95a09c6be5172cd8e7387970651d44d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 20:59:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c28742a8e834b5baed7f7d3d8a9fc13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2f7844018f407fb5e3678258b8c150 : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/1075ae3053914a5593cb54b3139e4620 : SUCCESS in 1h 02m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ef7168eb8b4448f9036bf7bad2f4191 : SUCCESS in 1h 03m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7977d63a732344dca55d16037b22b76b : SUCCESS in 53m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82d04dfcd7f64b988246d7f80ac63251 : SUCCESS in 1h 06m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e4a6604df3ae491abec59cf5b46c2b77 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18a5005917404bcaa0c2c3c2390d5c39 : SUCCESS in 19m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/147287973fbe4d6d8e6831d15705eccf : SUCCESS in 22m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/86b80f39b8b94036afe9c807e4d2207e : SUCCESS in 28m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7e1759f7c2b44e2f8f98adbb0fb6addf : SUCCESS in 12m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/97c41ec809924673b6a6ebfd965e1432 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7fc35324f4104b9aa0ed2fd35bb3f80f : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b4fc26a96c34a12960c9601570be70c : SUCCESS in 4m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c95e0a1d22e24e9089f73b73f23872a1 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ef6905c72e84f24bfc801063b62b4de : SUCCESS in 53m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23eeaa8ea3f946d3a73b1379c3d7f221 : SUCCESS in 49m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34c7932d51604ae79bbf105547d9186d : SUCCESS in 33m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12b06768941a416f95b24b36babdebcb : SUCCESS in 22m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06387190b65d468793a7545c1a4647b9 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/70792558c63040bfae8202cf223d4ddf : SUCCESS in 15m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4507afdcb2fb4b42bd70273ba294288f : SUCCESS in 34m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9beafad9cca24ff5a94406a761cda14e : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":5},{"id":"f267cc2ede6025037c8c95577f7e6d21940c14d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 20:59:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"4c99a9e56c4b90c6d7a9dbe15f793fd45dcca58e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 21:00:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40cda86f9f4b473290a3baa86a8a8c24\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95efe77447f64a76bf155baabfaed98a : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d1e8833e024c469e9232d4e2a035a393 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3d8fb046cbc439eb5e0b18d97b874c7f7b9ca489","revisions":{"7df86909f33c2ce8f2195e3cadec74c06988fd8e":{"kind":"REWORK","_number":1,"created":"2024-09-03 17:50:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/927857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/927857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/927857/1"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-30 21:03:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-03 17:49:58.000000000","tz":-420},"subject":"obj: Add option to tune down etag validation in object-server","message":"obj: Add option to tune down etag validation in object-server\n\nHistorically, the object-server would validate the ETag of an object\nwhenever it was streaming the complete object. This minimizes the\npossibility of returning corrupted data to clients, but\n\n- Clients that only ever make ranged requests get no benefit and\n- MD5 can be rather CPU-intensive; this is especially noticeable\n  in all-flash clusters/policies where Swift is not disk-constrained.\n\nAdd a new `etag_validate_pct` option to tune down this validation.\nThis takes values from 100 (default; all whole-object downloads are\nvalidated) down to 0 (none are).\n\nNote that even with etag validation turned off, the object-auditor\nshould eventually detect and quarantine corrupted objects.\n\nChange-Id: Iae48e8db642f6772114c0ae7c6bdd9c653cd035b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7df86909f33c2ce8f2195e3cadec74c06988fd8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7df86909f33c2ce8f2195e3cadec74c06988fd8e"}]},"branch":"refs/heads/master"},"99fa0bbcd9a8c02a5a53b1d6369580622a764b36":{"kind":"REWORK","_number":2,"created":"2024-09-03 20:21:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/927857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/927857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/927857/2"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-30 21:03:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-03 20:21:00.000000000","tz":-420},"subject":"obj: Add option to tune down etag validation in object-server","message":"obj: Add option to tune down etag validation in object-server\n\nHistorically, the object-server would validate the ETag of an object\nwhenever it was streaming the complete object. This minimizes the\npossibility of returning corrupted data to clients, but\n\n- Clients that only ever make ranged requests get no benefit and\n- MD5 can be rather CPU-intensive; this is especially noticeable\n  in all-flash clusters/policies where Swift is not disk-constrained.\n\nAdd a new `etag_validate_pct` option to tune down this validation.\nThis takes values from 100 (default; all whole-object downloads are\nvalidated) down to 0 (none are).\n\nNote that even with etag validation turned off, the object-auditor\nshould eventually detect and quarantine corrupted objects.\n\nChange-Id: Iae48e8db642f6772114c0ae7c6bdd9c653cd035b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99fa0bbcd9a8c02a5a53b1d6369580622a764b36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99fa0bbcd9a8c02a5a53b1d6369580622a764b36"}]},"branch":"refs/heads/master"},"64d216607f9fb948613ce1d74ecf17ec472c721b":{"kind":"REWORK","_number":3,"created":"2024-09-04 00:29:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/927857/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/927857/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/927857/3"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-30 21:03:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-04 00:29:15.000000000","tz":-420},"subject":"obj: Add option to tune down etag validation in object-server","message":"obj: Add option to tune down etag validation in object-server\n\nHistorically, the object-server would validate the ETag of an object\nwhenever it was streaming the complete object. This minimizes the\npossibility of returning corrupted data to clients, but\n\n- Clients that only ever make ranged requests get no benefit and\n- MD5 can be rather CPU-intensive; this is especially noticeable\n  in all-flash clusters/policies where Swift is not disk-constrained.\n\nAdd a new `etag_validate_pct` option to tune down this validation.\nThis takes values from 100 (default; all whole-object downloads are\nvalidated) down to 0 (none are).\n\nNote that even with etag validation turned off, the object-auditor\nshould eventually detect and quarantine corrupted objects.\n\nChange-Id: Iae48e8db642f6772114c0ae7c6bdd9c653cd035b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64d216607f9fb948613ce1d74ecf17ec472c721b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64d216607f9fb948613ce1d74ecf17ec472c721b"}]},"branch":"refs/heads/master"},"845394b4d2a4ac8db63733de90ff15979a1c30e7":{"kind":"REWORK","_number":4,"created":"2024-10-07 23:22:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/927857/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/927857/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/927857/4"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-30 21:03:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-07 23:18:57.000000000","tz":-420},"subject":"obj: Add option to tune down etag validation in object-server","message":"obj: Add option to tune down etag validation in object-server\n\nHistorically, the object-server would validate the ETag of an object\nwhenever it was streaming the complete object. This minimizes the\npossibility of returning corrupted data to clients, but\n\n- Clients that only ever make ranged requests get no benefit and\n- MD5 can be rather CPU-intensive; this is especially noticeable\n  in all-flash clusters/policies where Swift is not disk-constrained.\n\nAdd a new `etag_validate_pct` option to tune down this validation.\nThis takes values from 100 (default; all whole-object downloads are\nvalidated) down to 0 (none are).\n\nNote that even with etag validation turned off, the object-auditor\nshould eventually detect and quarantine corrupted objects. However,\ntransient read errors may cause clients to download corrupted data.\n\nChange-Id: Iae48e8db642f6772114c0ae7c6bdd9c653cd035b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/845394b4d2a4ac8db63733de90ff15979a1c30e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/845394b4d2a4ac8db63733de90ff15979a1c30e7"}]},"branch":"refs/heads/master"},"3d8fb046cbc439eb5e0b18d97b874c7f7b9ca489":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-01-08 18:21:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/927857/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/927857/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/927857/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/927857/5"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-30 21:03:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 18:21:30.000000000","tz":0},"subject":"obj: Add option to tune down etag validation in object-server","message":"obj: Add option to tune down etag validation in object-server\n\nHistorically, the object-server would validate the ETag of an object\nwhenever it was streaming the complete object. This minimizes the\npossibility of returning corrupted data to clients, but\n\n- Clients that only ever make ranged requests get no benefit and\n- MD5 can be rather CPU-intensive; this is especially noticeable\n  in all-flash clusters/policies where Swift is not disk-constrained.\n\nAdd a new `etag_validate_pct` option to tune down this validation.\nThis takes values from 100 (default; all whole-object downloads are\nvalidated) down to 0 (none are).\n\nNote that even with etag validation turned off, the object-auditor\nshould eventually detect and quarantine corrupted objects. However,\ntransient read errors may cause clients to download corrupted data.\n\nHat-tip to Jianjian for all the profiling work!\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Iae48e8db642f6772114c0ae7c6bdd9c653cd035b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d8fb046cbc439eb5e0b18d97b874c7f7b9ca489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d8fb046cbc439eb5e0b18d97b874c7f7b9ca489"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
