)]}'
{"id":"openstack%2Fswift~928291","triplet_id":"openstack%2Fswift~master~I820339934f895557ebcde4bb32168b0cfc494948","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-09 08:59:24.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-09 13:39:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I820339934f895557ebcde4bb32168b0cfc494948","subject":"proxy-logging: Clean up some timing assertions","status":"MERGED","created":"2024-09-05 20:46:18.000000000","updated":"2024-09-09 13:40:47.000000000","submitted":"2024-09-09 13:39:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"928291","meta_rev_id":"30c29120032c6c16a2c15abfd88a4c0192f69697","_number":928291,"virtual_id_number":928291,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-09 13:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-09-09 08:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-09-09 08:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-05 22:05:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-06 10:10:40.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"51a67f0ecfcb731604570d84e1f5bdd22a5956af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-05 20:46:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d4c1d6205999ebd19ae4870117fcde258bc7e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 22:05:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/180c07c5bee640218bb5d38f93b1be0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8990d58cf50e4d2990211e054d93683a : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/24f46b99af8947b59bad78322cdc5ff8 : SUCCESS in 1h 13m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f7822da50cc414da244512f9cce8ca4 : SUCCESS in 39m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e289fcc8e9244edb900d497e824b6cd : SUCCESS in 55m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84795de2a21d406680502c025b90f467 : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f24bd5ceb3c04049991aa90e0bb950fc : SUCCESS in 24m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b81dc768cf324ab09485f86fab1e132e : SUCCESS in 20m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9bac4d827405435483912fab23fd04a2 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/050a722e608a4e578cf81995b4be10f4 : SUCCESS in 18m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e264ae3d44f14bc09188efc6446472e6 : SUCCESS in 20m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/995cb5f868f34a4886facfa3d6f3cf47 : SUCCESS in 20m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/42b24e4229d94fa3853cfb6f0e56849e : SUCCESS in 20m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dd619a1f4e76484eaa389b4ea9bdfc03 : SUCCESS in 26m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4cfc5ec69acd4c7da0e7b96863c82966 : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8e664e67c3b0447cab6d530f08365a02 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f71a9f8a34fc4a8a86dd48085e0dc4a4 : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f97f2e92675f4bdcb5b64bd759353f19 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/13f0b9daace4477fb854adc814f58f2b : SUCCESS in 22m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b884286b43644355aad9c5039c4cce7b : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0ea61a6999b34c17b11a4ae1c7fecd08 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/68226dbf8e7d491790769347e1ad29e7 : SUCCESS in 15m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/723ebecf19434d03b19792ed2a484a61 : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd67ea650bbe4385b76644b4b3ee0ab1 : SUCCESS in 57m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67da20810cbc43a9b65feb0f90ae5f32 : SUCCESS in 45m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93fd705294b84c4d80ead87cd6b98c12 : SUCCESS in 44m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff5455430bd7427fae34af7778ed2211 : SUCCESS in 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9daa049bbaa045299a5e4240245254b3 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f9fa12a21ce84d9fa2e22cd094520aaf : SUCCESS in 20m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0efe8a13e88423081cd2eb9ed09faf5 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b4f95766cdb0d3874f097d6c3f45532fd45e9bcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 22:13:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0fb36ed15434250aa5e1d1a92beea90\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fd95f9b9fe84841ada64fb396278d93 : SUCCESS in 1h 15m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c209ecdd147492093a1510c08ff2f5b : SUCCESS in 1h 04m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b577c1ca6e1d494aab8ad916b8429a78 : SUCCESS in 1h 25m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/26ad7a0cb78a420ea8c21970437f9ed8 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":1},{"id":"99a6907e7fc02f11aa7787b81e1ad84d7b45e8cb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-06 10:10:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f4452d899a0f0ed18e8ea91967b55c9d0579f54d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 23:12:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4658c05d6dab41f4b28979d0d04c588ff9c563b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 00:04:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3938962ee8134feba61473ff1a555fdb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/127a46ad2de249c3ad9a21d3c01be334 : SUCCESS in 8m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f466ef732184338aa8ba8dd0c73e54c : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73d332d12e514e0899a9410bc70d145e : SUCCESS in 13m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c6c2fd027794fa2b83ad844c045df8a : SUCCESS in 23m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe78e70171284f55980817d2856b5b03 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/093e36ac248b49359f713eafd04ec71a : SUCCESS in 22m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1816bd576b3461280f0cce343be004f : SUCCESS in 16m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1b535edb87f840ccaf8a3b17ab712145 : SUCCESS in 20m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3c2596049ea541f0a9731bc11ff73954 : SUCCESS in 18m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b513b1c987b148d09e7ae42f82bdb7f7 : SUCCESS in 27m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c77f273ff094bce867ec18b2c2e4dd5 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/696a2c69bf6e49c493c9a58f5ca0e950 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/660e65ecbfed4b8e8ad37fe3c9dafc30 : SUCCESS in 18m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d2324bed405e43ee9b09ed817b9127be : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2735fb8bb9394f189f40482f3d9ff27c : SUCCESS in 18m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/659980379b6e4838ae89dd4ac4e07e61 : SUCCESS in 18m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e8997e2b470048d9a757fee61c478446 : SUCCESS in 48m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85c65a651f234946b3170c2bd010f6ea : SUCCESS in 32m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab24a54178704cd1aa720f1c445bc533 : SUCCESS in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/399690ee119740638523281766000ac2 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1c81a0bcc7184fa5a6e5379a732e1f38 : SUCCESS in 18m 56s","accounts_in_message":[],"_revision_number":2},{"id":"3bc3ee48e353331aa2020bef28609018e640f472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 00:21:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3be0eb28c0f47428250ab624ade26e8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81cd1e59574447b4950ff410d0e911da : SUCCESS in 1h 07m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/333130aa3f4c4f43952fa4d2b35fb071 : SUCCESS in 46m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/813506e336284f678f0769fae92c2a4a : FAILURE in 58m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ce5e5169bdd43f7aca0020690129f57 : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":2},{"id":"f3683f2bbe12a18346ebf37dadedc5e271bbf031","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-09 08:59:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e88cd74d6e3f25f5eb0e5a060288a81db2d8b92d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 08:59:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"95bdad393c022655bdca21e7967a1ceedc7dc696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 09:57:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa46a7f81e5a476c81f348a97dd44a63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c63f6dbd93490b87a56f3e543493c4 : SUCCESS in 7m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ce7ea99d57bd46b6baebea7ee0ecceee : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5fabf8d082a34f5d930fc77e44c83fe8 : SUCCESS in 23m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d062a60f0a44217ba7ff3d1bf10777c : SUCCESS in 15m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/05910d08fef742d3a89a15582ac7d455 : SUCCESS in 21m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c760b9f2c2c74afcb6c28750a5177768 : SUCCESS in 23m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/49a02c8271664c3c838ffa6633c840cd : SUCCESS in 21m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1921a9eadfb74566b17d015407f7b93c : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/259aa407001344198d55a11088e4f680 : SUCCESS in 23m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5a4c36e96f7c455f81ffd12d8afa3359 : SUCCESS in 15m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d2026b4015c4424eb26042035f0a68a7 : SUCCESS in 12m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4d9db5c06e74414ba02c8f21dc406399 : SUCCESS in 20m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ba2cc8ff9ee490d8cc9fad00b4b4323 : SUCCESS in 18m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2dad33a605c47c4aac406930d646f88 : SUCCESS in 8m 36s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cef4f58996534bdaa79f763980b5a84f : SUCCESS in 6m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31d33da2493c49d0929a91ee7dcea11b : SUCCESS in 50m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ccf65d5bbcff47fc8b475034a703af31 : SUCCESS in 53m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce007638c5b941239e3d56e620b99ca3 : SUCCESS in 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf8d996beb9494e9883a74d187e2bf9 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e324b0e08eaf4d30af98e4c26f93d0b7 : FAILURE in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"30899537b8b4434a631561500442b9ab31cb061a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-09 10:19:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cebf3553c4bff8462819adf22e8fbbe149f8264e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 11:57:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0fde658f1074403a74c15100420b077\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e03f25e88d4d5790cfb72f8cbb04af : SUCCESS in 8m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a9e5645fe714345af95ae94cf9c091a : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96fe174f19f146acacd273f753d100f5 : SUCCESS in 22m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c2a31aa5df343e9918240b8329c5e16 : SUCCESS in 22m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d6280d8b4cb742f28ca24fe1c42923ce : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d9a5a94463414ddbbc358e6053e742bf : SUCCESS in 18m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13489695d7fe4cb384b2150ab32b11c6 : SUCCESS in 19m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b987e39d5209410da5bb2268a6e48401 : SUCCESS in 21m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f8f7635421ba45d89ebb89302a042e2e : SUCCESS in 21m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d5e5c2b6de964b76a6202fdaeacd5f39 : SUCCESS in 15m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c95bdf65f5ef48e4820c37d0d7417273 : SUCCESS in 18m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ebf036a4bbd54155837706b1e1f99136 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/72456e81a33f4c428cb72a9dd4a60357 : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c1d7e9e7cf4943b5816dd5b40901ed0f : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7d3c34aa6ba743da870d17e3261cf3f2 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d14c3e7f0f145028f793dbf3eeddadc : SUCCESS in 18m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/870f7959885d45f685bd162024899c4b : SUCCESS in 46m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d7e28923a2734032970761b667a5cef0 : SUCCESS in 53m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5441734773c94866955c4a1fe0add6ec : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eb8ca0f32754be7af4d1748855d8634 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4bc4c41d02b841abbf67d40a1d2aabb5 : SUCCESS in 18m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4513579f7cd942f6642e7fd22500eb391ab66ba2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 11:58:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"21f091427ff65e1659c82e30cd333c7eb9485a86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 12:23:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a09c8aa4f24add951a55f04399f283\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/843a29a9809244719583cce059c8154e : SUCCESS in 1h 05m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e34efa3839f24b7498594dc6e48747e5 : SUCCESS in 1h 10m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4aff9a4102d419ba6ab15e0259e9f6d : SUCCESS in 1h 01m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ec533a175e243d18d9fc74fd6f3d738 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":2},{"id":"da36903425322f92a8c4b7c8fefdec0c74f86d45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 13:39:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6acdd1b74ab84b65ba047d31d5cfcc9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e51520e53d2e446ba280e777a2ffb026 : SUCCESS in 9m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9bc401285f204c8ab460b8659fa47ab7 : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/86b85b1060414be886e1d8726be9d860 : SUCCESS in 19m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df90871c93024ed7b7b6f1101f6aac1f : SUCCESS in 17m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cd962da151d347ecbca49b97610880e8 : SUCCESS in 22m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ee860a47f3ce4045abb5673da112b9ce : SUCCESS in 20m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c5632d5aca6d4f9aab7cdc76c3e0b789 : SUCCESS in 29m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e041f8268948428cacf3151b34950040 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/459484d4d01740cb9dab070cb2949afd : SUCCESS in 24m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/105a12d3e64c4c3db0fa8500d33eb183 : SUCCESS in 22m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2f2b65a02735400f946e6b6e900c4011 : SUCCESS in 13m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/37617c0df6304ea3a75a2c653bd11591 : SUCCESS in 13m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c5651d64ef8840698373b9e0e3ea8d63 : SUCCESS in 20m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/94eb86bc187f42f2b49817549f4e7043 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e462626072d2495baebbb28a4b791571 : SUCCESS in 7m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cbf6e192934b428194b2600e6d7d9b9e : SUCCESS in 51m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c210c651a56467a948721798ca0228c : SUCCESS in 45m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4158cb5b550c4313b34399b0cd70d765 : SUCCESS in 22m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c40b670d554ddfb87ad59374f6d50e : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/79af5864e561482cae64e31666499398 : SUCCESS in 17m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5bbda46dcb3bbb5c26c0ba60801b0f25301b20d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 13:39:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"30c29120032c6c16a2c15abfd88a4c0192f69697","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 13:40:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc829d25373440559c834ea03ec7df94\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7113748f6ac1444599e80a7c78404cf4 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3de9e0d0a53c4a1e82a65491e7d9dc6b : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7be25604db38c2f5752cceb72ea28e5f00d7cbe1","revisions":{"867d37c478253f94589e8a06d1c9680121952b47":{"kind":"REWORK","_number":1,"created":"2024-09-05 20:46:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/928291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/928291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/928291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/928291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/928291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/928291/1"}}},"commit":{"parents":[{"commit":"48e087b1cc7ea564d0f205baf5d625cad34be810","subject":"Merge \"object-expirer: refactor creating InternalClient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48e087b1cc7ea564d0f205baf5d625cad34be810"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-05 20:31:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-05 20:37:24.000000000","tz":-420},"subject":"proxy-logging: Clean up some timing assertions","message":"proxy-logging: Clean up some timing assertions\n\nRelated-Change: I526bbcc59c9eb5923c3784d5d06bc38998cb48db\nChange-Id: I820339934f895557ebcde4bb32168b0cfc494948\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/867d37c478253f94589e8a06d1c9680121952b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/867d37c478253f94589e8a06d1c9680121952b47"}]},"branch":"refs/heads/master"},"7be25604db38c2f5752cceb72ea28e5f00d7cbe1":{"kind":"REWORK","_number":2,"created":"2024-09-06 23:12:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/928291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/928291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/928291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/928291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/928291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/928291/2"}}},"commit":{"parents":[{"commit":"99e1c019b5a8627fd6c370bac6e024a8f7d9f70e","subject":"Merge \"common: Stop translating a bunch of printed messages and exceptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99e1c019b5a8627fd6c370bac6e024a8f7d9f70e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-05 20:31:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 21:51:23.000000000","tz":-420},"subject":"proxy-logging: Clean up some timing assertions","message":"proxy-logging: Clean up some timing assertions\n\nRelated-Change: I526bbcc59c9eb5923c3784d5d06bc38998cb48db\nChange-Id: I820339934f895557ebcde4bb32168b0cfc494948\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7be25604db38c2f5752cceb72ea28e5f00d7cbe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7be25604db38c2f5752cceb72ea28e5f00d7cbe1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
