)]}'
{"id":"openstack%2Fswift~928332","triplet_id":"openstack%2Fswift~master~I37b01677f7d9cf46ca20c9ff0ba64717158c49c9","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-06 23:16:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I37b01677f7d9cf46ca20c9ff0ba64717158c49c9","subject":"sq? functional test for missing path elements","status":"ABANDONED","created":"2024-09-06 09:49:47.000000000","updated":"2024-09-06 23:16:52.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"83764bc1d6e69100ea453b0777f4348f23f8f2ca","_number":928332,"virtual_id_number":928332,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-06 11:18:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-06 11:18:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-06 21:50:44.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"6a6f6092481d940ecc9a6531105d5ebc20dccf83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-06 09:49:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccab5d63907acabf78c979d6534af103e0a296e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 11:18:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0ea8b7f20a14adeb710850da113cb3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca5a6a442fa4411868cbf2ce81d5946 : SUCCESS in 9m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e09daceb0ac74f57a140ac85b1e1c76d : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/faca2896bde64a66a4310ebf03e24a84 : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6f682d8a5cb4f88a7c6bc404c476792 : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b4ba77905b34597b06bf9c44ca43a49 : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c12a05f6b5494fd4bba00bb3284241d9 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c87ebee91ad241169f2d9aefc860f1d2 : SUCCESS in 20m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/798d75a57b22420699d6f07ba4ec568a : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/459579a4e61d4380bd683a3d680f9cc0 : SUCCESS in 23m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c028784a8a6b4b52bcbcefe0bb808d54 : SUCCESS in 17m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f191421676b34a928a291b07d461685b : SUCCESS in 46m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3966ab0438f34027be5fd688f9bb4e6c : SUCCESS in 43m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7a76e9f58f4b40ac5484fe3f51c515 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d75581c2cdb41249bddfd673d6671b7 : SUCCESS in 17m 49s","accounts_in_message":[],"_revision_number":1},{"id":"cb220e4aac548d24234a07298629bb54cafa28ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 11:43:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa2e4cf1373543358ca66090c14c0e8e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/319a4fa538c74f31a33280f69407cd62 : SUCCESS in 1h 17m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81faa8abfe0b4178ab6702bfe4e1630c : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/32d4f47369314123ba1aa42f1bc6a38c : TIMED_OUT in 1h 31m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf2a8ceb512f4b489e75d02e09d345d2 : SUCCESS in 1h 21m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cc36e7e58abd5bd688304019af972b52afe77fb6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 21:50:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83764bc1d6e69100ea453b0777f4348f23f8f2ca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 23:16:52.000000000","message":"Abandoned\n\nSquashed into parent -- thanks!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c613945952a00a91a668e31fcbb60b1583a09dc","revisions":{"0c613945952a00a91a668e31fcbb60b1583a09dc":{"kind":"REWORK","_number":1,"created":"2024-09-06 09:49:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/928332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/928332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/928332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/928332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/928332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/928332/1"}}},"commit":{"parents":[{"commit":"59f0b5cfe18f80253a001c62271fc2cabf7b1713","subject":"utils: paths with empty components are invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59f0b5cfe18f80253a001c62271fc2cabf7b1713"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-06 09:48:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-06 09:48:44.000000000","tz":60},"subject":"sq? functional test for missing path elements","message":"sq? functional test for missing path elements\n\nChange-Id: I37b01677f7d9cf46ca20c9ff0ba64717158c49c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c613945952a00a91a668e31fcbb60b1583a09dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c613945952a00a91a668e31fcbb60b1583a09dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
