)]}'
{"id":"openstack%2Fswift~928464","triplet_id":"openstack%2Fswift~master~I374abcdd93664322cb440cd083f1ea4edf04d730","project":"openstack/swift","branch":"master","topic":"release-2.34.0","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-09-11 22:10:12.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-11 22:10:12.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-11 21:22:55.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I374abcdd93664322cb440cd083f1ea4edf04d730","subject":"AUTHORS/CHANGELOG for 2.34.0","status":"MERGED","created":"2024-09-06 17:54:15.000000000","updated":"2024-09-11 23:27:10.000000000","submitted":"2024-09-11 22:10:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928464-release-2.34.0","meta_rev_id":"1ab9b01bafaa3db951eed16bdc7a6425d0601d55","_number":928464,"virtual_id_number":928464,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-11 22:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-09-11 21:22:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":2,"date":"2024-09-11 21:22:14.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-11 21:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-11 21:22:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-06 18:18:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 08:20:10.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-09-11 00:16:38.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-09-11 16:20:01.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2024-09-11 21:08:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"c39565f7f81d405eaf16a4736f8b48d7f073e3a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 17:54:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd66d05ecea1365c686fe99dd92085b61b22f12f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 18:18:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/738260d89fe949f8af5d886a0cdf736b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4bd22911b2460aacdfe6268fbfed33 : SUCCESS in 9m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/925f1c8785ba438c91d4c0d8409ef5aa : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/73fa23da351c447b93ab574542a365d3 : SUCCESS in 19m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56ef7f19e53742318f9ade5ed7527115 : SUCCESS in 18m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/866c5e1fc0bb407d89d4f8d6708918ea : SUCCESS in 11m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5dcb81a5160247a19bb1b3050f807f53 : SUCCESS in 18m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8f413bf50f25425c8574408cabddd78f : SUCCESS in 17m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3253a8ff72a84d36b761172f2a861635 : SUCCESS in 17m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/445f134a36044c61af06bb4c2eff2d54 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c79d0eea079de3a7d1ac27b034e41cd2246a1cdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 19:48:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4276d3fba82c45638d8de5af7cbf7e8b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/30bb41d948c348819b8dd57da189242e : SUCCESS in 49m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7779546c5d7b4e5683ecaa040fda2d8f : SUCCESS in 1h 00m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c13e5888ad734ec585d112f387e5c03f : SUCCESS in 1h 08m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/600495618c564b3b8dccd5f43a4f2b16 : SUCCESS in 1h 12m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c3b7c012badebd658e88fa53361b7db4c1ac6c22","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 21:30:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5f78b0e060b91d445e6e96b3783baeff38b8423","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 22:15:08.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4de68fd501154742a7daf622e1b0563b\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fa5ccafc149a4e208b058e654ad02a69 : SUCCESS in 14m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4bbb54b6b9a645d28f36a6ce46f2993a : SUCCESS in 18m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d82ea97a80f744579f3a014f297972eb : SUCCESS in 19m 17s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/eaec8396663e4b1684558b8a623bb336 : SUCCESS in 20m 11s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/653ac440d82741e4b776483e78779cae : SUCCESS in 15m 35s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a966225347a04ad78480a2a7b3c0d5ef : SUCCESS in 18m 11s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/fb9fe87a6922438b9e864851b76f073d : SUCCESS in 15m 55s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/4ff95914796e4430a457d0df21059b9b : FAILURE in 5m 37s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/883e2fa5644a4e0ba992f681a9941d16 : SUCCESS in 23m 30s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/7b3cec9607294e45a9e913769d39eac5 : SUCCESS in 16m 31s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/a4f9a5403d164040a5692425221236e3 : SUCCESS in 22m 12s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/4b12c56cdf6e40c1abb3d0d2ac6fe77d : SUCCESS in 16m 56s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/c7f2c25823f14157a0728ff08d4996a8 : SUCCESS in 18m 12s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/95f60c520a664461a8fd28a8661636ea : SUCCESS in 17m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5d045f4e96749824616d2f1b4bb5573bd50e02ba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 23:29:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d58c05093fb77d1c3de63e32127815c53244590d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 23:57:10.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b0a41aadf44b508dac2b3dc226c0d5\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3375bd024d814ed9ac946dd326992ef2 : SUCCESS in 11m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b41c0e3bed184ffd81acfb74c348399a : SUCCESS in 11m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8d2b6e04df244dab99f37f3ddaa41f9f : SUCCESS in 11m 28s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/83b23540413143cbb47e92094b4ac065 : SUCCESS in 11m 19s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b927cab0c9b940378f3028e85db3a66e : SUCCESS in 11m 57s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/125687f6865c402ca4eb1acb2ab6e1d4 : SUCCESS in 11m 47s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/3a0b3b9e2ef045808658fbc2127c0231 : SUCCESS in 15m 34s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/a6f2193751b747daa71276407f9de944 : SUCCESS in 19m 52s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/b9c46532375d4722a8f066643454c509 : SUCCESS in 22m 22s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/110a5730c6e041aab25ce16c85699804 : FAILURE in 5m 22s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/e1659a40799749d584cebe36a344a505 : SUCCESS in 19m 19s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/5acaba666a9544aa9108be28c28699c8 : SUCCESS in 22m 28s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/b1a96c2389b84fdaa20c44d9ea7a9e2a : SUCCESS in 16m 32s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/a070f524fc994fe68877da46c3e112a7 : SUCCESS in 22m 44s","accounts_in_message":[],"_revision_number":1},{"id":"911e54c89af7d311948d17c2898afda54bdc52a6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-09 08:20:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ebacfd57a8903d52d30acae27a4f4bfdb3980766","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 21:08:31.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"155fa423161bd9c621936887118bd24c0d6ef130","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 21:08:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3d73e77d19eda1d5a03063e6e924bcfa5ca17990","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 21:12:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7330870f81bb37f5f43f875bc9fa877c8350691c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:44:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b92fa866fdeb4c709950ff877b848cf2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bb6e3a8a9e48e58821f90657549f62 : SUCCESS in 10m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6a2347d8b424174ae3a5253cdb1023f : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/afe0153132da44c9bd7972addd15122e : SUCCESS in 19m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62eef7797cac46f6accfd016efe22db8 : SUCCESS in 18m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d54f459be5ab49cb86d3f66972c6415f : SUCCESS in 17m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/07d230a7a1b6481090e0add082332134 : SUCCESS in 18m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/439f397d296348609fd2793cf2dfbd3f : SUCCESS in 20m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d74266aeafaf436ea9976cc6e4c811c4 : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f87ba7d28e48b6a8fa6ae2c7919a56 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"2765662e891412d597854baa556fb275b33606a8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:56:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b03565a16d9e426181e0e370bcd81baa\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/905e3d4aa1764562b7cac51333ef33d7 : SUCCESS in 17m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/47ad0d13d45c410783834c81d9f72cba : SUCCESS in 20m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d33305a300304b2fa694d52fb989cb4f : SUCCESS in 18m 32s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbc7d0116317473e8969760417a46b90 : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/23e39301aa63454699348b2385516356 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8b6c70acaad48bba943b9b1eb8cae78 : SUCCESS in 11m 22s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/515095def3d3490b86c7d34b26f4f0d4 : SUCCESS in 11m 12s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/11fcf1be7e934cd7be7c1f1eb6467a9f : SUCCESS in 16m 55s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/31cce93953d24ae087026f1abc50603e : SUCCESS in 24m 18s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/92a218489ad04abbb336dbaf0d3d485d : SUCCESS in 21m 24s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/b7fd3aeb77074c33b56d3d353a132d70 : SUCCESS in 16m 46s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/6640ffd0c91247bfadf20567cc24c554 : SUCCESS in 11m 08s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/a0ef03266bcb439e931959596a6a8f52 : SUCCESS in 16m 00s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/1ddd1d41d7174ef5a14dd819f2d9da97 : SUCCESS in 19m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ffbaca61fcc29ae2024c458a499f92f2339e8ebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 01:56:06.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87af50b3ed0a41b2a5fc1e247bef2e8c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/567b2b65b8534b28a43df7fbb1e1be3f : SUCCESS in 1h 10m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/010a4a77866b4fc0abf21fdc037f1c02 : SUCCESS in 1h 29m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c45034c642549d489fe645d4a86542e : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d42d777ef10476ebda1bf1b685249b2 : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c9e2424169787597aa1914cd8e217be9397dd0f9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-10 07:55:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eccb9500e0605409cacacfae1598f7a3ee32bd35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-10 23:41:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6e0880980b6a6d0dccb88247db8854dc14ee6ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 00:07:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b14904cfc37c4912bab8a1c31bbcde7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8dfd34352b443f280d6a69013f7efe5 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49e68b0135984146aa1a8b646fd11681 : SUCCESS in 5m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/510aa20c76f74d79982027384c1c1217 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/230f6dec7a9f4586958e3dbb44459c08 : SUCCESS in 21m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad5ce8910e174d119774716568c3d630 : SUCCESS in 10m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/baae557c804e4c599484dc2d364ef723 : SUCCESS in 20m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0cc5f4c8dda749edb639d672096c7365 : SUCCESS in 15m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/576222537ee14d509091df7816e12d06 : SUCCESS in 20m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3eb04645352e43069232723c8f5ad3ed : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7144e7381c87424ab6944e584263485b : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":3},{"id":"16a6b5b6c962bf79c39629e64bc025a670e7b19a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-11 00:16:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f12daab239b0aa9a05ba9e61176f53dea3b3a9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 00:52:47.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/655db1ce76404dbfa042f9dda2044341\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fef705d01354121bb59c8ac6f312c8c : SUCCESS in 46m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4e4ba4d916043b49063052ff756b016 : SUCCESS in 1h 00m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b553f44023604e0abf4677d999a34f68 : FAILURE in 1h 10m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a6103dd55b04551b0e0c93ce7316172 : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":3},{"id":"f6e5e7403b9d52d61ccdbfbfb23f2c09725d807a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-11 03:39:11.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d3e87916b08e41fffc9315cf2574a5627a46293c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-11 03:46:06.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d8d1a09eab03e7ba3347cdaaa54a1d0022f46ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 04:30:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/154eb97e2d0c44f9b2b9ec9f71c10854\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf0016e8d5d4afc80e9b73ffb45eaa4 : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/985f2547c11e48ad9df9efc934a9f1ef : SUCCESS in 4m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f6d390423e94fb286c56d34c5844b05 : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e22475bfc11c4459a570bebdb57ddf71 : SUCCESS in 17m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0936e123b8c94f0d91bcf5a1aa6672ac : SUCCESS in 19m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3931f70d01bd46d4baafa4cbb99228b5 : SUCCESS in 19m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e7932fa63f264ac598515787d07a31ae : SUCCESS in 15m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1c0be452b6c44b13a2af4f2c43687e7b : SUCCESS in 35m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/995c0f0fdd3c432183c745175367aced : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ede22f432949f287b2e3837e4a520f : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":4},{"id":"f9bca2cacc28185b3f0c00ce243e3f3f168e84e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 04:43:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f24bf9f3c884568bd2f893a1e0e1fdd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4374fb1fcf184eb6aa4794db0b4dfcb9 : SUCCESS in 47m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d48679301564b2f8666b82727a99574 : SUCCESS in 55m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/45e116295ef5429b9bb9c4d33c41e204 : SUCCESS in 47m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3af57bc5de244ff79f852c48f9b73db2 : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":4},{"id":"a0041f57e0785fd6c7a93be15d40ada43421bfdf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-11 15:40:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"782c7ae0c766175ad02b8d4fecca74d47c1bab3f","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-09-11 16:20:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"44290fa66aef806bb4fed7bf5a53302ea595623d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 16:22:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24f9acfed8af493483cd61375577caf2\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1527a023f23e4dffbf9d62d425f760cb : SUCCESS in 17m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a6b16d1c0b5f46b7a14e1081b76d97d0 : SUCCESS in 20m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/987b5d58f77e408290f51f6f51d8344b : SUCCESS in 20m 11s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/af6be8d81d634ebd91a09c41a1a7723e : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce166517f2d84f3681b6a4e34269f74a : SUCCESS in 20m 13s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6841574a89294e8a871337aac17995b6 : SUCCESS in 18m 25s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/6363b6c5d81f440ea729ce9fe65184cc : SUCCESS in 19m 22s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/3e2eb8b3023f46c09479812dc66a96b4 : SUCCESS in 16m 56s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/29ff1ec6207246529aa38e7fc8737d74 : SUCCESS in 18m 07s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/a1ff7f31ac354c4babf3c817b7d3f3b5 : SUCCESS in 12m 51s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/59364701293346a2a4afc985e3764320 : SUCCESS in 15m 52s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/18134f0fe5d742408fcb6f6f52acff83 : SUCCESS in 18m 14s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/68eb7c56ee4145068e4eff7a9b1636a8 : SUCCESS in 18m 53s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/5231f0d5bbb84f2391c498d7969370c4 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":4},{"id":"ce6cf16a34be3d388750a6863162901d433be9cf","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-09-11 21:08:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"323534e26f602494fedebb51e1564be555b1268a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-11 21:22:14.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5ec4147d34d4cb16f8238e73dc8020771719adb2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-11 21:22:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"56fe1bfb90c00fcb14503a3d1ed5320a34012a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 21:43:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd927fc0a58a4057958ee05eb109329b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7e71c5a9dc4f9f9b33d08c35c7cee9 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/183f9aa8f1a54a34b1e1ad073c32d7a2 : SUCCESS in 5m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7cc1065e12c45eeba872ad384c63908 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbed577834794ef799ee0392d20962da : SUCCESS in 19m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28955f1b461b40db9c6b7f9cd873ff71 : SUCCESS in 18m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/929aee2aa8134cc1bc6ab672b62846d8 : SUCCESS in 18m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3d22401aaa8847edb7774c258328c1fc : SUCCESS in 18m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a244d6345026408397577c0c2cc21ab1 : SUCCESS in 21m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c67a186c4ce4cd9b6eaabc37a4c7d0d : SUCCESS in 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c1bc2fb03e496fbb8754111a8dc778 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":5},{"id":"ca8c19887185d9d2827902d5e5f4e9206e428441","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 21:43:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c68d507f3c9b65bfa18f1170c8e50645eff9f418","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:10:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd48babd629045e58f8b3d751f898914\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76dff9dc7a3b4269967755200bb11a7a : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8629551fe4e4ecda7c10133e054965f : SUCCESS in 5m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/16da6210d92445e6b4b891bee657e7ce : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aff76855e3c44457b69fc73578639554 : SUCCESS in 21m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8303c768cd78402f9125ef1e31414cdf : SUCCESS in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/57de310246ed41539bc0885747426be8 : SUCCESS in 20m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f53c8e6639774f48927ada9c69face46 : SUCCESS in 18m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/12b1f8d703b14e14ad84cfe69a667037 : SUCCESS in 21m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a40f8f9bf2e49558d8528c4145e6713 : SUCCESS in 21m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3937695573a4136921c61b22dd2879f : SUCCESS in 11m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ea74c95bfd84ad18b105ab891623c26 : SUCCESS in 23m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/053d78d096a0459cba18c2e2054629f8 : SUCCESS in 10m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/757d809ef6e049d3a0724a8a7e160ebf : SUCCESS in 23m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/daf6512b9ffb466baff0e50158c0d0d6 : SUCCESS in 24m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/00e9a86c11cb44419edae9054bf04294 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb7fe8696f2044b59c5301b01d27fbf8 : SUCCESS in 7m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b37cb033cb6d47a39617de2743427aba : SUCCESS in 23m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c42b62832ce049868bb864be1c6cf709 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":5},{"id":"53adc9b3a905aa93c30c8f799b038a9195f852a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:10:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d5141a876151810a00f5ac7eb3b79472a8b1a216","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:11:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08428a94b0ac4a23820e35541f1aa1a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0f5b07a09f4e679179b69832c51863 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c4181f105014fdb9885ca4d4241c35d : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5db8bd60382c43bf8ad714ccbb45e206 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1ab9b01bafaa3db951eed16bdc7a6425d0601d55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 23:27:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e45c3afca62a46fe82b3c5690dac3640\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2cf6785623a4092be1a02a8c992bee6 : SUCCESS in 1h 08m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7802a985a9e481f94f5a06258fb2f0f : SUCCESS in 1h 15m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f71fa0e1e6444b0886a9a42fd54b9c7 : SUCCESS in 1h 08m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a5e042d3ca74268b2b2dd2498e2d828 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0e41e179003fac2a0ca4bc4c913e9a6d574ddf68","revisions":{"387d9bc14e422a77abbde7d174e884797073f650":{"kind":"REWORK","_number":1,"created":"2024-09-06 17:54:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/64/928464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/928464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/928464/1"}}},"commit":{"parents":[{"commit":"48e087b1cc7ea564d0f205baf5d625cad34be810","subject":"Merge \"object-expirer: refactor creating InternalClient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48e087b1cc7ea564d0f205baf5d625cad34be810"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 17:53:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 17:53:49.000000000","tz":-420},"subject":"AUTHORS/CHANGELOG for 2.34.0","message":"AUTHORS/CHANGELOG for 2.34.0\n\nChange-Id: I374abcdd93664322cb440cd083f1ea4edf04d730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/387d9bc14e422a77abbde7d174e884797073f650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/387d9bc14e422a77abbde7d174e884797073f650"}]},"branch":"refs/heads/master"},"1b8166df0e9c01826e2fe516071759aeffbc874f":{"kind":"REWORK","_number":2,"created":"2024-09-09 21:08:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/64/928464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/928464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/928464/2"}}},"commit":{"parents":[{"commit":"f7db1ec3cfd9e130b67b97e6a4d5e24cfc92e4eb","subject":"CI: use private IPs for multinode tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7db1ec3cfd9e130b67b97e6a4d5e24cfc92e4eb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 17:53:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-09 21:08:09.000000000","tz":-420},"subject":"AUTHORS/CHANGELOG for 2.34.0","message":"AUTHORS/CHANGELOG for 2.34.0\n\nChange-Id: I374abcdd93664322cb440cd083f1ea4edf04d730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8166df0e9c01826e2fe516071759aeffbc874f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8166df0e9c01826e2fe516071759aeffbc874f"}]},"branch":"refs/heads/master"},"df4fd93af58758920f9d5b7c0caaa9f648cdc472":{"kind":"REWORK","_number":3,"created":"2024-09-10 23:41:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/64/928464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/928464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/928464/3"}}},"commit":{"parents":[{"commit":"6f7687b23c43b616b1402898e6657e28c6a49fbe","subject":"Merge \"trivial: Use already-parsed a/c/o\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7687b23c43b616b1402898e6657e28c6a49fbe"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 17:53:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-10 23:40:53.000000000","tz":-420},"subject":"AUTHORS/CHANGELOG for 2.34.0","message":"AUTHORS/CHANGELOG for 2.34.0\n\nChange-Id: I374abcdd93664322cb440cd083f1ea4edf04d730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df4fd93af58758920f9d5b7c0caaa9f648cdc472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df4fd93af58758920f9d5b7c0caaa9f648cdc472"}]},"branch":"refs/heads/master"},"f69635e11e1ff4d8794c65ab04f5be246ccb0cef":{"kind":"REWORK","_number":4,"created":"2024-09-11 03:46:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/64/928464/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/928464/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/928464/4"}}},"commit":{"parents":[{"commit":"6f7687b23c43b616b1402898e6657e28c6a49fbe","subject":"Merge \"trivial: Use already-parsed a/c/o\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7687b23c43b616b1402898e6657e28c6a49fbe"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 17:53:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-11 03:45:54.000000000","tz":-420},"subject":"AUTHORS/CHANGELOG for 2.34.0","message":"AUTHORS/CHANGELOG for 2.34.0\n\nChange-Id: I374abcdd93664322cb440cd083f1ea4edf04d730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f69635e11e1ff4d8794c65ab04f5be246ccb0cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f69635e11e1ff4d8794c65ab04f5be246ccb0cef"}]},"branch":"refs/heads/master"},"0e41e179003fac2a0ca4bc4c913e9a6d574ddf68":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-09-11 21:22:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/64/928464/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/928464/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/928464/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/928464/5"}}},"commit":{"parents":[{"commit":"ffdf96259814e554dc16ea29b3de6b4e37173158","subject":"object-expirer: fix unused _make_internal_client arg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffdf96259814e554dc16ea29b3de6b4e37173158"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 17:53:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-11 21:22:07.000000000","tz":-420},"subject":"AUTHORS/CHANGELOG for 2.34.0","message":"AUTHORS/CHANGELOG for 2.34.0\n\nChange-Id: I374abcdd93664322cb440cd083f1ea4edf04d730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e41e179003fac2a0ca4bc4c913e9a6d574ddf68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e41e179003fac2a0ca4bc4c913e9a6d574ddf68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
