)]}'
{"id":"openstack%2Fswift~928475","triplet_id":"openstack%2Fswift~master~Ied0ff6bea7e054fad3fe9579c85d9ae5c9c0b255","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-09 04:01:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied0ff6bea7e054fad3fe9579c85d9ae5c9c0b255","subject":"tests: Functionally test account quotas","status":"MERGED","created":"2024-09-06 20:54:27.000000000","updated":"2024-09-09 04:02:09.000000000","submitted":"2024-09-09 04:01:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928475","meta_rev_id":"a15850bfcc52f5ebc1d85c2f5fbec6f75f414b76","_number":928475,"virtual_id_number":928475,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-09 04:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-09-09 03:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-09-09 02:25:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-09-09 03:21:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-06 22:29:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 02:25:09.000000000","updated_by":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"reviewer":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"state":"REVIEWER"},{"updated":"2024-09-09 03:21:01.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"91ee2912625b591b250b41cfbece937e8209d858","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 20:54:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4669160e1dfc171e3dfcb24d21275a377489651","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 21:29:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b0c76912a2678fc7d871c43e4741ead405e58d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 22:29:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e791a3f35c48b89f9f9ebfc29a3b14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1821ab7c3b70415ca71a2ac5462f7cc4 : SUCCESS in 9m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96381901b2c449a2876a6a93c69b4267 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff41f1dddcd047ea90f7922ccc0afdd5 : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ade50a4892834c4d887e59ef75e5a446 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0fd77490af5405b872937bd5c374069 : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b3493de4e8443b3a424f887fe091a00 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/abafa2e40d3748a799ae8288895ebd23 : SUCCESS in 16m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a780a3b7c69a4e95a69ff42175c7b415 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/421c72d2bd8d451894a07a4297fee5d6 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/240603d2989d4e91a22bc6a1b7f70ffa : SUCCESS in 19m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c495dcb58845431db36612b3820531cd : SUCCESS in 31m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/afb6e70aead14d7fab9a58794acf9933 : SUCCESS in 53m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a395c6a210644cdad467b2fd36f869d : SUCCESS in 3m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2eaecada9b8431884442c25e897c4e4 : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6d26dc94e7c423b65870389a3ebdce2a07dd80b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 23:01:37.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47c9a3cc938b4066b2f1f21544606ee2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7f98611733e470496ba15872e6f0c43 : TIMED_OUT in 1h 30m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd1d9b030f354f3cbe789551b7521926 : SUCCESS in 46m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9e1daf7847b4216b23e0928e161a029 : SUCCESS in 55m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a444499e5b3f48c7805bbb33ac1e6dd9 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5ae2d5e5d2ee3264908c6f21a78a9b8272d915ec","author":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"date":"2024-09-09 02:25:09.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"506cd2e869cf84c623f3e09ab4800c0de9be0e23","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-09 03:21:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7050c20adc237c897ad00ebd873c957128223411","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 03:21:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8fe1cca1a97c514cf9cde5373a77eba43925c0bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 04:01:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d0719a59c8e4b338a943f008cc403d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1d17a7295c466f9f47d6ab73d9ff80 : SUCCESS in 8m 13s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/21508d3cf7004d48b8da0e2470fc2293 : SUCCESS in 19m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c619f474cc7d41e4ad12619a15716de0 : SUCCESS in 14m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f415143b536c43bf9b34630a564e09d9 : SUCCESS in 19m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cf20174a91e14c7d8660e376c1feb491 : SUCCESS in 14m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/38c6d8e18b034cfe9b580913158d296b : SUCCESS in 27m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d5140be78fb3476ab06a1cc802d501f8 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a42c145c3ea349859427f6782228ec9c : SUCCESS in 17m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/44378ed19c8148df8f76c18bc0d04240 : SUCCESS in 17m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bb18407e338647c0900930c12be22a0a : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/38ee9b2b881349759dac1c6867b82e28 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/693677f691e7400883010709ee6872f5 : SUCCESS in 19m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6d81ac42c014899ab846f64c6efde72 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6179d909c8df4a9988420a8e8a00ece8 : SUCCESS in 6m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39544daec2c94ddcbe9b7f8bba3e44f8 : SUCCESS in 38m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d500647d18794d7ab3417fed999bc086 : SUCCESS in 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71f3ee548cc495482cf8c91a9ecd32f : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e1fe1645477649598d1e057bbe549238 : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":2},{"id":"fd985948f8a23ec0c7f171bdee9b55cde5b9f2b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 04:01:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a15850bfcc52f5ebc1d85c2f5fbec6f75f414b76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 04:02:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc1c356450894728a75eb9092c4f34c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72408696e860479191e8e3b95bf9c77f : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5e32a54eb662492bae00aefe7babaa3a : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd288b183d1d5df6d8c504fa2303e83f49097f75","revisions":{"ac967f8e44104ff9620d8622adc9d96ebc4f4f6d":{"kind":"REWORK","_number":1,"created":"2024-09-06 20:54:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/928475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/928475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/928475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/928475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/928475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/928475/1"}}},"commit":{"parents":[{"commit":"349d225085cc3197742b480c1823cb81e9b8a5b8","subject":"Merge \"common: Stop translating log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/349d225085cc3197742b480c1823cb81e9b8a5b8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 19:26:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 20:47:35.000000000","tz":-420},"subject":"tests: Functionally test account quotas","message":"tests: Functionally test account quotas\n\nChange-Id: Ied0ff6bea7e054fad3fe9579c85d9ae5c9c0b255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac967f8e44104ff9620d8622adc9d96ebc4f4f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac967f8e44104ff9620d8622adc9d96ebc4f4f6d"}]},"branch":"refs/heads/master"},"cd288b183d1d5df6d8c504fa2303e83f49097f75":{"kind":"REWORK","_number":2,"created":"2024-09-06 21:29:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/928475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/928475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/928475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/928475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/928475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/928475/2"}}},"commit":{"parents":[{"commit":"99e1c019b5a8627fd6c370bac6e024a8f7d9f70e","subject":"Merge \"common: Stop translating a bunch of printed messages and exceptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99e1c019b5a8627fd6c370bac6e024a8f7d9f70e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 19:26:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 21:29:01.000000000","tz":-420},"subject":"tests: Functionally test account quotas","message":"tests: Functionally test account quotas\n\nChange-Id: Ied0ff6bea7e054fad3fe9579c85d9ae5c9c0b255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd288b183d1d5df6d8c504fa2303e83f49097f75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd288b183d1d5df6d8c504fa2303e83f49097f75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
