)]}'
{"id":"openstack%2Fswift~928476","triplet_id":"openstack%2Fswift~master~Ic28752d835e0b970f2baa4e68cbfcde4f500b3d4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-09 07:27:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic28752d835e0b970f2baa4e68cbfcde4f500b3d4","subject":"account_quotas: Fix X-Remove-Account-Quota-Bytes-Policy-\u003cname\u003e","status":"MERGED","created":"2024-09-06 20:54:27.000000000","updated":"2024-09-09 07:28:46.000000000","submitted":"2024-09-09 07:27:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928476","meta_rev_id":"4bc175561033986b4f970beeeb6776146f86aeab","_number":928476,"virtual_id_number":928476,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-09 07:27:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-09-09 06:09:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-09-09 06:06:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-09-09 06:09:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-06 22:53:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 06:06:52.000000000","updated_by":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"reviewer":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"state":"REVIEWER"},{"updated":"2024-09-09 06:09:21.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"a8303adf7a7421aacc3fb6a5ec96692564f1036a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 20:54:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe451fcef976dbe9c7a80809a7e8a28310296b25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 21:29:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ce8536957c5a33c4e2c624d6bdb02cbb6d7b7686","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 22:53:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90cc076ee1ee44eb80139377fab7a9a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/006a524ecc574a6398a312e532461d6e : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/069f7a947bf347138d74347da69dd4ce : SUCCESS in 42m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1af1614a22b24c9a8e3d86141fb3c305 : SUCCESS in 54m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc3350ef9c3c4f59b907098131deb045 : SUCCESS in 46m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af7521fee025459bb56219845b4623d0 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cb36a948a134cc18ee2d3da84f10926 : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f26281ecf4444879b0b61ff8f260918 : SUCCESS in 25m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ddb6bc5250e24078ad9c048f40401253 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c7040a861d64413aee5efbd73813221 : SUCCESS in 20m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee6c09a3924b4d148d37514d36b623ee : SUCCESS in 14m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ac23e18fa32445dba62afde2affebbc5 : SUCCESS in 17m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8b2bc5a97cf749b3a207b2539465ea04 : SUCCESS in 15m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c7884d33616e4553bedfb6ff41384d26 : SUCCESS in 29m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/687ce21c064141a58c9c3338e7732e2f : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/67ab94958a674a4e8248abb07f8aebe6 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e8c9db274e324488b0accf392ff647ef : SUCCESS in 18m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a29a4f92f4b949caa09083f2fe20af88 : SUCCESS in 10m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1c3d1ad7bd74f6e8232687e83581230 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/877a666e878d4a25a802ae4ac07e5823 : SUCCESS in 18m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c564b37d2d5948638ba5db0a705c7603 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1eb97680525428ba11503ac4988608c : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9cbcf6abba6a492a9e6f39797fd20874 : SUCCESS in 6m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b964e0d59aa451694ec29e53fbc18a2 : SUCCESS in 38m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aef02e74859a4b568aab3e46c800d062 : SUCCESS in 33m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed12a4a76a0842a08de0f377477409eb : SUCCESS in 52m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd21f783ce1447adb49d3e8dbf0599c2 : SUCCESS in 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74fe8a19df442b693271343b637a479 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12e8ad8b9b11425ca24f5e37cf1b26c8 : FAILURE in 5m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5955e0e20a534e1ca0b0ece341504258 : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":2},{"id":"645f1a2cb61972073ff093cf49be63f7cc017c0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 23:02:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da0894505d94a0982820f6658fb6205\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68d7593654164d67bac9ba296a01878e : SUCCESS in 53m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af7e95a52de84bad9cda123ddce12ef5 : SUCCESS in 59m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf5a421f6fe443629839898dd184c44e : SUCCESS in 1h 01m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f127ffc2944f49d5993c3a0c33a92457 : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b1cfa19282203309f41f75c5d9b94cbcea03b9b6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 23:15:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6ebf7bef20a0042e44a100538e61cf8a88dfa20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 00:30:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00e901001fa04c368a58e4c2ee7981e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6333531dce4f479986ad03ae24be7ba4 : SUCCESS in 11m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/4497407acbc54888869ad9364417c704 : SUCCESS in 44m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38ccbe55e93745818728711b6b34d439 : SUCCESS in 57m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d05807e8e1f8496d942d2f5268bf361f : SUCCESS in 1h 04m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a74fc77954484678ad4bb6b3dcb1a134 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee139860955345d68fa4300e85a05f34 : SUCCESS in 13m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32d48ab5be2e4f8a97b4f0b45507204d : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7e7d003a0ea46bd9379f5da20bdd94f : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5c27c2f9475945f3ac7e6c464aa8c844 : SUCCESS in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee5ff66353554249af2fe5f768269a59 : SUCCESS in 18m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f813effae630487f9fd7d53dd37abac0 : SUCCESS in 20m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cc9f21596fd14690ae612ab65ace1c11 : SUCCESS in 16m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6b9348b590054a7d99506d79851e90a2 : SUCCESS in 22m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f55cb9842c544d38dc94cdf66975116 : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3c4b5dbcfb2456eb632472c3c257104 : SUCCESS in 19m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b0afbd2ffb74df89c4096082394dae5 : SUCCESS in 19m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5cb8da603a424916abce2f0814fa545e : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9d11d9cba3c24d0e94ad97593cbacf60 : SUCCESS in 10m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b827799cbc2148968b0672c468f8753d : SUCCESS in 17m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b5c625e53f742f8917f9ef741dbd5f4 : SUCCESS in 5m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d287bde5546745f08898dc1a5dd50d1e : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9289c21896d4524bc4b9a00f1b7becf : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a01729d1ebb4031849ee2dd5cebcaff : SUCCESS in 48m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de3987cc9e5b4a09805cbc24d531ac8e : SUCCESS in 41m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/03ccc900322f4274a075f7a9ffd3892f : SUCCESS in 31m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0d80b8903534839b26002510751c7ff : SUCCESS in 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8860e0c767454261a13b3c8209c6c679 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a86e604b87f349c0a0b8e9963619745e : SUCCESS in 21m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66e370771a3c43c39aa0971238b5cc88 : SUCCESS in 43m 55s","accounts_in_message":[],"_revision_number":2},{"id":"2ed7a8b03c10ee951655b08a6e01d55db90eede4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 01:38:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c62e253170b4c05aad36c3de43f25d8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/296fcbd5b4b84ea9a48ed8dc789037fc : SUCCESS in 1h 06m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7a8f7cc4a06408e8e2e5b015b165e13 : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad3513651a824248b9b018ff64ca082f : SUCCESS in 1h 00m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d3e3a94441b4fb69c41bb34b888cea6 : SUCCESS in 1h 05m 19s","accounts_in_message":[],"_revision_number":2},{"id":"fa19571d55e65738c2b0ae32953b4fdbbfc1ee4f","author":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"date":"2024-09-09 06:06:52.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79e1420745186d153895ea5d3545d42ff2202eb4","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-09 06:09:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"642490c375ed7b5fc1f0ee116dd395c3a2912a13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 06:09:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a99f894ae50ece74f40ee3ab20e4f6f4662a75a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 07:27:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/920f6bb1000e4293b7bd8064100dda41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e3738eb30cb41078145eb63c1d6a88b : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/067c4326e5334d93b317fb6ace0c332d : SUCCESS in 1h 03m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c11bc10d98d541c1ae6389451aa27301 : SUCCESS in 45m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f63c479f895438dbbd5786ecb889239 : SUCCESS in 1h 00m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/66437742dc63490f9c7727fa7287414b : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/610cd6dc3dcf41bda53677159f60712e : SUCCESS in 20m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/50518397a69e4f31bc14a3aee4c13f2d : SUCCESS in 17m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/54e50585247949aea7589fdfcf3155ac : SUCCESS in 20m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/465b29b5970c45a99b6b1ad475dcdfb3 : SUCCESS in 19m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0199fccc8f70410a9ae78186042c106c : SUCCESS in 20m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b369de1f3ce240d599d37724a1e909b2 : SUCCESS in 26m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/06a59287b3fd43938f90eae26688d2ae : SUCCESS in 17m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d10ec96c687f4fc081d1e5fc134344a1 : SUCCESS in 13m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/490d173e2cf048f09f2c11d1eef3a790 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b3f6a17f4ef849fc953c85d77972bc14 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aeb0fb298775424b83391bf57ef3a47a : SUCCESS in 18m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf358cae545c42b6afc483d713e182cc : SUCCESS in 4m 45s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1831c85a02e5437ea4b04f31b86cdd1a : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/847faf4906974275bec23aba3ad63c72 : SUCCESS in 48m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a6ef59f6e97c4d259f9b38ab14f52b81 : SUCCESS in 50m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a19f930cb63147f4ad7495ef47432756 : SUCCESS in 40m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9bdd39e7d0e414cacda18ad728ab4ed : SUCCESS in 19m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7146357134ce4da2aebc9ecd048d8589 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f94cf8c7dd814ef4b926ae421bf6da8d : SUCCESS in 18m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e64e2bd292347808c65227bd65cf0fa : SUCCESS in 46m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f27238dbd60bfd5ddd360ba2b38868de660d42da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 07:27:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4bc175561033986b4f970beeeb6776146f86aeab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 07:28:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c251723f4ec4c369150a7405f6fc5e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65abd16f64954110b9c7188cd14bdfae : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4221812922ef493b81d4c71033fd8864 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd0fe25da13bb109c2e564036ed00eefc12b148b","revisions":{"b7690e69536ee9996bcfe852e688da5a52dd2829":{"kind":"REWORK","_number":1,"created":"2024-09-06 20:54:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/76/928476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/928476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/928476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/928476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/928476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/928476/1"}}},"commit":{"parents":[{"commit":"ac967f8e44104ff9620d8622adc9d96ebc4f4f6d","subject":"tests: Functionally test account quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac967f8e44104ff9620d8622adc9d96ebc4f4f6d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 20:30:38.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 20:48:25.000000000","tz":-420},"subject":"account_quotas: Fix X-Remove-Account-Quota-Bytes-Policy-\u003cname\u003e","message":"account_quotas: Fix X-Remove-Account-Quota-Bytes-Policy-\u003cname\u003e\n\nPreviously, this would reduce the account\u0027s quota to zero, which seems\nlike the opposite of what the operator intended.\n\nNow, remove the quota, similar to sending an empty quota header.\n\nChange-Id: Ic28752d835e0b970f2baa4e68cbfcde4f500b3d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7690e69536ee9996bcfe852e688da5a52dd2829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7690e69536ee9996bcfe852e688da5a52dd2829"}]},"branch":"refs/heads/master"},"cd0fe25da13bb109c2e564036ed00eefc12b148b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-06 21:29:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/76/928476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/928476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/928476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/928476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/928476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/928476/2"}}},"commit":{"parents":[{"commit":"cd288b183d1d5df6d8c504fa2303e83f49097f75","subject":"tests: Functionally test account quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd288b183d1d5df6d8c504fa2303e83f49097f75"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 20:30:38.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 21:29:01.000000000","tz":-420},"subject":"account_quotas: Fix X-Remove-Account-Quota-Bytes-Policy-\u003cname\u003e","message":"account_quotas: Fix X-Remove-Account-Quota-Bytes-Policy-\u003cname\u003e\n\nPreviously, this would reduce the account\u0027s quota to zero, which seems\nlike the opposite of what the operator intended.\n\nNow, remove the quota, similar to sending an empty quota header.\n\nChange-Id: Ic28752d835e0b970f2baa4e68cbfcde4f500b3d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd0fe25da13bb109c2e564036ed00eefc12b148b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd0fe25da13bb109c2e564036ed00eefc12b148b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
