)]}'
{"id":"openstack%2Fswift~928478","triplet_id":"openstack%2Fswift~master~Id9adcf94a8847d3062c99258602ecba4a8a45edd","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-09 16:18:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id9adcf94a8847d3062c99258602ecba4a8a45edd","subject":"proxy: Clean up some bad-URL error codes","status":"NEW","created":"2024-09-06 23:12:35.000000000","updated":"2024-09-09 21:17:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41731d6e7e9f57d76083ad2c0089c4e5c18fbc6d","_number":928478,"virtual_id_number":928478,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-09 17:21:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-07 00:14:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-07 00:15:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"765ed365e91b50ec7feea296e7a18e1865333a13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 23:12:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79dc45a0e3f08f8a144bc2e176112af0e79e0a63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 00:14:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84733209c25549968ae1bc88e501c562\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4e5e4714fde47adae7d20c13cf34340 : SUCCESS in 54m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/574580f02abf4871b895ca0642682b05 : SUCCESS in 51m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/80a26b1941164cc4854ea2deea34ee13 : FAILURE in 58m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/56c3a7d660ab4ecfb259bf39aef573f4 : FAILURE in 1h 00m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6b4782cb7df24341799ce8cda391b39704b8680d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 00:15:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3b4aa2475354166a8c2c27c3f10d123\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/389b6a4992024bf2b8207b3d8873380e : SUCCESS in 11m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/4b062efcdc5143318e87009953b7286a : SUCCESS in 55m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/edbef619c4244bbabcf655a785ea0fdb : SUCCESS in 51m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca7c317e63724ab6b628cfabd296da3f : SUCCESS in 54m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52e84fe0163c4a8fa4ccd3573fe20ba3 : FAILURE in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a201a3cb9ad4ac7b4e4bf88627dd633 : FAILURE in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/343cc98d79dc46439147227c03cb2a6d : FAILURE in 22m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9088a8bca70a48ebbfb985553ba99173 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9e05800d059849dc8a85b843ed9f3493 : SUCCESS in 22m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5fd865806f24de69963149630c471cd : SUCCESS in 10m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/473159e845e44d1d81d5b17b6499d335 : SUCCESS in 18m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d3fd8f950e73484687e0fc2e61f52312 : SUCCESS in 19m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/34b827cf73dc4021a2176619c92b2f3a : SUCCESS in 28m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41bbf18d74e64c638f814e8bcf9bf957 : FAILURE in 13m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e9c2b492d27f436f93d01c44066a2919 : FAILURE in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d813d6835e3a4f85b9450d2f8f1b4b9c : FAILURE in 23m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e5686e2f3044033b0b55b6af5e2849b : FAILURE in 14m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/189c6b77c1a14022b9dfc48499a8eace : FAILURE in 20m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/414f6aec07be4a1899596a17a1d51563 : FAILURE in 18m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/840f70e44fb248c59341b883f82bee27 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d9220dced610405e87ab3ab9278224c8 : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cc7e84b4c6fe4a99b1dbb0e58e08b124 : SUCCESS in 4m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4e875bbe91c410f9db5e1d501f8fb07 : SUCCESS in 46m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c89e13818524fb09ad1aa11de749007 : FAILURE in 28m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cb34c7f10464fadb7794df686291fe0 : FAILURE in 22m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a46b1e618024936bb5d143de1d58477 : SUCCESS in 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42130a92cd2242228be61225b6ff6879 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b6ce8375a6e4799a0c331b9a1fa391b : SUCCESS in 23m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/701318fda5ef415e8303b8abbaf319e3 : SUCCESS in 49m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8414af4355d34448b6868ffe77fb9b44b17b0a47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-07 04:01:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bab0e57319b3af74dc59933feebc3824987177d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 05:01:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32c86ae3ac2a49349e9745b075afd322\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a82d268b8a4fb1a5f37c6f0e0cfcbb : SUCCESS in 5m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f5b48e3b26b84b95a23d5007fd063ca2 : SUCCESS in 39m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/31dbd7d37cb548bda10eda437efef34c : SUCCESS in 38m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/829b052e6bb5478abf327140422fe5e9 : SUCCESS in 40m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd14bafc54d3412498405403b9baf988 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae14a8323ad34e5782531d1cdf99f143 : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db3998b7df2c4077b0672d9590a5a555 : SUCCESS in 21m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/258662424303464f963f6dae26519613 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d1ddb4fb2c0b40879bbfcb508c2c7a0b : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd520ab449c6464496cec1eb4451d219 : SUCCESS in 14m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/18e66b074055463e8b640a361c9325f7 : SUCCESS in 17m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f394e467219848d8acfd5da29dae52f7 : SUCCESS in 15m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dfeaa30d48f9404fb66beb54d5d31d95 : SUCCESS in 17m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19bfa7b0a832491aa3d1cdc8a998c029 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5a8f55e9809942f790bbba15b6df021d : SUCCESS in 16m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/905678b99a794e989b72028941e2a949 : SUCCESS in 16m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/91eeeb6d1e344f4dab63dd1b92ee6b4b : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ec17b7e950ba4562927f6fc5206d6276 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/093c610aa40443428a0c325a60fb34ab : SUCCESS in 11m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7676f3bfc05b45aba3c7f60cac552d4e : SUCCESS in 4m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f127cdc792274d5484e38a92d86b01d4 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/604d4cfbfa4e482e9dd43f236cb4e6fa : SUCCESS in 5m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b437fda354b64dbd9a4f9eb6590a57a0 : SUCCESS in 37m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdc8905075cf417b801a6d094f415921 : SUCCESS in 52m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5ed69288ce842f58c236172a38c8dd6 : SUCCESS in 30m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d64027c8e9c74312a16c9e04ee0f63db : SUCCESS in 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f483b1db63374e358d626d5589251118 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d82c91327fbd44809f461ab3776ddad1 : FAILURE in 5m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a562a5dd0bf14cc9bdde387701d391a0 : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4ead48728347ab875f16e21ad179ee1f31fa5e34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 05:19:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/996721434d9541be84278c76b9c59621\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8598351b3e7242a4b6711039cb7e2f3a : SUCCESS in 1h 15m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a95d48066b2e496595c22160b29a248d : SUCCESS in 1h 10m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/76f6757507874a6d96a87d42bfefd939 : SUCCESS in 55m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdac7b795e604137a6fcad5cafd7805b : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":2},{"id":"6167f4db6ca4ca70e20082423cc9c0f2499cb8f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 16:18:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f4003d567ecd2d678ae323937f1c93742a7cf5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 17:21:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dd2ee39bb5e422391d09a098c086ccb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6ba0248a1c4ac7afef88588045c587 : SUCCESS in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/d8d736274fd84e95a80b69a3dbdbb555 : SUCCESS in 58m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/140608e2926b4211b2f944061fec7f30 : SUCCESS in 53m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43565a2ce91643ed927b2e0edb6eca22 : SUCCESS in 54m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c883eb104be34821a35b9dedb5e0f3c5 : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68276c5ebe214fbe8221fdc64fd6f3cf : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6165ae8de5eb4fa0a95769535f6f2b61 : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/557ef46f98db4ef189e768f5b7437e37 : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8affe56e60114aa085c4cc1f6ab12000 : SUCCESS in 20m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/78113aea389f4e3591949df6cd9bfb1f : SUCCESS in 14m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c1501c57d5244d2cbeaf1e8a3a940ae6 : SUCCESS in 19m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8b352ad6930648dea09eb2cb6866a6e3 : SUCCESS in 19m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e19b6fc0fb1541aea4527e88c911500b : SUCCESS in 21m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/682024423cef49edb77cfb7cc5aa953a : SUCCESS in 23m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4cbeb92925a24ce093acefaf6b871185 : SUCCESS in 19m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/054a169c0862422d85a19e7bd2c32963 : SUCCESS in 17m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d1ccc33d091a43118b98d765091b8b70 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/013dedf4844c4f77b18b0e8ac7104a20 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/81444d76ebf54137bced0fe4a2ebe14e : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/690db6ea4adb438ab73ca62d479322ef : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f500bf6c017a41da8c3926268cc0bd0e : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/15bfd3cd40cc4fa0892ba64a6a67b84f : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6c5d688562e4843a433d8df7a60c1a6 : SUCCESS in 55m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1dd91951d6c4912bc60214243a36456 : SUCCESS in 39m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dbf3b2ffe33344d68264c15939736345 : SUCCESS in 47m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61a3a7a52a464da59e4cc94ecf760e67 : SUCCESS in 22m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b11a97643d93479cb89a98d1b8d1c5e0 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5418d1acdf1c43b7bdce02e105564e37 : SUCCESS in 21m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecb8a96e51c048f7a98393c75b8bda70 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":2},{"id":"41731d6e7e9f57d76083ad2c0089c4e5c18fbc6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:17:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d282fd7bdb642f19f3cbcb72715ffcf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a08159818e649a786d144bf71cfdb05 : SUCCESS in 56m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e9e8e9541e74248ba77bde1fdef89a2 : SUCCESS in 48m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c31b1d19826248888f93a08e52b3e069 : SUCCESS in 1h 01m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4aa42c0adff04863b6eee5e10d966882 : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"107c11acfefe54ff329d980b782971959bcd944e","revisions":{"88dc30263e7ea2a4fbe05eeacdaa7bd7852032f2":{"kind":"REWORK","_number":1,"created":"2024-09-06 23:12:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/928478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/928478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/928478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/928478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/928478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/928478/1"}}},"commit":{"parents":[{"commit":"015cbaac868494c076a4470bb4dcfa907922d870","subject":"utils: paths with empty components are invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/015cbaac868494c076a4470bb4dcfa907922d870"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 22:48:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 22:53:36.000000000","tz":-420},"subject":"proxy: Clean up some bad-URL error codes","message":"proxy: Clean up some bad-URL error codes\n\nPreviously, we had some really strange behaviors:\n\n  - A swift path (i.e., one whose first component is a valid API version)\n    with missing account component (or missing container but a present\n    object component) would 404. This is maybe defensible, but it\u0027s not\n    obvious that the problem lies in the URL used.\n  - A non-swift path with a missing component early-ish in the path\n    would 404, similar to above.\n  - A non-swift path with multiple components to it would 400, despite\n    there being no way that a client could turn it into a valid request\n    at the same URL.\n  - A non-swift request with exactly one component would 412, despite\n    there not being any conditions (If-Match, If-Modified-Since, etc.)\n    specified on the request.\n\nAs an example of the confusion this can cause, a client trying to use v1\nauth against a cluster with no v1-auth middleware would receive a 400,\nlikely causing the user to think there\u0027s a client bug rather than a\nserver misconfiguration.\n\nNow, 400 on swift paths with missing components, and 404 on all non-swift\npaths.\n\nChange-Id: Id9adcf94a8847d3062c99258602ecba4a8a45edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88dc30263e7ea2a4fbe05eeacdaa7bd7852032f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88dc30263e7ea2a4fbe05eeacdaa7bd7852032f2"}]},"branch":"refs/heads/master"},"107c11acfefe54ff329d980b782971959bcd944e":{"kind":"REWORK","_number":2,"created":"2024-09-07 04:01:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/928478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/928478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/928478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/928478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/928478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/928478/2"}}},"commit":{"parents":[{"commit":"015cbaac868494c076a4470bb4dcfa907922d870","subject":"utils: paths with empty components are invalid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/015cbaac868494c076a4470bb4dcfa907922d870"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-06 22:48:15.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-07 04:01:41.000000000","tz":-420},"subject":"proxy: Clean up some bad-URL error codes","message":"proxy: Clean up some bad-URL error codes\n\nPreviously, we had some really strange behaviors:\n\n  - A swift path (i.e., one whose first component is a valid API version)\n    with missing account component (or missing container but a present\n    object component) would 404. This is maybe defensible, but it\u0027s not\n    obvious that the problem lies in the URL used.\n  - A non-swift path with a missing component early-ish in the path\n    would 404, similar to above.\n  - A non-swift path with multiple components to it would 400, despite\n    there being no way that a client could turn it into a valid request\n    at the same URL.\n  - A non-swift request with exactly one component would 412, despite\n    there not being any conditions (If-Match, If-Modified-Since, etc.)\n    specified on the request.\n\nAs an example of the confusion this can cause, a client trying to use v1\nauth against a cluster with no v1-auth middleware would receive a 400,\nlikely causing the user to think there\u0027s a client bug rather than a\nserver misconfiguration.\n\nNow, 400 on swift paths with missing components, and 404 on all non-swift\npaths.\n\nChange-Id: Id9adcf94a8847d3062c99258602ecba4a8a45edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/107c11acfefe54ff329d980b782971959bcd944e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/107c11acfefe54ff329d980b782971959bcd944e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
