)]}'
{"id":"openstack%2Fswift~928499","triplet_id":"openstack%2Fswift~master~I6ebb19c90dfb1cfbe0535ed3860f2319e5153c05","project":"openstack/swift","branch":"master","topic":"account_quota_meta_migrate","attention_set":{},"removed_from_attention_set":{"22042":{"account":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"last_update":"2024-09-09 22:45:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ebb19c90dfb1cfbe0535ed3860f2319e5153c05","subject":"functest: add checks for quota count API","status":"MERGED","created":"2024-09-09 06:42:59.000000000","updated":"2024-09-10 01:04:14.000000000","submitted":"2024-09-09 22:45:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928499-account_quota_meta_migrate","meta_rev_id":"99689a4489a3fe419df0ff3dc8511fffb215b5ac","_number":928499,"virtual_id_number":928499,"owner":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-09 22:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-09 15:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-09 15:39:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 07:40:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 15:37:04.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-09-09 15:39:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"3badac1e2519e749220ba3683a96036a874cef86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"date":"2024-09-09 06:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a96e61c7df963c7e4647d724e25a7534e3ce2a60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 07:40:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b30848aeff7c49a1a98cc21550d8840c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e785b42c985542d9886bdf528bee6844 : SUCCESS in 8m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9eaea5ff6924845a0a1e35f43d50830 : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c424dfc17a0d49128ddd3a9d710cccc0 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44927c7a26d34472ba8b02c2e57984df : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b4217328b2f44113a0d888ae18525491 : SUCCESS in 20m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20fcdd3fcd4a40cd9d8d8123465924cd : SUCCESS in 20m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cdd8233219e74058b2f353f1aa8550d2 : SUCCESS in 21m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/94c834c02b2d410dad2f40ad61510c63 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/61c88b27215648ac8085a1742aa962a1 : SUCCESS in 20m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/264a1c6e9d124005ae269f6c419b323e : SUCCESS in 20m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0abfbb57d4a457887ab19d1e18dc77a : SUCCESS in 51m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb512512c11344389c76863b1f16aa9d : SUCCESS in 52m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05463082e4b34a15b40e3c4096a041fd : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2bf135b3268f41f8b0490b2e28ea64b3 : FAILURE in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"fbedff16b47a9fe9aacb3fddc50bece69184918e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 07:59:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8838b9af48174aee9dddc46dd794e4f8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b02872f2d1e49c0822160514be5f3a3 : SUCCESS in 58m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d50a7837ea34a05a3316e0463939f18 : SUCCESS in 1h 14m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea938921200f4b02851d9b13569ec3c6 : SUCCESS in 1h 02m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f37ee50bf8524d6192a591306b5a129d : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":1},{"id":"89fda1bfc233a29b37f73d46d141178eb09481e2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 15:37:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da880c53b382dbb24c6c45c99753bb854278ccb4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 15:39:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95413cf61c2d5c25a2657bd0567b49d475c8b969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 16:32:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79dcb18b111b4503814fba342b98cc58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa54d7a2390f45ca80b28f83784a3fab : SUCCESS in 8m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a363db960ac74e92bc86a1187a9cb440 : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5ab552916cd4e7489a3d89fa6785080 : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/273fd589908b4a7fbadda55340c3f5d1 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f99c380e1a444eb69a5b08af2b2fc19f : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2567e7050a53402e8e327278784082c0 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ebe1fdc4cf1d49c097b324fd269f5c85 : SUCCESS in 14m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2b6e2f54144a4b9796fe426d2ab90799 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e4a86f403cfe46f780c26fb55cd88acc : SUCCESS in 20m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fc20768c95c44220896b11eb62deb635 : SUCCESS in 15m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/94bf404c45fa46ebad1af051a9834a66 : SUCCESS in 50m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7057f54ec0b34e0b9fd8bc15e3fe9596 : SUCCESS in 45m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3592ad3945ce4f9d9a68a6bc6adfb6a1 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a542155c6c3415aa9f13820f622f304 : FAILURE in 24m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0da6f3740253e99e0458143591dd8b383fc236fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 20:13:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eebcc8503e5e431e892a5b4f742c678d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a1d4fb8216f409284bb6fb21169392e : SUCCESS in 1h 20m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8567c6208704e44be7a108daf90d1d3 : SUCCESS in 1h 11m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9781554ca9ac4929ab18281af3e07b5f : SUCCESS in 1h 08m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d004fd551ac64b5cafd57455057d1f0c : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6cb06c1371ec393075c85f37ac2e67bd246923fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 20:56:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2db5ebcb4fd9a7db5e3130591fdd58cd51e64d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:45:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53bca569d64940de8cbd071bda7dc4bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c9996baf6148768d9e5ef27fe826d2 : SUCCESS in 8m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a1a7a34252f467596530ccc855109b7 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a63d29c4cf2340b9a4c725ed34bd7947 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b55b84520d72489a98fcdd343956c91c : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed18d34eaab5431ebbd77f2ad1869559 : SUCCESS in 22m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/173183ac4e5c4e919ade0d6ad2741e85 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b7c26704c034240a5ae768c0174e08d : SUCCESS in 19m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/633eab896994432697d69c414891ef27 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/21503083a6b44f0b9578d9945b3f732f : SUCCESS in 17m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b4671941aeca4cd49af82c90e3e53e35 : SUCCESS in 17m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc77d85dfb104f0eb7a484f155bf0335 : SUCCESS in 48m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9a89b3aa8884785b6745483c0d2eb3a : SUCCESS in 38m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351c9dea6ddd41869de79a813b4ad09b : SUCCESS in 7m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/43071e6706bb44f68e480c00a7baeb65 : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a84ed0fdf6950feb52543385c59d7361e4404563","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:45:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3dec53ca5be086c39cd16f489c2ece0173aff47a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 22:45:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ae948272ec040b3ae048a454041d80e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89e8c77a32e84bf18590dc670dc041da : SUCCESS in 9m 27s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/899af455c38a48a19a95e7b49d43c80b : SUCCESS in 20m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d05900372e0c46ccbe46f5bdd297bb99 : SUCCESS in 18m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dfa8e5dcb88a4187877553bd690b83c5 : SUCCESS in 21m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a4166540091644199d579efe340bb597 : SUCCESS in 36m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/538da89cbc0e4fdd85601470ecc7597b : SUCCESS in 28m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b19914a296842b282e3158c5095d9d7 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a6c027180b3b4484b8a40c11da73e73d : SUCCESS in 12m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b2b5c479aba463db43d4d8b1d3ba0d4 : SUCCESS in 20m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/59daba71aa1248e0a29e4add3e15a396 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/811d5da4d6b74b17a469dd1668491cc6 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5ee50867f8b54cd8ab85528a15369520 : SUCCESS in 10m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/622ac3d26c5a4de6b30508207ec7dd86 : SUCCESS in 5m 44s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/788e935dc5d949d5ad37f4d6325cf167 : SUCCESS in 6m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2dc1fc231714a749b706f578e96191c : SUCCESS in 48m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70867c499a0441c6877a916ecd2e8494 : SUCCESS in 29m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22e8e0eb493b4a5095e0dd05310ac86d : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c6419af54394806b5066f2847fa7c84 : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":1},{"id":"297f42a90c81217a274d43b59461acd81057bfb8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 22:45:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"24404c5aa21b846c55de45a25d70c35d498dfcff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 22:46:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51482d310bcb4c2dba86d0cda4744859\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933c9561df2846ce956c9153fa2b4b09 : SUCCESS in 57s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/65520052784a4e4696d5356a6ae13dbe : FAILURE in 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99689a4489a3fe419df0ff3dc8511fffb215b5ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 01:04:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c8a920c5634cc39aabe189d2413878\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b9ddefd3cd940b186f3e116e8875a4d : SUCCESS in 50m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/360123db48fb4b17942a9378c64d0b9f : SUCCESS in 48m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c4c06280d1648df9b55f4292bcb9022 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c1c56f6c51b4b848a04321bccc5b706 : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"112423f59cb46dc897cd9ee7c7fc4de1260d6595","revisions":{"112423f59cb46dc897cd9ee7c7fc4de1260d6595":{"kind":"REWORK","_number":1,"created":"2024-09-09 06:42:59.000000000","uploader":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"ref":"refs/changes/99/928499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/928499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/928499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/928499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/928499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/928499/1"}}},"commit":{"parents":[{"commit":"4eefae2482aeacbf27b58dae44b7f5808bbfa3de","subject":"account_quota: migrate quota_bytes and quota_count to the sysmeta namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eefae2482aeacbf27b58dae44b7f5808bbfa3de"}]}],"author":{"name":"Chris Smart","email":"distroguy@gmail.com","date":"2024-09-09 06:32:58.000000000","tz":600},"committer":{"name":"Chris Smart","email":"distroguy@gmail.com","date":"2024-09-09 06:39:24.000000000","tz":600},"subject":"functest: add checks for quota count API","message":"functest: add checks for quota count API\n\nThe account quota middleware was missing tests for the new quota count\nAPI and it was also only testing the legacy API for quota bytes.\n\nThis adds the new APIs quota bytes and quota count to the functional\ntests.\n\nChange-Id: I6ebb19c90dfb1cfbe0535ed3860f2319e5153c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/112423f59cb46dc897cd9ee7c7fc4de1260d6595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/112423f59cb46dc897cd9ee7c7fc4de1260d6595"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
