)]}'
{"id":"openstack%2Fswift~928667","triplet_id":"openstack%2Fswift~master~I5669a814377242e9939a09a42bb36642358b85b3","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I5669a814377242e9939a09a42bb36642358b85b3","subject":"CI: use private IPs for multinode tests","status":"MERGED","created":"2024-09-09 16:39:52.000000000","updated":"2024-09-09 21:53:54.000000000","submitted":"2024-09-09 20:51:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928667","meta_rev_id":"47e0474d26a856b6cadf621622b0cda4819b5c29","_number":928667,"virtual_id_number":928667,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-09 20:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-09 17:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-09 17:24:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 17:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-09 18:29:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b65071c3715ee0abb9e970cbeb1dd242a3572f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 16:39:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"382fb4430cb46146ec6cc0283f22a604627957a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 16:40:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0617cb0767dd6926597209e0574bac8243e0f3f3","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 17:13:29.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4ef43b4473a4172a65fc12a9258c037\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/718e607cd00d4543bf703988bbf4b101 : SUCCESS in 9m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c383a6a6e714460f9d89f305fd3ea938 : SUCCESS in 11m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fd0e3028811e4bd195f15ad7a45e7cf2 : SUCCESS in 10m 37s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d1a4b5039004cd5872225a0d4325ca0 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6936a2a340c9456997919f1a48057b10 : SUCCESS in 12m 03s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/059b6f081d3f48d4be793176924708f2 : SUCCESS in 17m 55s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/810caa616ee84d309cbbe3c8ab6390c5 : FAILURE in 18m 51s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/c672d792b96e412eb7e7ec69b4372a2e : SUCCESS in 11m 57s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/f145f60b4e4d45999e268722750a4ed7 : SUCCESS in 18m 30s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/6cde5abf6e45453691108e0fc8ca98b1 : SUCCESS in 18m 33s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/b7cd978cd20046ea9bef400d962728dd : SUCCESS in 22m 47s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/8f043f6a8b4243a2b4a15beb1a54123c : SUCCESS in 18m 17s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/4352192e94a04c12a6eac2504e058de2 : SUCCESS in 23m 39s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/06bb30904e4046e1907c2ba509ca3a60 : SUCCESS in 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ed19196f7328ed3edd2c10d1e5437b188bb4d4c6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-09 17:24:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16277b1166ab10a742334b0cbaf1d9693c718b39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 18:29:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81d52b33c0cd4055a7324af3e045034c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d02d1c87ae884104a7040047b2ffdaef : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/c0664f82af35420d91bf40a01649de2c : SUCCESS in 56m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/07168cbcc38c48d6af6c91553a55b735 : SUCCESS in 53m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75d74e234b0b4843b5160c6288836c1e : SUCCESS in 59m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e3bb4415f47470b80ac61fbb91ad920 : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a45c75f919b41798db02212ceda4b56 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8055ed96fe84a5dbbc0195311e29b10 : SUCCESS in 23m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa4022edbffc4d4590500f05714fba7e : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d06dd39990104b43876965f802cf2dd6 : SUCCESS in 16m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f72429b18d164dcb9346f922bdac6ea4 : SUCCESS in 14m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/888d5c10ad854c7c855037e71287c252 : SUCCESS in 18m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/792ff827a26c4454be6d6055cc5cc1d7 : SUCCESS in 14m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/49b60c5ac94f434ca2f77f177eb1e610 : SUCCESS in 32m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30d68977d0844fc183934e1a0dd3acf6 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4be344948fb54e7caef9907652742b02 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b19acfbc9c52452098b374e93f48b1d5 : SUCCESS in 16m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/875e03bf6aac469ba37dec3238834bba : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/895c47f4ddde46e3bd89820caaac31d3 : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c6f17a4c322a46b6b73ee9d1285a9123 : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/99ab899093c7447bbe5f767a69678b85 : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ed66f7dacc34521bf1e75c58d03f8eb : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a7d3cf828f6c464ba804a7703378a9fb : SUCCESS in 6m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5006e4c0aa1d4a62a9d5e12021026deb : SUCCESS in 56m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0647377b1e70483295c97606b8ea1968 : SUCCESS in 47m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/705a3b99638a4f2185c094f55fa684e8 : SUCCESS in 1h 05m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f79ed1c6c85b49bcb250599faaaebf74 : SUCCESS in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81dddd4d05c641398c5ef5a8a74fff1d : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/962aebb1866a4827bd36232eac2fe28e : SUCCESS in 16m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caa5420cf3d74c529f78289be906537b : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":1},{"id":"31c8d4ea9fbeaef1d5698ce7a0d56e2e6b7f0a8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 18:29:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1449fd37e87b8501d7660cb78362392fd292d117","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 20:51:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5edde79720841bb941b604f0516c01c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6e51db8931490e923f76b2fff204b1 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/2eeb3b63beea480092b3dca10838a621 : SUCCESS in 1h 01m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0187da8c617c456386bca6c6b01fc9bf : SUCCESS in 50m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd95f872d2e9486fb142aa4bb7097de8 : SUCCESS in 57m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f71d90efb5ff4946afec65d8b8d3a7cc : SUCCESS in 7m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a3b2478d6659410a801f66ed23bc21d4 : SUCCESS in 11m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c65f1339345446ee9321b50221ca615e : SUCCESS in 17m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8d7b78c42e4c4cb8b2865ef8f816c015 : SUCCESS in 10m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b8f94e693a174343b3cecdc9b922ea7c : SUCCESS in 10m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a8f4688953804c75801bbc11a40b62dd : SUCCESS in 13m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e6e22ac637584503955fbbe435220b85 : SUCCESS in 12m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/176019480aa547bf89394e6b15fef67b : SUCCESS in 19m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd621d5932a84d5dab1022ef8750b7ac : SUCCESS in 20m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1db87c318c67459b8cca6e77c6eb3509 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f10f4b0226c54d7f9303382241881a07 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ffe754c6c71c4c3bb2752cd4bcde123e : SUCCESS in 18m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4fc86395ff024c32b73737ced094aace : SUCCESS in 6m 38s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c5d5962f78794d01ad00e5bed5be7e8c : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/46153420e70346cc89f8dd3d2e8e6942 : SUCCESS in 53m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3bffda76adb146b8a24a6e5d0f914de4 : SUCCESS in 50m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0aed7aec38754c5491c6e736f2c5f507 : SUCCESS in 51m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3d61d3405d845bdae7fe5d5ce0b8c89 : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0f87b64c08c44548f29d112df41c980 : SUCCESS in 3m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ba0e9ca9fd7458bab2ccfe4184e147a : SUCCESS in 22m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf2d6a0f8ce54d0da9abd42e5d603f83 : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d25c779d155fc63362437b2996c9913703def141","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 20:51:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7858f6612dd37d5a4877e37821df18c4af555abc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 20:52:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8841ebbae7724e228aa2657023f4c099\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a248d8137fcc46199078b07981d046ec : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c8832379680e45dd8d04f1aea08e9bfd : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47e0474d26a856b6cadf621622b0cda4819b5c29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:53:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97b94022b9fb418197e337e175bb63a3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ef98409edfc427cba59bec9a78354c4 : SUCCESS in 45m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f39f081163b24babad565b2a1c7b35ba : SUCCESS in 1h 28m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d6a2145871d4d488d684b19d9f2ac41 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/327c9e7b9f9a43b4948f4ff0d9bfe210 : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7db1ec3cfd9e130b67b97e6a4d5e24cfc92e4eb","revisions":{"f7db1ec3cfd9e130b67b97e6a4d5e24cfc92e4eb":{"kind":"REWORK","_number":1,"created":"2024-09-09 16:39:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/928667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/928667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/928667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/928667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/928667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/928667/1"}}},"commit":{"parents":[{"commit":"146bfeb643648605927185afaddd17514f76f198","subject":"Merge \"proxy-logging: Clean up some timing assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/146bfeb643648605927185afaddd17514f76f198"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-09 16:39:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-09 16:39:21.000000000","tz":-420},"subject":"CI: use private IPs for multinode tests","message":"CI: use private IPs for multinode tests\n\nIf the nodepool provider uses NAT\u0027d floating IPs, the public IP won\u0027t\nactually be available locally for binding. This has caused a bunch of\nfailures on raxflex lately.\n\nChange-Id: I5669a814377242e9939a09a42bb36642358b85b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7db1ec3cfd9e130b67b97e6a4d5e24cfc92e4eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7db1ec3cfd9e130b67b97e6a4d5e24cfc92e4eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
