)]}'
{"id":"openstack%2Fswift~928773","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ic23348f234c308ae8db395ad06bc7e27e3576e9e","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-id-timestamp","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-01 05:13:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic23348f234c308ae8db395ad06bc7e27e3576e9e","subject":"Refuse part upload to aborted or completing session","status":"MERGED","created":"2024-09-10 09:26:01.000000000","updated":"2024-10-01 05:15:31.000000000","submitted":"2024-10-01 05:13:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928773-p-mpu-id-timestamp","meta_rev_id":"49f8bfba57eafa7685739c31a3bbb1fc5d05276c","_number":928773,"virtual_id_number":928773,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-01 05:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-01 03:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-01 03:36:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-10 10:18:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-01 03:36:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"e9198060dea1b0537647fcc0b7e65f627dff94ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-10 09:26:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c85fd1d64e12c64aa55f8d0e00c840af6424946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 10:18:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9aaad29f98e34fde95a4efd536b57027\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28bc5b187e0441ec8faf042d356cea68 : SUCCESS in 9m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f8334f9051a40459128d1930bc5b92a : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2731f07b113b47c2a0afc6b5f0296426 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5591a370e95f4539a8df8438c00bc294 : SUCCESS in 20m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e8c42f628c4f465096d4ab6f9318938b : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/903731e99f5e4c64a722da3507960788 : SUCCESS in 19m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3e348c08d6b54fdeb0da9248f353ead2 : SUCCESS in 19m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bf59b21cb1fe4f20b84ca926fa37cec5 : SUCCESS in 20m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/419aa318d1c549589fd532f0b82e41be : SUCCESS in 22m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dfc24d851b4f46e0b2e7606dda94ed7c : SUCCESS in 32m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75a4db4d20464297ba5ad6eae987f5b6 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6d07eaadcf984f899e4884531f63cd81 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1955acfaac784e8eb5861a7a9ff890f7 : SUCCESS in 18m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a317c10442f34b7f9c3746e6b024eab9 : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/386191d1a82a46f49dbe4de4d94f30da : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d92e385ec744191a60d2c37042038e2 : SUCCESS in 20m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2054910a96b4355b6910393e2f2738c : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/27d14ed8e5e8413da97da1972bcecf0a : SUCCESS in 13m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb052cff7d5341f799930e61f043b5b0 : SUCCESS in 8m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcafd93f744f4fc8b8487737714bb4e4 : SUCCESS in 47m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9b8e4dbf574454ca617b0080db79452 : SUCCESS in 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb86fb9b6511453194b1c7cf94c71a39 : FAILURE in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3032dade97b4796bf66fb081b87d83e : SUCCESS in 16m 29s\n\nWarning:\n  Comments left for invalid file test/functional/test_mpu.py","accounts_in_message":[],"_revision_number":1},{"id":"c22f89ff01ef3f9d9192756850e7d571ed7385bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-10 10:45:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"26ac5af35759e8b2580c2d12824201ab89f5319e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 11:41:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72163659c52b40b8971b0cd56e55c55e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd2365f1f30427a8d0e9759e34a28ed : SUCCESS in 8m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00b7ef72ad3444deb0dacec582bbb4d0 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ffe0299f22248c68bf577f2492a9442 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cdec248225a4a368f858ebb68647f40 : SUCCESS in 22m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28a5fe22a33047c0bd5e63ae5048d8cd : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4d0e0727a6654cbda350fc5d3bb1a072 : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a2a56c770628472fb97c32a148e6acce : SUCCESS in 19m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/650dd76767d5488d9ed986df209a291f : SUCCESS in 19m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/69e6102c11bd4bc98530022261ea73fe : SUCCESS in 17m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/807fdda6f6d74d70b62094d1d72b4a67 : SUCCESS in 32m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24fba7c872a743be917d9f8933471020 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b9582e9e726437dba4131eeda0b02a7 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/22d98901288f4aca937c160622092d0e : SUCCESS in 21m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/91ae4205fab3457ca3f37200f7ce95c7 : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1999b37b110747d9916caf1a76006ad9 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d9d39d209bf2488f85a106a193f9895f : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d269bbdb9e2f4a8aafbaa749c90180ec : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd5031661fef40208d1caf62e6a71c4d : SUCCESS in 12m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1e0282bbbff4da5a542f7ff5564ce8f : SUCCESS in 6m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/70e61db8a100498e85fc57e24c5c8d60 : SUCCESS in 47m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb97d7bd141349fda57e31d29a92943b : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c570b4c54f2458f9165f8d45a9c6f2c : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68bd4363abcd441480baa5d900b7e408 : SUCCESS in 16m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a84227a057cdcf790464e1d79dfa533040a9b98c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 15:13:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9075deed5df843e78aab3edc597e5a2d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97d37073cc994fbb8d7d6efc4d43a241 : SUCCESS in 1h 25m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1dcbc003c1c04239b40a41554d74e373 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1336a21447d140d8988103a48fabae17 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7a3f01d2828433c95df2a15d79f654e : TIMED_OUT in 1h 30m 59s","accounts_in_message":[],"_revision_number":2},{"id":"0f071b1a2f97df541112c6ccba9d1c57715fbf83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-17 14:14:47.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"643282d59aea3c8863eb9bb5a61025757c48a612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:16:05.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9903d72b83e54abeb8216993ac513909\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6abc37beae754576a19829fd27fc8cf8 : SUCCESS in 9m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdd07f3ec94144b2880276d82874c0c8 : SUCCESS in 22m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1e60527985d4ff09e4e4a948f133122 : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9278d87e3bd644b0abf8d3b734ec20c3 : SUCCESS in 20m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe1b1c0c2b2a46519cc0e6e208e9d722 : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/946e8c1bbbf24a3a83352a82a407c5eb : SUCCESS in 26m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5935f51e294d4a5692403a1d0d17a617 : SUCCESS in 17m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/87e99ec954374e5d8bc28cea5576a262 : SUCCESS in 20m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/88900885e50c45f792cdab1bc9c4d1ff : SUCCESS in 19m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/391c5e11c7e34758a9cb89b006b08d86 : SUCCESS in 27m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/738c16dc7d1043b9a579dcd7850b4b8c : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/333c2138c98241ae851aabdf2dad7798 : SUCCESS in 19m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7186f36d2f6942969bfed0a2d88122ce : SUCCESS in 20m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f9d9f677e9784e9b8a0437712eb92847 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e15b05cdd8d54729a2376c1873c882e6 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/626502b1cd5e491ba384254989feb644 : SUCCESS in 17m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03813ed622764c3fb817efd95b5be0ac : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/857c2551b5044a758227c4ecec93d60d : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b09d2e5dedaf4ef0a054bd164a225598 : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/30bd894f5a2f467293c22e841f7c929c : FAILURE in 55m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bf32b27a7f642b7bf8d30fc7db4f4fe : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/016ebde9553b4239a8cd1f07ddb3a56b : FAILURE in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6342def4d8f241c9a1bb0fabc33328b7 : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":3},{"id":"14d61e3ed0045e2895b7904d2734fb7261b43f79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 16:13:44.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21100796b5e54e038698452a1c886db2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15c64c783f1647e0addc7a020c73abec : SUCCESS in 1h 06m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d3325f5defd4ba8b6d615991f6c0e7a : SUCCESS in 1h 05m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec4df94339d347faa6b74ece2948c217 : SUCCESS in 57m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/df95a9861dd043b4bd5d805d77db1dbe : FAILURE in 1h 19m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b5ff546acebb8a5d18ccaa24f486c5ee0f12485f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-20 15:52:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"84fe6ae97d12d44a7ba9f39402fc393989bf0c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 16:57:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84d157b8916644debbef9be949a3ff50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e9f63f19cf4f19908e0ebba000366f : SUCCESS in 9m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33e5a1108fa6437fa1d7b5ec142ef05c : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fa54e994c4b4c09a5bafe940d7af72f : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ddfc293779741f48b6694b6c0530502 : SUCCESS in 23m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d50c504e6f6b4f6c84e02d1c0368b895 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c5b6044e6b064458ae7981cc1da5840e : SUCCESS in 20m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26011a091d4f49eeb1c1d252a0999698 : SUCCESS in 10m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fbc5fa9a48984d0aa8cbf20b291fb961 : SUCCESS in 16m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1917ed50320d4ed096663a48619ba13e : SUCCESS in 12m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5ba17e7ecf5746838aeefda518403c5d : SUCCESS in 28m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5f7159d39b834bdc8f52474d5de3c600 : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ceaad848f1e47d0b8f81deaf947bdcb : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8544051646fc4d7ebdab06f551c4082d : SUCCESS in 19m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b3af5f93a9584e068e51afc99be28d36 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4aa8b1094bf543cbb81b097d7d1ff9cb : SUCCESS in 18m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/45b2dc323abc42b1be7e08da83303837 : SUCCESS in 18m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9068769e7bde47ec8e51208bf468478b : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f9cf7976607c4990915b7178bd3bdf99 : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8add2c77164246a7a08ec7470baf58d1 : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/229fdf261a7340cd866cc6b890363870 : SUCCESS in 56m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cac448c87414a3381fc07108f6b50e8 : SUCCESS in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d98c89f43394364b4881cf4815ed9a8 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e2fdc000cdcd48dfbb8a891d7f3b5502 : SUCCESS in 19m 14s","accounts_in_message":[],"_revision_number":4},{"id":"5de53c6cbea724e3580a2d8f689f7f01d8ac1740","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 18:42:33.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327fbe6af2bb4c38bd411a41d15e17cc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d343e72a9cac4adab2f0e1b4792f1bf7 : TIMED_OUT in 1h 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d65a23434594358bbbad56f5f6c85a2 : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6106f6a4d83c4f3f8cf877b74244c5af : SUCCESS in 1h 30m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c5ad596ee4243a5a1425ee95e859603 : SUCCESS in 1h 15m 55s","accounts_in_message":[],"_revision_number":4},{"id":"f84cdc8a2adff2f54d0b0a83c11f6761ac8ad77e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-23 18:26:30.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7eeac90c5efba70376d62364d40fc5dc234c66a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 19:21:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54c48699783f4e2291907a61fbff49e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60064ff3fe70468f8c83b8cb4d4ae487 : SUCCESS in 7m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ac2db61f381421590bf36818bc0f340 : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2feeb70542a94976895b2af96cf2c8c1 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53a0ea3c2bc84a4f84570d843aac9205 : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a6313d9049c54e0d82389c6a8bba8efc : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c9f4b1120356496786292afeaef78ae1 : SUCCESS in 13m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/492dc929f61b4cbd9b04adfec3bbcee1 : SUCCESS in 18m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d3b36bbad751435caeefe1fbfdc21c2e : SUCCESS in 21m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/83b69f9bf5ce4e2a9d8e6d6099ae8268 : SUCCESS in 22m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2b31163516a34c2eb74897260500f04b : SUCCESS in 27m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ddd45f4f95944328816d0b0c09140b9 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/45870fb22e744dd6bfeb68c173ae75f7 : SUCCESS in 19m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6bfcc9c8a774ef8a3c999f4d3d3ecd7 : SUCCESS in 20m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/905f4aed35d74b5fa9653205bd6d775c : SUCCESS in 12m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7328967a3af04d0881c99659ac987c5c : SUCCESS in 19m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/74509498ac004cc3a374fc67cad5f8f1 : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38059f9175594291973783fef02048eb : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/556b57c7683b42e99794fae55904a281 : SUCCESS in 14m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2da1e68e3de8478694d4e65bfb28f3a0 : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/73d974b8b943455e89150ac421a57d87 : SUCCESS in 51m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7af1a38daab34866a15564c91fa6a08d : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962480a539a1467a89481d74ecfc4b07 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7542f5b70af74b188ec6ae4fdfdba7ed : FAILURE in 5m 27s","accounts_in_message":[],"_revision_number":5},{"id":"df4a196cf1034630f4ff8308c5ecbfc1ee0f4061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 01:13:37.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe776758fac453a895821b9f519f3fe\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b3642265e2948ff9589711caa0eae90 : TIMED_OUT in 1h 31m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/995208ff6cfa4099a8556ffff6f66bbf : SUCCESS in 1h 22m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/46e7af8200d5438aacda84b9554cfbde : SUCCESS in 1h 14m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1517b329d6ea4dc4a7ae2a487aa6ca95 : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":5},{"id":"c057e13edd7044037f14eb33a4f3ee0ac280f208","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-27 15:58:04.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"7e06ab3fd10c515c7c0dd1ef0a383274c73e6df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 17:10:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd43d9d05ac4f2f93e7d4218865b4cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a411d016684b45ad8e6498c15e46cfbf : SUCCESS in 16m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b267e11f45d749e9ad0876fed9f67bb2 : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8baca4b8a4744fabebe8f845017d8b4 : SUCCESS in 21m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f26e59b9fac343cb9dd0e1c191f5087c : SUCCESS in 22m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/190ab878e4bb41a7881a573b3d7713e2 : SUCCESS in 11m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b54e6c82b31f47119c071a92d16c8db7 : SUCCESS in 19m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b91c0ada7e0a43cfb24fab69a45ee038 : SUCCESS in 23m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3fc46dae197e44e89319c9ee64b45cef : SUCCESS in 23m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1adf34fc2c2f484bad8516a567f15bb9 : SUCCESS in 24m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1254d334d5c5467080c67ed687a7a0b3 : SUCCESS in 29m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e2912da859a14e65846203a75aad9b75 : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e3459c641d31466ea0e262dc7e81a70b : SUCCESS in 22m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/afe01207a21144489ef40ba19c3872bb : SUCCESS in 19m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/45b667fe462a4819ae7c1469ad806afc : SUCCESS in 8m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/00c6effbe88948d3844c9b48e6f74fe8 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7d9505f3cf2547368ee54fbda4d49781 : SUCCESS in 20m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd3dd4f5b75d486fb680ed2d99987d2f : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13597499fd8c4829a61c50645c10568a : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f75d32972ad745afba67e207fd4bed44 : SUCCESS in 9m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f3beaed161244128982ec54a987a6c0 : SUCCESS in 49m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a2af30241384b4b9a77150db1ce5233 : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a7c93a67504c94b3ba02cb26116444 : SUCCESS in 8m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f19d96ef3354e3ea0ef9a81426e6257 : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":6},{"id":"69e626e6524ce4ff9065b8a99e37269a52001de4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 19:19:37.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85a7bfe9f15d49bdb8b3601d6ace662a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d5bdde146ed46e09f6292a82089bbb8 : SUCCESS in 1h 24m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b47c8bb3a7c46b29050a713f34ec857 : SUCCESS in 1h 22m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1aa1a427ce0146579dea38b8d4f82806 : FAILURE in 1h 05m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e2c67781c2b46b08116aa872dcc2838 : TIMED_OUT in 2h 00m 56s","accounts_in_message":[],"_revision_number":6},{"id":"f3bc5a8870bbd24bce5c61f661361fe40eba8b7d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-01 03:36:29.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e17f49961267d4948415ee8bac65233e23bc17ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 04:16:38.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1ef727dd9ca4d5c2bdcafd7cedca13fa46a055ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 05:13:36.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e30f4e3212b34533a1a3aa73b7958090\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b302859e87aa45cfab8f6470b7c61982 : SUCCESS in 8m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/223fcb5c442f4535bec4c07e29823956 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3099e47e970f48c7b538d4ddc9545ebf : SUCCESS in 20m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/92198004bcb34ae3892402f7eb9a324b : SUCCESS in 15m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dee1a9762b814c94b55327e3b14c3dd3 : SUCCESS in 10m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d398cd6fc8164e10a0711f16f68aedb6 : SUCCESS in 15m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/661e882c03d84377a4b46eee33bb3d18 : SUCCESS in 13m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a6e02ec0622b429fb480d43f0d5cdccd : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9affc435eecf4b5fa1e9698661668375 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/136388a85cac4bf69a748a75046ff2a7 : SUCCESS in 16m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9d8fefa5deef4d4082b784ba87883148 : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b99950577f69438dac18722d8967804c : SUCCESS in 8m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d5f903a10f92498c9e810b4ca080e1b0 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8fb294d0669a42ddb3fad6a4769e2e14 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e67555eef2546f8b8bf9ec4d89a232d : SUCCESS in 9m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/508c248ad19f47f3b864fd4c6902dc51 : SUCCESS in 46m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86ebaf2752bf43c696e70dff4bd767b1 : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56687972ad3846528b6748c27ce2c3d4 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3dc9e9529674474f9d7f468af98b983c : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":6},{"id":"4e0bc7e9dfeeca3a698fa03aadeea1e05b78aa7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 05:13:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"49f8bfba57eafa7685739c31a3bbb1fc5d05276c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 05:15:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36b7570cbca04a86962880cda704b8c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03884adad4ff41479cae529ca4a871a3 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/357670c9c21140558b2e7ae2636af27e : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f90094eff0ff6c83b4138a0f5b29b7179315c18f","revisions":{"39baeb09e6d1e67cb722e37f8e7d3a1c9ff17b55":{"kind":"REWORK","_number":1,"created":"2024-09-10 09:26:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/928773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/928773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/928773/1"}}},"commit":{"parents":[{"commit":"d247a0f8c85d65a408d56cbce4ce693b4a5e09f4","subject":"Exclude aborted MPUs from session listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d247a0f8c85d65a408d56cbce4ce693b4a5e09f4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-10 08:50:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-10 09:21:54.000000000","tz":60},"subject":"Refuse part upload to aborted or completing session","message":"Refuse part upload to aborted or completing session\n\nChange-Id: Ic23348f234c308ae8db395ad06bc7e27e3576e9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39baeb09e6d1e67cb722e37f8e7d3a1c9ff17b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39baeb09e6d1e67cb722e37f8e7d3a1c9ff17b55"}]},"branch":"refs/heads/feature/mpu"},"2e4c99a62310d155cc85e34ff8e700c10af24040":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-10 10:45:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/928773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/928773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/928773/2"}}},"commit":{"parents":[{"commit":"e4369f01610fe06250d24de33884e33852a19dd3","subject":"Exclude aborted MPUs from session listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4369f01610fe06250d24de33884e33852a19dd3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-10 08:50:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-10 10:42:48.000000000","tz":60},"subject":"Refuse part upload to aborted or completing session","message":"Refuse part upload to aborted or completing session\n\nChange-Id: Ic23348f234c308ae8db395ad06bc7e27e3576e9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e4c99a62310d155cc85e34ff8e700c10af24040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e4c99a62310d155cc85e34ff8e700c10af24040"}]},"branch":"refs/heads/feature/mpu"},"0009c0d8dc34883c9f18e75fffa70bad679d3605":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-09-17 14:14:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/928773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/928773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/928773/3"}}},"commit":{"parents":[{"commit":"2b73112e1249797c639fb3aaf12624abbf46ce08","subject":"Exclude aborted MPUs from session listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b73112e1249797c639fb3aaf12624abbf46ce08"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-10 08:50:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-17 10:43:26.000000000","tz":60},"subject":"Refuse part upload to aborted or completing session","message":"Refuse part upload to aborted or completing session\n\nChange-Id: Ic23348f234c308ae8db395ad06bc7e27e3576e9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0009c0d8dc34883c9f18e75fffa70bad679d3605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0009c0d8dc34883c9f18e75fffa70bad679d3605"}]},"branch":"refs/heads/feature/mpu"},"d6dba5905d8dd665cc0e1ae62f53d9274ba501cd":{"kind":"REWORK","_number":4,"created":"2024-09-20 15:52:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/928773/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/928773/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/928773/4"}}},"commit":{"parents":[{"commit":"2b73112e1249797c639fb3aaf12624abbf46ce08","subject":"Exclude aborted MPUs from session listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b73112e1249797c639fb3aaf12624abbf46ce08"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-10 08:50:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-20 14:06:41.000000000","tz":60},"subject":"Refuse part upload to aborted or completing session","message":"Refuse part upload to aborted or completing session\n\nChange-Id: Ic23348f234c308ae8db395ad06bc7e27e3576e9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6dba5905d8dd665cc0e1ae62f53d9274ba501cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6dba5905d8dd665cc0e1ae62f53d9274ba501cd"}]},"branch":"refs/heads/feature/mpu"},"90a5f397333c2a4926d2a9b3aae9e1372b1acfe8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-09-23 18:26:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/928773/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/928773/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/928773/5"}}},"commit":{"parents":[{"commit":"3cf00a120836cb2ab77014975e0a4f9f83ca959f","subject":"Exclude aborted MPUs from session listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cf00a120836cb2ab77014975e0a4f9f83ca959f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-10 08:50:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-23 17:56:28.000000000","tz":60},"subject":"Refuse part upload to aborted or completing session","message":"Refuse part upload to aborted or completing session\n\nChange-Id: Ic23348f234c308ae8db395ad06bc7e27e3576e9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90a5f397333c2a4926d2a9b3aae9e1372b1acfe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90a5f397333c2a4926d2a9b3aae9e1372b1acfe8"}]},"branch":"refs/heads/feature/mpu"},"f90094eff0ff6c83b4138a0f5b29b7179315c18f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-09-27 15:58:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/928773/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/928773/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/928773/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/928773/6"}}},"commit":{"parents":[{"commit":"99daef8f764b410434f2d8f282df9c593989d9a5","subject":"Exclude aborted MPUs from session listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99daef8f764b410434f2d8f282df9c593989d9a5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-10 08:50:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-27 15:57:08.000000000","tz":60},"subject":"Refuse part upload to aborted or completing session","message":"Refuse part upload to aborted or completing session\n\nChange-Id: Ic23348f234c308ae8db395ad06bc7e27e3576e9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f90094eff0ff6c83b4138a0f5b29b7179315c18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f90094eff0ff6c83b4138a0f5b29b7179315c18f"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
